)]}'
{"id":"openstack%2Foslo.config~877084","triplet_id":"openstack%2Foslo.config~master~I9e3c1e5f98e2ecfb5564d8dbd608d19d4a66cfce","project":"openstack/oslo.config","branch":"master","topic":"oslo-bump-bandit","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2023-06-27 14:01:02.000000000","reason":"Change was submitted"},"28522":{"account":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"last_update":"2023-06-27 14:01:02.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I9e3c1e5f98e2ecfb5564d8dbd608d19d4a66cfce","subject":"Bump bandit and make oslo.config compatible with latest rules","status":"MERGED","created":"2023-03-10 09:23:40.000000000","updated":"2023-06-27 14:02:10.000000000","submitted":"2023-06-27 14:01:02.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":2,"has_review_started":true,"submission_id":"877084-oslo-bump-bandit","meta_rev_id":"4e94c66b27af46a7689f80cb504c16bdb8eae7be","_number":877084,"virtual_id_number":877084,"owner":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-06-27 14:01:02.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":2,"date":"2023-06-27 12:04:41.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":1,"date":"2023-06-27 12:04:41.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-03-10 11:48:51.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-04-10 02:11:25.000000000","updated_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"reviewer":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"state":"REVIEWER"},{"updated":"2023-06-27 12:04:41.000000000","updated_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"reviewer":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"state":"REVIEWER"}],"messages":[{"id":"d69e96cdeb69e68571b3fc4b57f24f5260cc5e87","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2023-03-10 09:23:40.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"e79628973a998b61aae03d63bd733101929d74ab","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2023-03-10 09:24:49.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.","accounts_in_message":[],"_revision_number":2},{"id":"dcf89c2e4d90733e38f134cac00726c404bee7a7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2023-03-10 10:27:26.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"c5015f121c1ebf012a41825c01ab58f18f63acec","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-10 11:48:51.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5353cc7e7e354981991cb0399ebb618c\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/6067c1a38a0f4f3cb1c5a73f2c2ab2c1 : SUCCESS in 2m 13s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/15b6bb2378ff4068be510e547aba295d : SUCCESS in 1h 19m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/541bf3fc46c645dea4a8955ebfaf2a38 : SUCCESS in 3m 17s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5ea78fef77e6463c875b749a66159196 : SUCCESS in 3m 15s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/f0cea925251146a4969bf130011e32b8 : SUCCESS in 3m 27s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/53863e70fa364a0580b677478edbf6b0 : SUCCESS in 3m 10s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e6d2c69d39344113b8aed7468553591b : SUCCESS in 3m 16s","accounts_in_message":[],"_revision_number":3},{"id":"9ed100d7514bf4401b0f64dfd005b9d38d9d9cc6","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2023-04-10 02:11:25.000000000","message":"Patch Set 3: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"f0f077c6f0d1225515c62db2c7d5c8ce4e2ac735","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2023-05-17 09:19:32.000000000","message":"Uploaded patch set 4.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":4},{"id":"78d1668490668d8e6e065f75946a266c35fae223","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2023-05-17 09:19:56.000000000","message":"Patch Set 3:\n\n(2 comments)","accounts_in_message":[],"_revision_number":3},{"id":"828c918ea8ff6c0cdcdf47824e432ab6e357658e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-17 10:47:37.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/56e9aadd5eaa4bb5b9669d7f5167a134\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/5ac80630d1e44f5da76dd358150d46b7 : SUCCESS in 2m 14s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/0148827def9448079fd0494d84b29c3c : SUCCESS in 1h 27m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ab7a979573fd41439031ff989d3e7eca : SUCCESS in 3m 37s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/db75d773ff5745ae98ad38ef7d19f195 : SUCCESS in 2m 53s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/56d6cdb7f34c4b65a017f933b794e6d5 : SUCCESS in 4m 13s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/42ec8521759d4f8599bc7173b9317d77 : SUCCESS in 2m 55s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a268eb9a09234a7b9fbe04045db51089 : SUCCESS in 3m 17s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3a5d1a202658406b99af8ee82444608b : SUCCESS in 3m 30s","accounts_in_message":[],"_revision_number":4},{"id":"8622c668a6cab94f5d38461fefaf2545f0640a1f","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2023-06-27 12:04:41.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"cac751fe4498688cdb1e5ec5038e722f40138785","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-27 12:05:10.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"ac0b5372fd20ac5e54b4e1214aa9b95bd685c956","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-27 14:01:02.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3dfab0f4bec541ddb2ba39b01e44810f\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/b1ff58054d9e4cc5b9c45e14c612988f : SUCCESS in 2m 14s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/68b1383c33dd4fe8a5f27d7719c0ee9e : SUCCESS in 1h 48m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cb7126547e61402884efc0fea599f409 : SUCCESS in 3m 33s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/48be6a1faea94034b1a1bae7abaa4406 : SUCCESS in 4m 50s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b0fd2f2a77ea4cf2b5d914ee6d733ec6 : SUCCESS in 3m 30s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/909f6dd520e9429d9a376978e9297d49 : SUCCESS in 3m 27s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/52a6f0c634f54e7f8233bae583761b9a : SUCCESS in 3m 23s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e670bf119f664ae38389ac77fc14c910 : SUCCESS in 3m 07s","accounts_in_message":[],"_revision_number":4},{"id":"0975bc821271bb372c130bac486aec2d8f7686ce","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-27 14:01:02.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":4},{"id":"4e94c66b27af46a7689f80cb504c16bdb8eae7be","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-27 14:02:10.000000000","message":"Patch Set 4:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/794869c4168842bab0869eecffab681d\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f19d0ee031d641dba7b2623688b8de89 : SUCCESS in 42s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f23ac9329f66458b8fcd935c1ef4b8d9 : SUCCESS in 42s","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"3dff974eea1a509a63aeaccddbdc90b3695c90ef","revisions":{"db3cafba18a00b49108e5f12e8f730a5d7e9a86a":{"kind":"REWORK","_number":1,"created":"2023-03-10 09:23:40.000000000","uploader":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"ref":"refs/changes/84/877084/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.config","ref":"refs/changes/84/877084/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.config refs/changes/84/877084/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.config refs/changes/84/877084/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.config refs/changes/84/877084/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.config refs/changes/84/877084/1"}}},"commit":{"parents":[{"commit":"515daab81acccbea4015afc5c3c3316d88903a10","subject":"Fix issues related to tox4","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.config/commit/515daab81acccbea4015afc5c3c3316d88903a10"}]}],"author":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2023-03-10 09:21:35.000000000","tz":60},"committer":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2023-03-10 09:21:35.000000000","tz":60},"subject":"Bump bandit and make oslo.config compatible with latest rules","message":"Bump bandit and make oslo.config compatible with latest rules\n\nApply a timeout to requests calls to avoid uncontrolled\nresource consumption (CWE400) [1].\n\n[1] https://cwe.mitre.org/data/definitions/400.html\n\nChange-Id: I9e3c1e5f98e2ecfb5564d8dbd608d19d4a66cfce\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.config/commit/db3cafba18a00b49108e5f12e8f730a5d7e9a86a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.config/commit/db3cafba18a00b49108e5f12e8f730a5d7e9a86a"}]},"branch":"refs/heads/master"},"643f42f32c00dc3115c34c164f0c84a189953d05":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2023-03-10 09:24:49.000000000","uploader":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"ref":"refs/changes/84/877084/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.config","ref":"refs/changes/84/877084/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.config refs/changes/84/877084/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.config refs/changes/84/877084/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.config refs/changes/84/877084/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.config refs/changes/84/877084/2"}}},"commit":{"parents":[{"commit":"1654dcc91c5d5179c5b83d3c1ef5a2c18f655f4c","subject":"Update master for stable/2023.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.config/commit/1654dcc91c5d5179c5b83d3c1ef5a2c18f655f4c"}]}],"author":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2023-03-10 09:21:35.000000000","tz":60},"committer":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2023-03-10 09:24:38.000000000","tz":60},"subject":"Bump bandit and make oslo.config compatible with latest rules","message":"Bump bandit and make oslo.config compatible with latest rules\n\nApply a timeout to requests calls to avoid uncontrolled\nresource consumption (CWE400) [1].\n\n[1] https://cwe.mitre.org/data/definitions/400.html\n\nChange-Id: I9e3c1e5f98e2ecfb5564d8dbd608d19d4a66cfce\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.config/commit/643f42f32c00dc3115c34c164f0c84a189953d05"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.config/commit/643f42f32c00dc3115c34c164f0c84a189953d05"}]},"branch":"refs/heads/master"},"39d91cb6ee410528efcfd0757d5ca61e8dc6f123":{"kind":"REWORK","_number":3,"created":"2023-03-10 10:27:26.000000000","uploader":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"ref":"refs/changes/84/877084/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.config","ref":"refs/changes/84/877084/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.config refs/changes/84/877084/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.config refs/changes/84/877084/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.config refs/changes/84/877084/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.config refs/changes/84/877084/3"}}},"commit":{"parents":[{"commit":"1654dcc91c5d5179c5b83d3c1ef5a2c18f655f4c","subject":"Update master for stable/2023.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.config/commit/1654dcc91c5d5179c5b83d3c1ef5a2c18f655f4c"}]}],"author":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2023-03-10 09:21:35.000000000","tz":60},"committer":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2023-03-10 10:27:20.000000000","tz":60},"subject":"Bump bandit and make oslo.config compatible with latest rules","message":"Bump bandit and make oslo.config compatible with latest rules\n\nApply a timeout to requests calls to avoid uncontrolled\nresource consumption (CWE400) [1].\n\n[1] https://cwe.mitre.org/data/definitions/400.html\n\nChange-Id: I9e3c1e5f98e2ecfb5564d8dbd608d19d4a66cfce\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.config/commit/39d91cb6ee410528efcfd0757d5ca61e8dc6f123"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.config/commit/39d91cb6ee410528efcfd0757d5ca61e8dc6f123"}]},"branch":"refs/heads/master"},"3dff974eea1a509a63aeaccddbdc90b3695c90ef":{"kind":"REWORK","_number":4,"created":"2023-05-17 09:19:32.000000000","uploader":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"ref":"refs/changes/84/877084/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.config","ref":"refs/changes/84/877084/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.config refs/changes/84/877084/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.config refs/changes/84/877084/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.config refs/changes/84/877084/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.config refs/changes/84/877084/4"}}},"commit":{"parents":[{"commit":"1654dcc91c5d5179c5b83d3c1ef5a2c18f655f4c","subject":"Update master for stable/2023.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.config/commit/1654dcc91c5d5179c5b83d3c1ef5a2c18f655f4c"}]}],"author":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2023-03-10 09:21:35.000000000","tz":60},"committer":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2023-05-17 09:19:22.000000000","tz":120},"subject":"Bump bandit and make oslo.config compatible with latest rules","message":"Bump bandit and make oslo.config compatible with latest rules\n\nApply a timeout to requests calls to avoid uncontrolled\nresource consumption (CWE400) [1].\n\n[1] https://cwe.mitre.org/data/definitions/400.html\n\nChange-Id: I9e3c1e5f98e2ecfb5564d8dbd608d19d4a66cfce\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.config/commit/3dff974eea1a509a63aeaccddbdc90b3695c90ef"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.config/commit/3dff974eea1a509a63aeaccddbdc90b3695c90ef"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
