)]}'
{"id":"openstack%2Foslo.config~915396","triplet_id":"openstack%2Foslo.config~master~Iec977e21088b1ced9ad0df05638a459af8cdda2a","project":"openstack/oslo.config","branch":"master","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2024-06-25 10:24:26.000000000","reason":"Change was submitted"},"31245":{"account":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},"last_update":"2024-06-25 08:31:17.000000000","reason":"\u003cGERRIT_ACCOUNT_31245\u003e replied on the change","reason_account":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"}}},"hashtags":[],"change_id":"Iec977e21088b1ced9ad0df05638a459af8cdda2a","subject":"Dict: Support customizing key-value separator","status":"MERGED","created":"2024-04-10 06:24:12.000000000","updated":"2024-06-25 10:25:23.000000000","submitted":"2024-06-25 10:24:26.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"915396","meta_rev_id":"8a549e5b07218e12eb4f379bf5b90bd9cb91ec07","_number":915396,"virtual_id_number":915396,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-06-25 10:24:26.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2024-06-25 08:31:22.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2024-06-25 08:31:22.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-04-10 07:55:08.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-06-24 07:54:42.000000000","updated_by":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},"reviewer":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},"state":"REVIEWER"}],"messages":[{"id":"3761f07c605f5a15c54e4de6e114f215a2c15a58","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-04-10 06:24:12.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"63a5df38e70613441b7b9bbb34ffe75aefe5b4fc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-10 07:55:08.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/56e8e4abb062417b9b8b772b53b899df\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/dec04720dd804cd48c120fb97f4f3a4c : SUCCESS in 1h 25m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6cd8436881d14f33bd5877c7934f544c : SUCCESS in 3m 13s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/8adbd54f45fa41dcb8d4d7e4dbecc01b : SUCCESS in 3m 17s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/336bd89e438f48fa805d0ea7c0841b11 : SUCCESS in 3m 19s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/8fe8cb28e6e842978c9c58fd188814a1 : SUCCESS in 4m 43s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9a065849149241fe95ed91c48bc6e98f : SUCCESS in 2m 31s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3a8d0e6d83574fcea68c6a2df3b30327 : SUCCESS in 2m 32s","accounts_in_message":[],"_revision_number":1},{"id":"19615b0b8c1840e42de7d00e4cadd654731099f3","author":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},"date":"2024-06-24 07:54:42.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"4d6ab6feb8660c6283f1869480db840847b34d90","author":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},"date":"2024-06-24 12:58:12.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"c253bc0c545a3de7eb160ca2937a8f6bceb268c3","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-06-24 13:43:08.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"1c0a888a3a8fca27b0d2ed34a0186e4683c9139d","author":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},"date":"2024-06-25 08:31:17.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"c27e30e1968acea57faedee52fba19b30fa43b8a","author":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},"date":"2024-06-25 08:31:22.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"371345b5348ae2936ec7684ea58f6710b17c545d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-25 08:31:50.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"ac890bc990fd8b546c349525435cfbd7dcc4b753","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-25 10:24:26.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b8d42672ee86442b8e37bb19e47fb501\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/0df79e697c4144ec90fb30f943dc10fe : SUCCESS in 1h 34m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0feba205854c40fd874071bbf871fcbf : SUCCESS in 3m 19s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/3e282a2e5ecb40fc9015b661df54c6a0 : SUCCESS in 3m 48s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/a46885b62c54409984addcdbce308fa3 : SUCCESS in 4m 04s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/077bb29b449340478c1c9d1e53f0d5f0 : SUCCESS in 3m 16s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/4fc023dc6308492ba4a9cc2edfeb6505 : SUCCESS in 2m 31s","accounts_in_message":[],"_revision_number":1},{"id":"b6c4ffdd4a14313d75e969431f66d3af80d7259e","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-25 10:24:26.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"8a549e5b07218e12eb4f379bf5b90bd9cb91ec07","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-25 10:25:23.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f786bede8c7242abb62e6822e21f599d\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c48d91dea4a44b3293cc07d88a4ffd96 : SUCCESS in 39s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e551f0fe78dd443fba974f4e0320dac6 : SUCCESS in 38s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"d193c657441f24506f75b9ce1248c8fdebd295f3","revisions":{"d193c657441f24506f75b9ce1248c8fdebd295f3":{"kind":"REWORK","_number":1,"created":"2024-04-10 06:24:12.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/96/915396/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.config","ref":"refs/changes/96/915396/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.config refs/changes/96/915396/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.config refs/changes/96/915396/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.config refs/changes/96/915396/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.config refs/changes/96/915396/1"}}},"commit":{"parents":[{"commit":"c290dd7fcf9142df49e7375d127e5845060ae6c7","subject":"Merge \"doc: Drop oslo.db.concurrency namespace\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.config/commit/c290dd7fcf9142df49e7375d127e5845060ae6c7"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-04-10 06:21:29.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-04-10 06:21:29.000000000","tz":540},"subject":"Dict: Support customizing key-value separator","message":"Dict: Support customizing key-value separator\n\nSome projects implements their own mechanism to parse dict values from\nlist because they use different separators between key and value.\n\nFor example, nova provides the [libvirt] disk_cachemode option, and\nthis option takes a value written in the following format.\n\n[libvirt]\ndisk_cachemode\u003dkey1\u003dvalue1,key2\u003dvalue2\n\nIntroduce the new key_value_separator parameter to allow using\na different separator, so that DictOpt can be used for such usage.\n\nChange-Id: Iec977e21088b1ced9ad0df05638a459af8cdda2a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.config/commit/d193c657441f24506f75b9ce1248c8fdebd295f3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.config/commit/d193c657441f24506f75b9ce1248c8fdebd295f3"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
