)]}'
{"id":"openstack%2Foslo.config~974557","triplet_id":"openstack%2Foslo.config~master~I7993043e36ad527887d0a50caea50664d95901e2","project":"openstack/oslo.config","branch":"master","topic":"bug/2138960","attention_set":{},"removed_from_attention_set":{"36966":{"account":{"_account_id":36966,"name":"Myles Penner","display_name":"Myles Penner","email":"mj.penner@live.ca","username":"mylesjp"},"last_update":"2026-01-26 13:46:36.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I7993043e36ad527887d0a50caea50664d95901e2","subject":"Adapt subcommand test for Python 3.14","status":"MERGED","created":"2026-01-23 19:45:24.000000000","updated":"2026-01-26 13:48:48.000000000","submitted":"2026-01-26 13:46:36.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"974557-bug/2138960","meta_rev_id":"278cdcccf6401bcde936ec77107a9a76388fd598","_number":974557,"virtual_id_number":974557,"owner":{"_account_id":36966,"name":"Myles Penner","display_name":"Myles Penner","email":"mj.penner@live.ca","username":"mylesjp"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-01-26 13:46:36.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":2,"date":"2026-01-26 12:44:48.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":1,"date":"2026-01-26 12:44:48.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-01-23 21:25:13.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-01-26 12:44:48.000000000","updated_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"reviewer":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"state":"REVIEWER"}],"messages":[{"id":"494c3a5af6e725a140fb3e2db519d2589f1ac10f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36966,"name":"Myles Penner","display_name":"Myles Penner","email":"mj.penner@live.ca","username":"mylesjp"},"date":"2026-01-23 19:45:24.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"c2b223392038e91a563e333e03048c895392a9ef","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-23 21:25:13.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e5f312e4a7214c0eab7604e3f9859a32\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/2a58115a5dc84b1587a64b5057f326d6 : SUCCESS in 1h 37m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8463b1a1318d4ec4839de5415abefd08 : SUCCESS in 3m 22s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/eedde19f85fc46aaa4ff186d8ee3eab5 : SUCCESS in 2m 28s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/6f98966756bc43c1b1aa753c65fcc8f8 : SUCCESS in 3m 34s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5f22f8e22e5b4bdfa3747065d9ae0d73 : SUCCESS in 2m 53s","accounts_in_message":[],"_revision_number":1},{"id":"7a5903d5247856273faf790229e58ee36f80d9fc","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2026-01-26 12:44:48.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"db50da469f9de2175d9008ed16cc80f27d23cb47","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-26 12:45:08.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"4a430b70ff0232e9f0b157d0b7ae52fb8ddcfda2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-26 13:46:36.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9ba67c0d35fa46439bdcc3b43e013b88\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/46cb6149fdc347269775afc833a1af2b : SUCCESS in 1h 00m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/24e0a0d3135e4c5a80adbf47d12a7a7d : SUCCESS in 3m 24s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/892f56c576bc483a9d21044f43b15403 : SUCCESS in 3m 14s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/380a9416e05b403dab1dac18cb3d0319 : SUCCESS in 3m 44s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0eb076370cbb4627ab22590efda12a44 : SUCCESS in 2m 43s","accounts_in_message":[],"_revision_number":1},{"id":"c5233b1aa30709ce8b1d8124114fc0f0d76bd7f0","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-26 13:46:36.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"278cdcccf6401bcde936ec77107a9a76388fd598","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-26 13:48:48.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3e5b9b32c41247e3a3a56b1b24faece2\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1d4efd3bb581451ba4fe364bd3ecec47 : SUCCESS in 48s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"1e3e45c6b035993593d718235f3f2d44f9a8bbe5","revisions":{"1e3e45c6b035993593d718235f3f2d44f9a8bbe5":{"kind":"REWORK","_number":1,"created":"2026-01-23 19:45:24.000000000","uploader":{"_account_id":36966,"name":"Myles Penner","display_name":"Myles Penner","email":"mj.penner@live.ca","username":"mylesjp"},"ref":"refs/changes/57/974557/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.config","ref":"refs/changes/57/974557/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.config refs/changes/57/974557/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.config refs/changes/57/974557/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.config refs/changes/57/974557/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.config refs/changes/57/974557/1"}}},"commit":{"parents":[{"commit":"ba58fbd74115a596afdb2eb2853042e331d5d98e","subject":"Add min_length parameter","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.config/commit/ba58fbd74115a596afdb2eb2853042e331d5d98e"}]}],"author":{"name":"Myles Penner","email":"myles.penner@canonical.com","date":"2026-01-23 19:39:41.000000000","tz":-480},"committer":{"name":"Myles Penner","email":"myles.penner@canonical.com","date":"2026-01-23 19:45:22.000000000","tz":-480},"subject":"Adapt subcommand test for Python 3.14","message":"Adapt subcommand test for Python 3.14\n\nThe `test_sub_command_multiple` test fails on Python 3.14 because\n`argparse` now raises a `ValueError` when multiple subparsers are\nadded, whereas previous versions raised `argparse.ArgumentError`\nor `SystemExit`.\n\nUpdate the test to expect `ValueError` on Python 3.14 and newer,\nretaining the existing logic for older Python releases.\n\nChange-Id: I7993043e36ad527887d0a50caea50664d95901e2\nSigned-off-by: Myles Penner \u003cmyles.penner@canonical.com\u003e\nCloses-bug: 2138960\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.config/commit/1e3e45c6b035993593d718235f3f2d44f9a8bbe5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.config/commit/1e3e45c6b035993593d718235f3f2d44f9a8bbe5"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
