)]}'
{"id":"openstack%2Foslo.config~979495","triplet_id":"openstack%2Foslo.config~master~Ie02589adae98773f9896d2529a4fb261c5a201e6","project":"openstack/oslo.config","branch":"master","topic":"linter-sync","attention_set":{},"removed_from_attention_set":{"15334":{"account":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"last_update":"2026-03-09 14:13:53.000000000","reason":"removed on reply"}},"hashtags":[],"change_id":"Ie02589adae98773f9896d2529a4fb261c5a201e6","subject":"Update packaging configuration","status":"MERGED","created":"2026-03-09 11:32:25.000000000","updated":"2026-05-02 15:54:34.000000000","submitted":"2026-05-02 15:53:47.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"979495-linter-sync","meta_rev_id":"b27e93e83e86a345959df041272dad48f9702557","_number":979495,"virtual_id_number":979495,"owner":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-05-02 15:53:47.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":2,"date":"2026-05-02 12:27:05.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":1,"date":"2026-05-02 12:27:05.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-03-09 14:00:39.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"46102c3384ceef113c878af73737a12cd9df20da","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2026-03-09 11:32:25.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"4ff2df26fea64a041a2ffb1326801469f1be7178","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2026-03-09 11:35:51.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"1a68ed6371b08f3930bacbb96ffab73a0986df5a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-09 14:00:39.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/6bd0fd46a68f48b5b4c82ef67f44d2ff\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/83f8c34740bd437f94dba2dbc9b988c1 : FAILURE in 2m 30s\n- requirements-check https://zuul.opendev.org/t/openstack/build/588a7c621a034ec1b3f8cbe8a97a5bee : SUCCESS in 2m 27s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/37372b0a64d6481caa0d45e86ede68a1 : SUCCESS in 2h 01m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/635ad5ebd9e44b5c82ff56751245ffa5 : SUCCESS in 3m 24s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/19daadf19b5a49bda882a926071dbe94 : SUCCESS in 4m 12s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/7d3a888b85c84b9d886372fc67db85ab : SUCCESS in 5m 22s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/059d69a1f9eb4e6a9e22607bb13bdb1d : SUCCESS in 6m 20s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2afebcf77e04445c99ac238b1b1ebf62 : SUCCESS in 3m 54s","accounts_in_message":[],"_revision_number":2},{"id":"0aa713fb3988752706fae246fd7b5eb386c57f43","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2026-03-09 14:13:53.000000000","message":"Patch Set 2: Workflow-1","accounts_in_message":[],"_revision_number":2},{"id":"961725cd5e97c88bf1935cb2d8fa8cd58c7792d2","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2026-05-02 12:27:05.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"33d4ea7d1cf96c185500ab848050ee03a412b9a6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-02 14:12:03.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0db8c3bd1d6e4d9086ff83a41ed4d6fb\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/f845a28228bd470c96a45560d88d1409 : SUCCESS in 2m 55s\n- requirements-check https://zuul.opendev.org/t/openstack/build/931c820052c04274918fb04d5e14c6a9 : SUCCESS in 2m 07s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/d8d22786ea8f495ba4a25e7ad0595155 : SUCCESS in 1h 39m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a74d802092be418aa781f1e5ef8196db : SUCCESS in 3m 48s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/8818b2c9910040db9d6dfb628e2f10a2 : SUCCESS in 3m 06s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/8dfa1fc41d914f7187628e0509020f82 : SUCCESS in 3m 09s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/a57f3eecc42047e785d43b96a1654216 : SUCCESS in 6m 29s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/eb800f1ae1184955886eeb12e44dca1e : SUCCESS in 3m 31s","accounts_in_message":[],"_revision_number":2},{"id":"626dc82f6009f05032a57048ef189ee074ddd2d4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-02 14:12:28.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"ef6f8a189c32e698903f20b902bb55e8d87ca37b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-02 15:53:47.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ec8a6de613404c4ca0e2dfcefeadd049\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/c2eea0cb878c4f0fa39a4bbbb5d3a2d0 : SUCCESS in 2m 25s\n- requirements-check https://zuul.opendev.org/t/openstack/build/3d1f9d79f45f4cc882f1b8f09e680900 : SUCCESS in 2m 03s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/0618edf50a984eb4a7bf744aaaf031a7 : SUCCESS in 1h 36m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fa7fb89365be44f3893f414ae2a8d0b3 : SUCCESS in 2m 06s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/4cb4dae1538644a5863db88b4b537e7c : SUCCESS in 3m 07s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/19c2ab79df3546fa80eb1b73991b73d3 : SUCCESS in 3m 21s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e9a998bfdc814001b55c01994b004176 : SUCCESS in 4m 33s","accounts_in_message":[],"_revision_number":2},{"id":"4b979229635788941173a3b4513797e40a106aba","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-02 15:53:47.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"b27e93e83e86a345959df041272dad48f9702557","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-02 15:54:34.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1df112df2f774f72a9a66667c04ac385\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/78f7a68bdf35473496425280d9eb0d29 : SUCCESS in 37s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"5c7349f26323ba28efbf512466b15cd6eabfcf1e","revisions":{"c0fa326391ce89eb22b40c9a17784390ddc5d81f":{"kind":"REWORK","_number":1,"created":"2026-03-09 11:32:25.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/95/979495/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.config","ref":"refs/changes/95/979495/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.config refs/changes/95/979495/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.config refs/changes/95/979495/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.config refs/changes/95/979495/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.config refs/changes/95/979495/1"}}},"commit":{"parents":[{"commit":"b2a5182bf5489be677a65637e191c98d9690e695","subject":"ruff: Configure hacking as external linter","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.config/commit/b2a5182bf5489be677a65637e191c98d9690e695"}]}],"author":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2026-03-09 11:32:19.000000000","tz":0},"committer":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2026-03-09 11:32:19.000000000","tz":0},"subject":"Update packaging configuration","message":"Update packaging configuration\n\nResolve setuptools deprecation warnings around license configuration.\n\nChange-Id: Ie02589adae98773f9896d2529a4fb261c5a201e6\nSigned-off-by: Stephen Finucane \u003csfinucan@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.config/commit/c0fa326391ce89eb22b40c9a17784390ddc5d81f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.config/commit/c0fa326391ce89eb22b40c9a17784390ddc5d81f"}]},"branch":"refs/heads/master"},"5c7349f26323ba28efbf512466b15cd6eabfcf1e":{"kind":"REWORK","_number":2,"created":"2026-03-09 11:35:51.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/95/979495/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.config","ref":"refs/changes/95/979495/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.config refs/changes/95/979495/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.config refs/changes/95/979495/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.config refs/changes/95/979495/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.config refs/changes/95/979495/2"}}},"commit":{"parents":[{"commit":"b2a5182bf5489be677a65637e191c98d9690e695","subject":"ruff: Configure hacking as external linter","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.config/commit/b2a5182bf5489be677a65637e191c98d9690e695"}]}],"author":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2026-03-09 11:32:19.000000000","tz":0},"committer":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2026-03-09 11:35:30.000000000","tz":0},"subject":"Update packaging configuration","message":"Update packaging configuration\n\nEnable package discovery and resolve setuptools deprecation warnings\naround license configuration.\n\nChange-Id: Ie02589adae98773f9896d2529a4fb261c5a201e6\nSigned-off-by: Stephen Finucane \u003csfinucan@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.config/commit/5c7349f26323ba28efbf512466b15cd6eabfcf1e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.config/commit/5c7349f26323ba28efbf512466b15cd6eabfcf1e"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
