)]}'
{"id":"openstack%2Foslo.config~990585","triplet_id":"openstack%2Foslo.config~master~Iad15176ec4d8c50604227c8e51f9db9a07f5aec8","project":"openstack/oslo.config","branch":"master","attention_set":{},"removed_from_attention_set":{"5890":{"account":{"_account_id":5890,"name":"Doug Goldstein","email":"cardoe@cardoe.com","username":"cardoe"},"last_update":"2026-06-02 11:38:51.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Iad15176ec4d8c50604227c8e51f9db9a07f5aec8","subject":"fix: setuptools.package.find is a glob","status":"MERGED","created":"2026-05-29 02:21:07.000000000","updated":"2026-06-02 11:39:57.000000000","submitted":"2026-06-02 11:38:51.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"990585","meta_rev_id":"6795644ddcfc76726fc18645e16dec4c194b87e6","_number":990585,"virtual_id_number":990585,"owner":{"_account_id":5890,"name":"Doug Goldstein","email":"cardoe@cardoe.com","username":"cardoe"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-06-02 11:38:50.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":2,"date":"2026-06-02 09:43:27.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":1,"date":"2026-06-02 09:43:27.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-05-29 03:56:23.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-05-30 23:56:13.000000000","updated_by":{"_account_id":5890,"name":"Doug Goldstein","email":"cardoe@cardoe.com","username":"cardoe"},"reviewer":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"state":"CC"},{"updated":"2026-06-02 09:43:27.000000000","updated_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"reviewer":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"state":"REVIEWER"}],"messages":[{"id":"5f07e7733a67dcd6e87374cac5ddaf08b5d8f7bc","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":5890,"name":"Doug Goldstein","email":"cardoe@cardoe.com","username":"cardoe"},"date":"2026-05-29 02:21:07.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"9b28f7ce66428ef43e9bbfae1815d9ab3c57248e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-29 03:56:23.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/61b8ae4ffd0249eca156700af08581cc\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/fcbb74f37e984ebea351c7b48bba7bd0 : SUCCESS in 2m 28s\n- requirements-check https://zuul.opendev.org/t/openstack/build/b4bd9e654bce48edb4481e1e6d56b34c : SUCCESS in 2m 26s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/11e2b943d8be4f65a48508be73d7e2a4 : SUCCESS in 1h 06m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/68932a2996f143fc9b6e61702e37537d : SUCCESS in 4m 41s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/9b78e4c981184405922e39ce0926292c : SUCCESS in 3m 54s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/be4cc1a84ff24733bf615ac1f8d10e4c : SUCCESS in 3m 23s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/2d038d2ef0514fa4b28ac5ea07046c11 : SUCCESS in 5m 23s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/152fe0c943e2495a8fe6ec05645a41da : SUCCESS in 3m 57s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f4281ef4d42e42779678730c0753a9dc : SUCCESS in 3m 25s","accounts_in_message":[],"_revision_number":1},{"id":"2c67eadd29dd98d44f0d5121fc4d7a5265571511","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2026-06-02 09:43:27.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"cb8fbd8ef7532e8d9edc16c175776bbe1fef65d0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-02 09:43:53.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"2e372cd540bf6390046a35a7c7174e5fcd13bb65","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-02 11:38:50.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0a27e31ebe7d4e90865c4881fb4ab78c\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/9c1a9336f7d64da28e194bc2c08e341a : SUCCESS in 4m 39s\n- requirements-check https://zuul.opendev.org/t/openstack/build/80e697b2814049a5ae4f7b70cb7c26b2 : SUCCESS in 3m 08s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/60eb9f43d98c455281c8abc742e58106 : SUCCESS in 1h 49m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b587d8c490934b18ba0c1922cec595e2 : SUCCESS in 6m 27s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/f01e4eb82cc3474d8e9481205cc539b9 : SUCCESS in 3m 46s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/08b7def59de240608e1705d599098d7b : SUCCESS in 3m 50s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1dc380f8e0b04bbb9646b26cabe29c6b : SUCCESS in 3m 26s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ffce800542564c4d91d4a0c54c62a2cb : SUCCESS in 3m 55s","accounts_in_message":[],"_revision_number":1},{"id":"9324fa2c80ef3f5b5c9211f6ce33644d9f07cdc7","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-02 11:38:51.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"6795644ddcfc76726fc18645e16dec4c194b87e6","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-02 11:39:57.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3d51d5fdcad14f9aa83cdd6ada671838\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/af4a50b511064923bc8137fe69b7e325 : SUCCESS in 50s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"aa5f161db19f64f61dd8f9e67c93deb7537b5ebe","revisions":{"aa5f161db19f64f61dd8f9e67c93deb7537b5ebe":{"kind":"REWORK","_number":1,"created":"2026-05-29 02:21:07.000000000","uploader":{"_account_id":5890,"name":"Doug Goldstein","email":"cardoe@cardoe.com","username":"cardoe"},"ref":"refs/changes/85/990585/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.config","ref":"refs/changes/85/990585/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.config refs/changes/85/990585/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.config refs/changes/85/990585/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.config refs/changes/85/990585/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.config refs/changes/85/990585/1"}}},"commit":{"parents":[{"commit":"8c9d80325454a6d3c2f2f79518552b18c22b3d55","subject":"Merge \"Drop Python 3.10\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.config/commit/8c9d80325454a6d3c2f2f79518552b18c22b3d55"}]}],"author":{"name":"Doug Goldstein","email":"cardoe@cardoe.com","date":"2026-05-29 02:20:03.000000000","tz":-420},"committer":{"name":"Doug Goldstein","email":"cardoe@cardoe.com","date":"2026-05-29 02:21:01.000000000","tz":-420},"subject":"fix: setuptools.package.find is a glob","message":"fix: setuptools.package.find is a glob\n\nCorrect the usage of setuptools.package.find which takes a glob for the\ninclude path.\n\nChange-Id: Iad15176ec4d8c50604227c8e51f9db9a07f5aec8\nSigned-off-by: Doug Goldstein \u003ccardoe@cardoe.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.config/commit/aa5f161db19f64f61dd8f9e67c93deb7537b5ebe"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.config/commit/aa5f161db19f64f61dd8f9e67c93deb7537b5ebe"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
