)]}'
{"id":"openstack%2Foslo.context~1006300","triplet_id":"openstack%2Foslo.context~master~Ia6c366fc6c36874dbe2bec8b8cf882b9d303cd4b","project":"openstack/oslo.context","branch":"master","attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2026-09-18 19:14:12.000000000","reason":"A robot voted negatively on a label"}},"removed_from_attention_set":{},"hashtags":[],"change_id":"Ia6c366fc6c36874dbe2bec8b8cf882b9d303cd4b","subject":"Use built-in typing.Self","status":"NEW","created":"2026-09-18 17:37:59.000000000","updated":"2026-09-18 22:53:42.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":true,"submittable":false,"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"2d2c4d83e15247b5b0198ebb1e1c0089186dbf59","_number":1006300,"virtual_id_number":1006300,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2026-09-18 22:53:42.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-09-18 19:14:12.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"d301aae881eea3dcd3f217aa48f6a8d24d2484d3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-09-18 17:37:59.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"270d0f19e6211e87b4ac334efbfd2ea5c51839c4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-09-18 17:39:44.000000000","message":"Patch Set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"c96d2d6908bcdb061e1a3998a9cc252e62aa14d3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-18 19:14:12.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/ba4a21e07a414e078d85bb1993e8bbd6\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/4cf36012a9ed4b729736e68a7c974c83 : SUCCESS in 2m 04s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/ba33e32b159d4e19abaa1cb2ba63034d : SUCCESS in 1h 23m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8f18ef1a09b64e84b56d0b640ffc8a5b : FAILURE in 3m 25s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/aff30c2dcd7c47099ed9d01eef1170db : SUCCESS in 2m 22s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/1f763851c0ae40e88432df6e1e4e3637 : SUCCESS in 2m 40s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/8dcb05fbebd34372a3a17694c83af117 : SUCCESS in 4m 28s\n- openstack-tox-py315 https://zuul.opendev.org/t/openstack/build/83665ad10d1c4f3eb6cccc858b90c3b5 : SUCCESS in 4m 30s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/10351f9bbb5245f4a549fc269a1e83f2 : SUCCESS in 2m 32s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/4e4568ae366e4fba96605115b8bf6fdd : SUCCESS in 2m 45s","accounts_in_message":[],"_revision_number":2},{"id":"9ad4c155eeb0affa7a8a903154299d6743b09bb8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-09-18 21:54:33.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":3},{"id":"2d2c4d83e15247b5b0198ebb1e1c0089186dbf59","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-18 22:53:42.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1999758cfa7948ed8d63ee11a3d9a35a\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/3944c4a3537a4ad8b8d51e359e566bab : SUCCESS in 2m 12s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/579bb97ca60f4dd38311791f4892295d : SUCCESS in 57m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f013ef9478ae4b44bebe1ebad51a28b4 : SUCCESS in 3m 34s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/aa18bf520eb641caabd3da99f82bcaaa : SUCCESS in 2m 44s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/472b75e1e0704b058a1ae4bde8097404 : SUCCESS in 4m 47s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/12b883ac97b5428d83a4325cd7e98bc0 : SUCCESS in 5m 42s\n- openstack-tox-py315 https://zuul.opendev.org/t/openstack/build/e3be83289c8e473e94c76143b3e24359 : SUCCESS in 5m 51s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b6b914d22a99440fb1f5198881cf2bff : SUCCESS in 3m 12s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e61e32ae00e746bdb1ad905838d5777b : SUCCESS in 4m 37s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"2b27ac9b860cc41cb396dabd2f9f3e196cb6b609","revisions":{"8fc252316e0ac43628977cfba75f0c471e69d2bc":{"kind":"REWORK","_number":1,"created":"2026-09-18 17:37:59.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/00/1006300/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.context","ref":"refs/changes/00/1006300/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.context refs/changes/00/1006300/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.context refs/changes/00/1006300/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.context refs/changes/00/1006300/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.context refs/changes/00/1006300/1"}}},"commit":{"parents":[{"commit":"af3126242027af83c6a0e2a8fd6b9859d25840c4","subject":"Merge \"Ensure recursive detection of package files\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.context/commit/af3126242027af83c6a0e2a8fd6b9859d25840c4"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-09-18 17:37:36.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-09-18 17:37:36.000000000","tz":540},"subject":"Use built-in typing.Self","message":"Use built-in typing.Self\n\nIt has been available since Python 3.11 .\n\nChange-Id: Ia6c366fc6c36874dbe2bec8b8cf882b9d303cd4b\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.context/commit/8fc252316e0ac43628977cfba75f0c471e69d2bc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.context/commit/8fc252316e0ac43628977cfba75f0c471e69d2bc"}]},"branch":"refs/heads/master"},"3a5c21bf81aa02bed42c5109d5ec7e44865be737":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2026-09-18 17:39:44.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/00/1006300/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.context","ref":"refs/changes/00/1006300/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.context refs/changes/00/1006300/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.context refs/changes/00/1006300/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.context refs/changes/00/1006300/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.context refs/changes/00/1006300/2"}}},"commit":{"parents":[{"commit":"af3126242027af83c6a0e2a8fd6b9859d25840c4","subject":"Merge \"Ensure recursive detection of package files\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.context/commit/af3126242027af83c6a0e2a8fd6b9859d25840c4"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-09-18 17:37:36.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-09-18 17:39:44.000000000","tz":0},"subject":"Use built-in typing.Self","message":"Use built-in typing.Self\n\nIt has been available since Python 3.11 support was removed.\n\nChange-Id: Ia6c366fc6c36874dbe2bec8b8cf882b9d303cd4b\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.context/commit/3a5c21bf81aa02bed42c5109d5ec7e44865be737"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.context/commit/3a5c21bf81aa02bed42c5109d5ec7e44865be737"}]},"branch":"refs/heads/master","description":"Edit commit message"},"2b27ac9b860cc41cb396dabd2f9f3e196cb6b609":{"kind":"REWORK","_number":3,"created":"2026-09-18 21:54:33.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/00/1006300/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.context","ref":"refs/changes/00/1006300/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.context refs/changes/00/1006300/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.context refs/changes/00/1006300/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.context refs/changes/00/1006300/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.context refs/changes/00/1006300/3"}}},"commit":{"parents":[{"commit":"af3126242027af83c6a0e2a8fd6b9859d25840c4","subject":"Merge \"Ensure recursive detection of package files\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.context/commit/af3126242027af83c6a0e2a8fd6b9859d25840c4"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-09-18 17:37:36.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-09-18 21:54:25.000000000","tz":540},"subject":"Use built-in typing.Self","message":"Use built-in typing.Self\n\nIt has been available since Python 3.11 .\n\nChange-Id: Ia6c366fc6c36874dbe2bec8b8cf882b9d303cd4b\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.context/commit/2b27ac9b860cc41cb396dabd2f9f3e196cb6b609"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.context/commit/2b27ac9b860cc41cb396dabd2f9f3e196cb6b609"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY"},{"label":"Workflow","status":"MAY"},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
