)]}'
{"id":"openstack%2Foslo.context~135641","triplet_id":"openstack%2Foslo.context~master~I548120770d0f11a2af56f07d0902433a8c7c98fe","project":"openstack/oslo.context","branch":"master","topic":"bp/graduate-oslo-context","hashtags":[],"change_id":"I548120770d0f11a2af56f07d0902433a8c7c98fe","subject":"Change instance_uuid to resource_uuid","status":"MERGED","created":"2014-11-19 16:32:15.000000000","updated":"2014-11-21 12:44:22.000000000","submitted":"2014-11-21 12:44:22.000000000","submitter":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"697350bcc1cdde6de243070cfbc94da2bd3f7c63","_number":135641,"virtual_id_number":135641,"owner":{"_account_id":2472,"name":"Doug Hellmann","email":"dhellmann@redhat.com","username":"doug-hellmann"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"all":[{"value":2,"date":"2014-11-21 12:44:22.000000000","_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":0,"_account_id":5638,"name":"Davanum Srinivas","email":"davanum@gmail.com","username":"dims-v"},{"value":0,"_account_id":6159,"name":"Flavio Percoco Premoli","display_name":"flaper87","email":"flavio.percoco@flyrlabs.com","username":"flaper87"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":5638,"name":"Davanum Srinivas","email":"davanum@gmail.com","username":"dims-v"},"all":[{"value":0,"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":2,"date":"2014-11-19 16:57:39.000000000","_account_id":5638,"name":"Davanum Srinivas","email":"davanum@gmail.com","username":"dims-v"},{"value":2,"date":"2014-11-21 11:37:39.000000000","_account_id":6159,"name":"Flavio Percoco Premoli","display_name":"flaper87","email":"flavio.percoco@flyrlabs.com","username":"flaper87"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":6159,"name":"Flavio Percoco Premoli","display_name":"flaper87","email":"flavio.percoco@flyrlabs.com","username":"flaper87"},"all":[{"value":0,"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":0,"_account_id":5638,"name":"Davanum Srinivas","email":"davanum@gmail.com","username":"dims-v"},{"value":1,"date":"2014-11-21 11:37:39.000000000","_account_id":6159,"name":"Flavio Percoco Premoli","display_name":"flaper87","email":"flavio.percoco@flyrlabs.com","username":"flaper87"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":0,"_account_id":5638,"name":"Davanum Srinivas","email":"davanum@gmail.com","username":"dims-v"},{"value":0,"_account_id":6159,"name":"Flavio Percoco Premoli","display_name":"flaper87","email":"flavio.percoco@flyrlabs.com","username":"flaper87"}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":3,"name":"Jenkins","username":"jenkins"},{"_account_id":5638,"name":"Davanum Srinivas","email":"davanum@gmail.com","username":"dims-v"},{"_account_id":6159,"name":"Flavio Percoco Premoli","display_name":"flaper87","email":"flavio.percoco@flyrlabs.com","username":"flaper87"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2014-11-19 16:57:39.000000000","updated_by":{"_account_id":5638,"name":"Davanum Srinivas","email":"davanum@gmail.com","username":"dims-v"},"reviewer":{"_account_id":5638,"name":"Davanum Srinivas","email":"davanum@gmail.com","username":"dims-v"},"state":"REVIEWER"},{"updated":"2014-11-21 11:37:39.000000000","updated_by":{"_account_id":6159,"name":"Flavio Percoco Premoli","display_name":"flaper87","email":"flavio.percoco@flyrlabs.com","username":"flaper87"},"reviewer":{"_account_id":6159,"name":"Flavio Percoco Premoli","display_name":"flaper87","email":"flavio.percoco@flyrlabs.com","username":"flaper87"},"state":"REVIEWER"},{"updated":"2014-11-21 12:44:22.000000000","updated_by":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"reviewer":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"state":"REVIEWER"}],"messages":[{"id":"6c1832b0743e5c6c05f24eb92efd7c90c72a7c50","author":{"_account_id":2472,"name":"Doug Hellmann","email":"dhellmann@redhat.com","username":"doug-hellmann"},"date":"2014-11-19 16:32:15.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"bc1011806b439280a2e907f235b9e53dca9f6bf7","author":{"_account_id":5638,"name":"Davanum Srinivas","email":"davanum@gmail.com","username":"dims-v"},"date":"2014-11-19 16:57:39.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"ea3b17e0531e84290e65e92811a2bbb68bbbbdee","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2014-11-19 17:33:15.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- gate-oslo.context-pep8 http://logs.openstack.org/41/135641/1/check/gate-oslo.context-pep8/0db7ae9 : SUCCESS in 19s\n- gate-oslo.context-docs http://docs-draft.openstack.org/41/135641/1/check/gate-oslo.context-docs/19b539a/doc/build/html/ : SUCCESS in 29s\n- gate-oslo.context-python27 http://logs.openstack.org/41/135641/1/check/gate-oslo.context-python27/18ba667 : SUCCESS in 28s\n- gate-oslo.context-python33 http://logs.openstack.org/41/135641/1/check/gate-oslo.context-python33/3ef8aad : SUCCESS in 1m 31s\n- gate-tempest-dsvm-src-oslo.context http://logs.openstack.org/41/135641/1/check/gate-tempest-dsvm-src-oslo.context/17c4074 : SUCCESS in 56m 33s\n- gate-tempest-dsvm-neutron-src-oslo.context http://logs.openstack.org/41/135641/1/check/gate-tempest-dsvm-neutron-src-oslo.context/9abe15f : SUCCESS in 58m 40s\n- gate-oslo.context-python26 http://logs.openstack.org/41/135641/1/check/gate-oslo.context-python26/d91d80a : SUCCESS in 26s","accounts_in_message":[],"_revision_number":1},{"id":"79089563899362040db2ec104b8a49c0badd4617","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2014-11-21 11:37:39.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.\nhttp://status.openstack.org/zuul/","accounts_in_message":[],"_revision_number":1},{"id":"caae4015024fc15c80c6f3af4b666795992aa27c","author":{"_account_id":6159,"name":"Flavio Percoco Premoli","display_name":"flaper87","email":"flavio.percoco@flyrlabs.com","username":"flaper87"},"date":"2014-11-21 11:37:39.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"b50acf3b673da8030981f6ce75e3d1e04ce479cf","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2014-11-21 12:44:22.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- gate-oslo.context-docs http://docs-draft.openstack.org/41/135641/1/gate/gate-oslo.context-docs/02ce676/doc/build/html/ : SUCCESS in 49s\n- gate-oslo.context-pep8 http://logs.openstack.org/41/135641/1/gate/gate-oslo.context-pep8/af1aa62 : SUCCESS in 24s\n- gate-oslo.context-python27 http://logs.openstack.org/41/135641/1/gate/gate-oslo.context-python27/ecc116a : SUCCESS in 34s\n- gate-oslo.context-python33 http://logs.openstack.org/41/135641/1/gate/gate-oslo.context-python33/f562cd6 : SUCCESS in 1m 30s\n- gate-tempest-dsvm-src-oslo.context http://logs.openstack.org/41/135641/1/gate/gate-tempest-dsvm-src-oslo.context/3a23594 : SUCCESS in 43m 09s\n- gate-tempest-dsvm-neutron-src-oslo.context http://logs.openstack.org/41/135641/1/gate/gate-tempest-dsvm-neutron-src-oslo.context/db15018 : SUCCESS in 1h 06m 41s\n- gate-oslo.context-python26 http://logs.openstack.org/41/135641/1/gate/gate-oslo.context-python26/e26ad63 : SUCCESS in 29s","accounts_in_message":[],"_revision_number":1},{"id":"33cf24060fc90c83094a4914664f812844f7c23e","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2014-11-21 12:44:22.000000000","message":"Change has been successfully merged into the git repository.","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"cac31d9c6f0dd11cde816947abc2f2a7f74932be","revisions":{"cac31d9c6f0dd11cde816947abc2f2a7f74932be":{"kind":"REWORK","_number":1,"created":"2014-11-19 16:32:15.000000000","uploader":{"_account_id":2472,"name":"Doug Hellmann","email":"dhellmann@redhat.com","username":"doug-hellmann"},"ref":"refs/changes/41/135641/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.context","ref":"refs/changes/41/135641/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.context refs/changes/41/135641/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.context refs/changes/41/135641/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.context refs/changes/41/135641/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.context refs/changes/41/135641/1"}}},"commit":{"parents":[{"commit":"384c0de994102a5fbe8d8da1b37a3f1890ee574d","subject":"Better information in the README","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.context/commit/384c0de994102a5fbe8d8da1b37a3f1890ee574d"}]}],"author":{"name":"Doug Hellmann","email":"doug@doughellmann.com","date":"2014-11-19 16:04:52.000000000","tz":-300},"committer":{"name":"Doug Hellmann","email":"doug@doughellmann.com","date":"2014-11-19 16:04:52.000000000","tz":-300},"subject":"Change instance_uuid to resource_uuid","message":"Change instance_uuid to resource_uuid\n\nUse an application-agnostic term for the resource the context is\nattached to.\n\nbp/app-agnostic-logging-parameters\n\nChange-Id: I548120770d0f11a2af56f07d0902433a8c7c98fe\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.context/commit/cac31d9c6f0dd11cde816947abc2f2a7f74932be"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.context/commit/cac31d9c6f0dd11cde816947abc2f2a7f74932be"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"384c0de994102a5fbe8d8da1b37a3f1890ee574d","is_merged_in_target_branch":true,"change_id":"I51f8d6626ce1623f043a896cd0cc82478512f908","change_number":135100,"patch_set_number":5,"change_status":"MERGED"}],"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
