)]}'
{"id":"openstack%2Foslo.context~455974","triplet_id":"openstack%2Foslo.context~master~Ic04e9852d6b8ef400722a903c4ca5feec32b719d","project":"openstack/oslo.context","branch":"master","topic":"openstack/requirements","hashtags":[],"change_id":"Ic04e9852d6b8ef400722a903c4ca5feec32b719d","subject":"Updated from global requirements","status":"MERGED","created":"2017-04-12 04:19:58.000000000","updated":"2017-04-13 06:45:59.000000000","submitted":"2017-04-13 06:45:59.000000000","submitter":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"74d6a1120a5e6df2643c83962698fd8372664e4e","_number":455974,"virtual_id_number":455974,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"all":[{"value":2,"date":"2017-04-13 06:45:59.000000000","post_submit":true,"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":0,"_account_id":9796,"name":"ChangBo Guo","email":"glongwave@gmail.com","username":"gcb"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9796,"name":"ChangBo Guo","email":"glongwave@gmail.com","username":"gcb"},"all":[{"value":0,"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":2,"date":"2017-04-13 05:34:22.000000000","_account_id":9796,"name":"ChangBo Guo","email":"glongwave@gmail.com","username":"gcb"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9796,"name":"ChangBo Guo","email":"glongwave@gmail.com","username":"gcb"},"all":[{"value":0,"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":1,"date":"2017-04-13 05:34:22.000000000","_account_id":9796,"name":"ChangBo Guo","email":"glongwave@gmail.com","username":"gcb"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":0,"_account_id":9796,"name":"ChangBo Guo","email":"glongwave@gmail.com","username":"gcb"}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":3,"name":"Jenkins","username":"jenkins"},{"_account_id":9796,"name":"ChangBo Guo","email":"glongwave@gmail.com","username":"gcb"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2017-04-13 05:34:22.000000000","updated_by":{"_account_id":9796,"name":"ChangBo Guo","email":"glongwave@gmail.com","username":"gcb"},"reviewer":{"_account_id":9796,"name":"ChangBo Guo","email":"glongwave@gmail.com","username":"gcb"},"state":"REVIEWER"},{"updated":"2017-04-13 06:45:59.000000000","updated_by":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"reviewer":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"state":"REVIEWER"}],"messages":[{"id":"faeade5eae75562f77872e7ff8dfa1d91d6b9bf4","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2017-04-12 04:19:58.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"10d3548204218c65d76d63a4590efa56bc83b558","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2017-04-12 05:40:39.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- gate-oslo.context-docs-ubuntu-xenial http://docs-draft.openstack.org/74/455974/1/check/gate-oslo.context-docs-ubuntu-xenial/2923f9f//doc/build/html/ : SUCCESS in 5m 00s\n- gate-oslo.context-pep8-ubuntu-xenial http://logs.openstack.org/74/455974/1/check/gate-oslo.context-pep8-ubuntu-xenial/708fa70/ : SUCCESS in 4m 36s\n- gate-oslo.context-python27-ubuntu-xenial http://logs.openstack.org/74/455974/1/check/gate-oslo.context-python27-ubuntu-xenial/0741ad1/ : SUCCESS in 3m 18s\n- gate-oslo.context-python35 http://logs.openstack.org/74/455974/1/check/gate-oslo.context-python35/371ef65/ : SUCCESS in 3m 11s\n- gate-oslo.context-requirements http://logs.openstack.org/74/455974/1/check/gate-oslo.context-requirements/182c712/ : SUCCESS in 2m 20s\n- gate-tempest-dsvm-neutron-src-oslo.context-ubuntu-xenial http://logs.openstack.org/74/455974/1/check/gate-tempest-dsvm-neutron-src-oslo.context-ubuntu-xenial/28bfff6/ : SUCCESS in 1h 07m 52s\n- gate-tempest-dsvm-neutron-src-oslo.context-ubuntu-xenial-newton http://logs.openstack.org/74/455974/1/check/gate-tempest-dsvm-neutron-src-oslo.context-ubuntu-xenial-newton/c3a5400/ : FAILURE in 15m 53s (non-voting)\n- gate-tempest-dsvm-neutron-src-oslo.context-ubuntu-xenial-ocata http://logs.openstack.org/74/455974/1/check/gate-tempest-dsvm-neutron-src-oslo.context-ubuntu-xenial-ocata/ee47d56/ : FAILURE in 16m 10s (non-voting)\n- gate-oslo.context-releasenotes http://docs-draft.openstack.org/74/455974/1/check/gate-oslo.context-releasenotes/5e41c98//releasenotes/build/html/ : SUCCESS in 3m 58s","accounts_in_message":[],"_revision_number":1},{"id":"881fd5baa8d6abd2fe707be5f4617a89822584c1","author":{"_account_id":9796,"name":"ChangBo Guo","email":"glongwave@gmail.com","username":"gcb"},"date":"2017-04-13 05:34:22.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"369732eb7160009d7113ce72239ee9137643ecff","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2017-04-13 05:34:32.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.\nhttp://status.openstack.org/zuul/","accounts_in_message":[],"_revision_number":1},{"id":"209e813c072e46c71e60f77ca35cc0e297e46732","date":"2017-04-13 06:45:59.000000000","message":"Change has been successfully merged into the git repository by Jenkins","accounts_in_message":[],"_revision_number":1},{"id":"759c52dc850f432275577b60ac8ba4d90fbb6b23","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2017-04-13 06:45:59.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- gate-oslo.context-docs-ubuntu-xenial http://docs-draft.openstack.org/74/455974/1/gate/gate-oslo.context-docs-ubuntu-xenial/f6e57ce//doc/build/html/ : SUCCESS in 1m 47s\n- gate-oslo.context-pep8-ubuntu-xenial http://logs.openstack.org/74/455974/1/gate/gate-oslo.context-pep8-ubuntu-xenial/7f05e10/ : SUCCESS in 4m 24s\n- gate-oslo.context-python27-ubuntu-xenial http://logs.openstack.org/74/455974/1/gate/gate-oslo.context-python27-ubuntu-xenial/8a0dec3/ : SUCCESS in 3m 13s\n- gate-oslo.context-python35 http://logs.openstack.org/74/455974/1/gate/gate-oslo.context-python35/af458eb/ : SUCCESS in 1m 56s\n- gate-oslo.context-requirements http://logs.openstack.org/74/455974/1/gate/gate-oslo.context-requirements/ef4b56c/ : SUCCESS in 1m 55s\n- gate-tempest-dsvm-neutron-src-oslo.context-ubuntu-xenial http://logs.openstack.org/74/455974/1/gate/gate-tempest-dsvm-neutron-src-oslo.context-ubuntu-xenial/29cfdc0/ : SUCCESS in 1h 10m 09s\n- gate-oslo.context-releasenotes http://docs-draft.openstack.org/74/455974/1/gate/gate-oslo.context-releasenotes/7f923b1//releasenotes/build/html/ : SUCCESS in 2m 58s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"68b0476999cb240a450e3cb6289e1eb9caa7b063","revisions":{"68b0476999cb240a450e3cb6289e1eb9caa7b063":{"kind":"REWORK","_number":1,"created":"2017-04-12 04:19:58.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/74/455974/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.context","ref":"refs/changes/74/455974/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.context refs/changes/74/455974/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.context refs/changes/74/455974/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.context refs/changes/74/455974/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.context refs/changes/74/455974/1"}}},"commit":{"parents":[{"commit":"be7b1bf3319de103488c1722812a7e5f3353c9d0","subject":"Updated from global requirements","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.context/commit/be7b1bf3319de103488c1722812a7e5f3353c9d0"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2017-04-12 04:19:58.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2017-04-12 04:19:58.000000000","tz":0},"subject":"Updated from global requirements","message":"Updated from global requirements\n\nChange-Id: Ic04e9852d6b8ef400722a903c4ca5feec32b719d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.context/commit/68b0476999cb240a450e3cb6289e1eb9caa7b063"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.context/commit/68b0476999cb240a450e3cb6289e1eb9caa7b063"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"be7b1bf3319de103488c1722812a7e5f3353c9d0","is_merged_in_target_branch":true,"change_id":"I3c2534e4b628ba7666dd68a626d03a7287ed881e","change_number":440085,"patch_set_number":3,"change_status":"MERGED"}],"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
