)]}'
{"id":"openstack%2Foslo.context~530509","triplet_id":"openstack%2Foslo.context~master~I02fdaccfdd002d60b0b51c5d3327c783009cf35e","project":"openstack/oslo.context","branch":"master","topic":"bp/system-scope","hashtags":[],"change_id":"I02fdaccfdd002d60b0b51c5d3327c783009cf35e","subject":"Implement system-scope","status":"MERGED","created":"2017-12-29 21:29:29.000000000","updated":"2018-05-30 22:26:01.000000000","submitted":"2018-05-30 22:26:01.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":40,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"530509-1527719161032-39dde6cf","meta_rev_id":"4c5c5bcf6f18ec6a7afb475142196e5e023963af","_number":530509,"virtual_id_number":530509,"owner":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},{"value":0,"_account_id":11589,"name":"Harry Rybacki","email":"hrybacki@redhat.com","username":"hrybacki"},{"value":0,"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},{"value":0,"_account_id":2903,"name":"Morgan Fainberg","email":"morgan.fainberg@gmail.com","username":"mdrnstm"},{"value":0,"_account_id":7,"name":"Jay Pipes","email":"jaypipes@gmail.com","username":"jaypipes"},{"value":0,"_account_id":7191,"name":"Jamie Lennox","email":"jamielennox@gmail.com","username":"jamielennox"},{"value":0,"_account_id":2472,"name":"Doug Hellmann","email":"dhellmann@redhat.com","username":"doug-hellmann"},{"value":2,"date":"2018-05-30 22:26:00.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15054,"name":"wangxiyuan","email":"wangxiyuan1007@gmail.com","username":"wangxiyuan"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},"all":[{"value":2,"date":"2018-05-30 20:35:28.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},{"value":0,"_account_id":11589,"name":"Harry Rybacki","email":"hrybacki@redhat.com","username":"hrybacki"},{"value":0,"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},{"value":1,"date":"2018-05-30 19:33:12.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":2903,"name":"Morgan Fainberg","email":"morgan.fainberg@gmail.com","username":"mdrnstm"},{"value":0,"_account_id":7,"name":"Jay Pipes","email":"jaypipes@gmail.com","username":"jaypipes"},{"value":0,"_account_id":7191,"name":"Jamie Lennox","email":"jamielennox@gmail.com","username":"jamielennox"},{"value":2,"date":"2018-05-30 18:39:38.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":2472,"name":"Doug Hellmann","email":"dhellmann@redhat.com","username":"doug-hellmann"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15054,"name":"wangxiyuan","email":"wangxiyuan1007@gmail.com","username":"wangxiyuan"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},"all":[{"value":1,"date":"2018-05-30 20:35:28.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},{"value":0,"_account_id":11589,"name":"Harry Rybacki","email":"hrybacki@redhat.com","username":"hrybacki"},{"value":0,"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},{"value":0,"_account_id":2903,"name":"Morgan Fainberg","email":"morgan.fainberg@gmail.com","username":"mdrnstm"},{"value":0,"_account_id":7,"name":"Jay Pipes","email":"jaypipes@gmail.com","username":"jaypipes"},{"value":0,"_account_id":7191,"name":"Jamie Lennox","email":"jamielennox@gmail.com","username":"jamielennox"},{"value":0,"_account_id":2472,"name":"Doug Hellmann","email":"dhellmann@redhat.com","username":"doug-hellmann"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15054,"name":"wangxiyuan","email":"wangxiyuan1007@gmail.com","username":"wangxiyuan"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},{"value":0,"_account_id":11589,"name":"Harry Rybacki","email":"hrybacki@redhat.com","username":"hrybacki"},{"value":0,"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},{"value":0,"_account_id":2903,"name":"Morgan Fainberg","email":"morgan.fainberg@gmail.com","username":"mdrnstm"},{"value":0,"_account_id":7,"name":"Jay Pipes","email":"jaypipes@gmail.com","username":"jaypipes"},{"value":0,"_account_id":7191,"name":"Jamie Lennox","email":"jamielennox@gmail.com","username":"jamielennox"},{"value":0,"_account_id":2472,"name":"Doug Hellmann","email":"dhellmann@redhat.com","username":"doug-hellmann"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15054,"name":"wangxiyuan","email":"wangxiyuan1007@gmail.com","username":"wangxiyuan"}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":7,"name":"Jay Pipes","email":"jaypipes@gmail.com","username":"jaypipes"},{"_account_id":2472,"name":"Doug Hellmann","email":"dhellmann@redhat.com","username":"doug-hellmann"},{"_account_id":2903,"name":"Morgan Fainberg","email":"morgan.fainberg@gmail.com","username":"mdrnstm"},{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},{"_account_id":7191,"name":"Jamie Lennox","email":"jamielennox@gmail.com","username":"jamielennox"},{"_account_id":11589,"name":"Harry Rybacki","email":"hrybacki@redhat.com","username":"hrybacki"},{"_account_id":15054,"name":"wangxiyuan","email":"wangxiyuan1007@gmail.com","username":"wangxiyuan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2018-03-12 03:26:35.000000000","updated_by":{"_account_id":15054,"name":"wangxiyuan","email":"wangxiyuan1007@gmail.com","username":"wangxiyuan"},"reviewer":{"_account_id":15054,"name":"wangxiyuan","email":"wangxiyuan1007@gmail.com","username":"wangxiyuan"},"state":"REVIEWER"},{"updated":"2018-04-24 23:31:15.000000000","updated_by":{"_account_id":7,"name":"Jay Pipes","email":"jaypipes@gmail.com","username":"jaypipes"},"reviewer":{"_account_id":7,"name":"Jay Pipes","email":"jaypipes@gmail.com","username":"jaypipes"},"state":"REVIEWER"},{"updated":"2018-04-25 23:21:52.000000000","updated_by":{"_account_id":7191,"name":"Jamie Lennox","email":"jamielennox@gmail.com","username":"jamielennox"},"reviewer":{"_account_id":7191,"name":"Jamie Lennox","email":"jamielennox@gmail.com","username":"jamielennox"},"state":"REVIEWER"},{"updated":"2018-05-16 15:08:35.000000000","updated_by":{"_account_id":11589,"name":"Harry Rybacki","email":"hrybacki@redhat.com","username":"hrybacki"},"reviewer":{"_account_id":11589,"name":"Harry Rybacki","email":"hrybacki@redhat.com","username":"hrybacki"},"state":"REVIEWER"},{"updated":"2018-05-30 18:39:38.000000000","updated_by":{"_account_id":2472,"name":"Doug Hellmann","email":"dhellmann@redhat.com","username":"doug-hellmann"},"reviewer":{"_account_id":2472,"name":"Doug Hellmann","email":"dhellmann@redhat.com","username":"doug-hellmann"},"state":"REVIEWER"},{"updated":"2018-05-30 19:33:12.000000000","updated_by":{"_account_id":2903,"name":"Morgan Fainberg","email":"morgan.fainberg@gmail.com","username":"mdrnstm"},"reviewer":{"_account_id":2903,"name":"Morgan Fainberg","email":"morgan.fainberg@gmail.com","username":"mdrnstm"},"state":"REVIEWER"},{"updated":"2018-05-30 20:35:28.000000000","updated_by":{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},"reviewer":{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},"state":"REVIEWER"},{"updated":"2018-05-30 22:26:00.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"c01456ae2106e6444bb65e4e27c0b21104b76b09","author":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"date":"2017-12-29 21:29:29.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"b48141e6b3e6b37319548665278fb4ea1f27cbe7","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2017-12-29 22:45:29.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 http://logs.openstack.org/09/530509/1/check/openstack-tox-pep8/2786e4f/ : SUCCESS in 2m 55s\n- openstack-tox-py27 http://logs.openstack.org/09/530509/1/check/openstack-tox-py27/6d8b68e/ : FAILURE in 3m 13s\n- openstack-tox-py35 http://logs.openstack.org/09/530509/1/check/openstack-tox-py35/7b6bf6e/ : FAILURE in 4m 10s\n- build-openstack-sphinx-docs http://logs.openstack.org/09/530509/1/check/build-openstack-sphinx-docs/fc2035c/html/ : SUCCESS in 5m 35s\n- legacy-tempest-dsvm-neutron-src http://logs.openstack.org/09/530509/1/check/legacy-tempest-dsvm-neutron-src/a872c4e/ : SUCCESS in 1h 14m 40s","accounts_in_message":[],"_revision_number":1},{"id":"358a1dd0e5d190c4b1258c347d90971496bac588","author":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"date":"2018-03-09 03:17:04.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"99eb2ff2f0463348c49aec296052af6f6518bf34","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-03-09 04:40:38.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/09/530509/2/check/openstack-tox-pep8/50bddaa/ : SUCCESS in 3m 09s\n- openstack-tox-py27 http://logs.openstack.org/09/530509/2/check/openstack-tox-py27/0aed99f/ : SUCCESS in 2m 44s\n- openstack-tox-py35 http://logs.openstack.org/09/530509/2/check/openstack-tox-py35/405de79/ : SUCCESS in 3m 31s\n- build-openstack-sphinx-docs http://logs.openstack.org/09/530509/2/check/build-openstack-sphinx-docs/034b72a/html/ : SUCCESS in 3m 16s\n- legacy-tempest-dsvm-neutron-src http://logs.openstack.org/09/530509/2/check/legacy-tempest-dsvm-neutron-src/9ef1f57/ : SUCCESS in 1h 22m 31s","accounts_in_message":[],"_revision_number":2},{"id":"2d121c4591b5570ae61f03b702aacfe6365b7baa","author":{"_account_id":15054,"name":"wangxiyuan","email":"wangxiyuan1007@gmail.com","username":"wangxiyuan"},"date":"2018-03-12 03:26:35.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"141ab3bafd42d5a907be12bef87ecbfbad66dd72","author":{"_account_id":2472,"name":"Doug Hellmann","email":"dhellmann@redhat.com","username":"doug-hellmann"},"date":"2018-04-10 20:05:16.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"97bee661cc37410afac7cefb03846e729c01753f","author":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"date":"2018-04-23 14:05:33.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"752aa76d7762f7380af472dd91c797ad78a14c4d","author":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"date":"2018-04-23 14:05:44.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"8dfd467419b71dfa1a1cce0be789f633f9e37650","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-04-23 15:38:29.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/09/530509/3/check/openstack-tox-pep8/afb2714/ : SUCCESS in 3m 53s\n- openstack-tox-py27 http://logs.openstack.org/09/530509/3/check/openstack-tox-py27/665278e/ : SUCCESS in 3m 26s\n- openstack-tox-py35 http://logs.openstack.org/09/530509/3/check/openstack-tox-py35/0df5eb3/ : SUCCESS in 4m 05s\n- build-openstack-sphinx-docs http://logs.openstack.org/09/530509/3/check/build-openstack-sphinx-docs/66bb70c/html/ : SUCCESS in 4m 10s\n- legacy-tempest-dsvm-neutron-src http://logs.openstack.org/09/530509/3/check/legacy-tempest-dsvm-neutron-src/de4c8d8/ : SUCCESS in 1h 18m 44s\n- openstack-tox-lower-constraints http://logs.openstack.org/09/530509/3/check/openstack-tox-lower-constraints/890ae31/ : SUCCESS in 4m 14s","accounts_in_message":[],"_revision_number":3},{"id":"5f3acc6c8dc93798e173af5ac87ab5637d60f41d","author":{"_account_id":7,"name":"Jay Pipes","email":"jaypipes@gmail.com","username":"jaypipes"},"date":"2018-04-23 16:15:37.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"1dcec1b54376e672053655ef2681e0522cbe7c4a","author":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"date":"2018-04-23 18:25:35.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"b6c4ff0d230c7c3bf6c8ad8cb532ec2b54f1880f","author":{"_account_id":7,"name":"Jay Pipes","email":"jaypipes@gmail.com","username":"jaypipes"},"date":"2018-04-23 19:22:39.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"53a3c64fccade53144355f7382c8d1fbc86efd89","author":{"_account_id":2903,"name":"Morgan Fainberg","email":"morgan.fainberg@gmail.com","username":"mdrnstm"},"date":"2018-04-23 22:34:59.000000000","message":"Patch Set 3: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"5a238068c6697a6c7ee686194e6979c5b48ff6a0","author":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"date":"2018-04-24 13:18:27.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"6be939839760815a8eb22881cc3fb59ad76997e3","author":{"_account_id":7,"name":"Jay Pipes","email":"jaypipes@gmail.com","username":"jaypipes"},"date":"2018-04-24 14:12:23.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"1b63c73eb9ca8a6f5c9b35e88ed4a4ea0d8741b3","author":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"date":"2018-04-24 14:15:51.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"2d214ec4f4eae1b9231c89dedc1f572e0eafb8ec","author":{"_account_id":2903,"name":"Morgan Fainberg","email":"morgan.fainberg@gmail.com","username":"mdrnstm"},"date":"2018-04-24 15:30:26.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"640f8bc5470cb6dd4944c3d03a00b4a7443667c9","author":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"date":"2018-04-24 15:39:08.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"fc33fe9d0d31684b899e8725c1c6c728ee23b7fe","author":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"date":"2018-04-24 15:39:19.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"3fbab222f848ec658a6268ae4f340dc5cd1a18cb","author":{"_account_id":7,"name":"Jay Pipes","email":"jaypipes@gmail.com","username":"jaypipes"},"date":"2018-04-24 16:15:28.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"049e5ba39f26ccac03823ea887a4106409c9cded","author":{"_account_id":2903,"name":"Morgan Fainberg","email":"morgan.fainberg@gmail.com","username":"mdrnstm"},"date":"2018-04-24 17:12:16.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"ac8be5f7078523656c100e9e13182dc3793e4d9b","author":{"_account_id":2903,"name":"Morgan Fainberg","email":"morgan.fainberg@gmail.com","username":"mdrnstm"},"date":"2018-04-24 17:12:43.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"cf3b4f00b9f1c5d6e02688b3a8f99cc01f5adfe2","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-04-24 18:03:20.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/09/530509/4/check/openstack-tox-pep8/1a1d82a/ : SUCCESS in 4m 34s\n- openstack-tox-py27 http://logs.openstack.org/09/530509/4/check/openstack-tox-py27/bcc6454/ : SUCCESS in 4m 05s\n- openstack-tox-py35 http://logs.openstack.org/09/530509/4/check/openstack-tox-py35/fd85b02/ : SUCCESS in 3m 28s\n- build-openstack-sphinx-docs http://logs.openstack.org/09/530509/4/check/build-openstack-sphinx-docs/786551f/html/ : SUCCESS in 4m 13s\n- legacy-tempest-dsvm-neutron-src http://logs.openstack.org/09/530509/4/check/legacy-tempest-dsvm-neutron-src/cbbce22/ : SUCCESS in 1h 16m 10s\n- openstack-tox-lower-constraints http://logs.openstack.org/09/530509/4/check/openstack-tox-lower-constraints/495ee01/ : SUCCESS in 4m 57s","accounts_in_message":[],"_revision_number":4},{"id":"dd4e93862a5bb5380ac9795fdd7e7309e2cc8344","author":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"date":"2018-04-24 20:52:59.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"e6bbbe17e21c5b361e8d4414d5d9887083107143","author":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"date":"2018-04-24 20:57:08.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"c0bee1ed7591f26145d47fc073f9b7492b7abed9","author":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"date":"2018-04-24 21:34:04.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"3be9b2a4c3d520bba2992428aa3cf895e7bdfcaa","author":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"date":"2018-04-24 21:34:42.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"d34ef2fb87ae760b8cd1f02be628a5f5fdd284c3","author":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"date":"2018-04-24 22:03:47.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"1b23fe5b853e74201f30cd502f85204baec7d392","author":{"_account_id":7,"name":"Jay Pipes","email":"jaypipes@gmail.com","username":"jaypipes"},"date":"2018-04-24 23:31:15.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"d4d940b18c4b170c959bc577738d4f13d74780d2","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-04-25 00:22:14.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/09/530509/5/check/openstack-tox-pep8/cdfa29b/ : SUCCESS in 4m 00s\n- openstack-tox-py27 http://logs.openstack.org/09/530509/5/check/openstack-tox-py27/2789845/ : SUCCESS in 3m 49s\n- openstack-tox-py35 http://logs.openstack.org/09/530509/5/check/openstack-tox-py35/c4a830a/ : SUCCESS in 4m 09s\n- build-openstack-sphinx-docs http://logs.openstack.org/09/530509/5/check/build-openstack-sphinx-docs/db01079/html/ : SUCCESS in 4m 07s\n- legacy-tempest-dsvm-neutron-src http://logs.openstack.org/09/530509/5/check/legacy-tempest-dsvm-neutron-src/09c1826/ : SUCCESS in 1h 16m 37s\n- openstack-tox-lower-constraints http://logs.openstack.org/09/530509/5/check/openstack-tox-lower-constraints/b3ebd5b/ : SUCCESS in 4m 22s","accounts_in_message":[],"_revision_number":5},{"id":"ccc86870f2b171419d042d2fb55529cb3936773f","author":{"_account_id":7191,"name":"Jamie Lennox","email":"jamielennox@gmail.com","username":"jamielennox"},"date":"2018-04-25 23:21:52.000000000","message":"Patch Set 5: Code-Review-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":5},{"id":"46ec4eda7b8db97429b400c59f80d3333d426ec6","author":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"date":"2018-04-26 14:05:16.000000000","message":"Patch Set 5:\n\n(2 comments)","accounts_in_message":[],"_revision_number":5},{"id":"105e6f15bb0b769022d44d0e1e10262c528a3e03","author":{"_account_id":2903,"name":"Morgan Fainberg","email":"morgan.fainberg@gmail.com","username":"mdrnstm"},"date":"2018-05-02 16:02:00.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"16ca014785bb62f5b3b6f7bf78b011f0978e0f9f","author":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"date":"2018-05-02 19:18:21.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"487b04fbd06e00abf8fafa490174664aae824d98","author":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"date":"2018-05-02 19:19:27.000000000","message":"Patch Set 6:\n\n\u003e (2 comments)\n\nGood call - I\u0027ve update the test here to showcase the behavior you can expect if consuming the header from keystonemiddleware [0].\n\n[0] https://review.openstack.org/#/c/564072/","accounts_in_message":[],"_revision_number":6},{"id":"d8cf5114ed2a120ace18518f2d91d327e9619fdd","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-05-02 23:24:31.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/09/530509/6/check/openstack-tox-pep8/371f594/ : SUCCESS in 3m 38s\n- openstack-tox-py27 http://logs.openstack.org/09/530509/6/check/openstack-tox-py27/62850d6/ : SUCCESS in 4m 00s\n- openstack-tox-py35 http://logs.openstack.org/09/530509/6/check/openstack-tox-py35/d2b55a8/ : SUCCESS in 4m 58s\n- build-openstack-sphinx-docs http://logs.openstack.org/09/530509/6/check/build-openstack-sphinx-docs/3f83296/html/ : SUCCESS in 4m 43s\n- legacy-tempest-dsvm-neutron-src http://logs.openstack.org/09/530509/6/check/legacy-tempest-dsvm-neutron-src/6dd9910/ : SUCCESS in 1h 43m 12s\n- openstack-tox-lower-constraints http://logs.openstack.org/09/530509/6/check/openstack-tox-lower-constraints/f3b9f86/ : SUCCESS in 4m 57s","accounts_in_message":[],"_revision_number":6},{"id":"ff3d81add6cc5686be8049d40a2732c0687ed8e4","author":{"_account_id":2903,"name":"Morgan Fainberg","email":"morgan.fainberg@gmail.com","username":"mdrnstm"},"date":"2018-05-09 15:51:57.000000000","message":"Patch Set 6: Code-Review+1\n\nGenerally looks good to me.","accounts_in_message":[],"_revision_number":6},{"id":"8224a2bf43c5b91cc0be95c90f2c59567e780b6f","author":{"_account_id":2472,"name":"Doug Hellmann","email":"dhellmann@redhat.com","username":"doug-hellmann"},"date":"2018-05-15 16:03:27.000000000","message":"Patch Set 6:\n\n(3 comments)","accounts_in_message":[],"_revision_number":6},{"id":"f4aed83c52d674590fa2a0131a7112edeafd5111","author":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"date":"2018-05-15 17:56:17.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"1bc6645bdfe591c577095a395ac224af39daa553","author":{"_account_id":2472,"name":"Doug Hellmann","email":"dhellmann@redhat.com","username":"doug-hellmann"},"date":"2018-05-15 18:06:56.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"c7ea10eeb98036af730548f0ae8336584bfcce55","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-05-15 18:12:50.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 http://logs.openstack.org/09/530509/7/check/openstack-tox-pep8/984b641/ : SUCCESS in 3m 51s\n- openstack-tox-py27 http://logs.openstack.org/09/530509/7/check/openstack-tox-py27/9d028f4/ : SUCCESS in 3m 52s\n- openstack-tox-py35 http://logs.openstack.org/09/530509/7/check/openstack-tox-py35/b3a82f9/ : SUCCESS in 4m 21s\n- build-openstack-sphinx-docs http://logs.openstack.org/09/530509/7/check/build-openstack-sphinx-docs/8b63b55/html/ : SUCCESS in 4m 03s\n- legacy-tempest-dsvm-neutron-src http://logs.openstack.org/09/530509/7/check/legacy-tempest-dsvm-neutron-src/723a2d9/ : FAILURE in 13m 27s\n- openstack-tox-lower-constraints http://logs.openstack.org/09/530509/7/check/openstack-tox-lower-constraints/6bba0b2/ : SUCCESS in 4m 00s","accounts_in_message":[],"_revision_number":7},{"id":"7507247db7d5267789b1dc1c87137002e44d8d34","author":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"date":"2018-05-15 19:10:10.000000000","message":"Patch Set 6:\n\n(3 comments)","accounts_in_message":[],"_revision_number":6},{"id":"ae2640d7c4b2ef556b2ad58b43c0a71157a94f94","author":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"date":"2018-05-15 21:52:41.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"782575ed7325de805b9ec0d5b57ee8e995bd85ac","author":{"_account_id":2472,"name":"Doug Hellmann","email":"dhellmann@redhat.com","username":"doug-hellmann"},"date":"2018-05-16 13:56:52.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"0a20ed7feacb8df02ca1f86921b498a706a3989d","author":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"date":"2018-05-16 14:40:29.000000000","message":"Uploaded patch set 8.","accounts_in_message":[],"_revision_number":8},{"id":"cfd616194bd8a7256af3413a246bd984dbea5ddc","author":{"_account_id":11589,"name":"Harry Rybacki","email":"hrybacki@redhat.com","username":"hrybacki"},"date":"2018-05-16 14:42:18.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"290acd227e3c22312cbd6a1a6bd539509483209b","author":{"_account_id":11589,"name":"Harry Rybacki","email":"hrybacki@redhat.com","username":"hrybacki"},"date":"2018-05-16 14:46:12.000000000","message":"Patch Set 8:\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"6f19c50b9e24c08f53b6c3b419ea840b0a5c269d","author":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"date":"2018-05-16 15:05:21.000000000","message":"Patch Set 8:\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"831d04142d21815772a08ed0e6a4c9f9c3401b93","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-05-16 15:06:06.000000000","message":"Patch Set 8: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 http://logs.openstack.org/09/530509/8/check/openstack-tox-pep8/20a9270/ : SUCCESS in 5m 14s\n- openstack-tox-py27 http://logs.openstack.org/09/530509/8/check/openstack-tox-py27/166ad55/ : SUCCESS in 3m 57s\n- openstack-tox-py35 http://logs.openstack.org/09/530509/8/check/openstack-tox-py35/3857b97/ : SUCCESS in 3m 57s\n- build-openstack-sphinx-docs http://logs.openstack.org/09/530509/8/check/build-openstack-sphinx-docs/5edad34/html/ : SUCCESS in 5m 22s\n- legacy-tempest-dsvm-neutron-src http://logs.openstack.org/09/530509/8/check/legacy-tempest-dsvm-neutron-src/5d6dee5/ : FAILURE in 16m 20s\n- openstack-tox-lower-constraints http://logs.openstack.org/09/530509/8/check/openstack-tox-lower-constraints/becdd02/ : SUCCESS in 3m 47s","accounts_in_message":[],"_revision_number":8},{"id":"4943b243395f5569b86f93ec7703d1482eb1279b","author":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"date":"2018-05-16 15:08:06.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"4d56df12d552757e07e01598a4bf01e6e19a6a85","author":{"_account_id":11589,"name":"Harry Rybacki","email":"hrybacki@redhat.com","username":"hrybacki"},"date":"2018-05-16 15:08:35.000000000","message":"Patch Set 8:\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"7c6650a11f2dce67bd75871c3baa4938b777ea71","author":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"date":"2018-05-16 15:08:44.000000000","message":"Patch Set 8:\n\nrecheck","accounts_in_message":[],"_revision_number":8},{"id":"fc2054d9c9c036df70cc4e1e491ae506cdbc353f","author":{"_account_id":2472,"name":"Doug Hellmann","email":"dhellmann@redhat.com","username":"doug-hellmann"},"date":"2018-05-16 15:31:51.000000000","message":"Patch Set 8:\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"e9c45f929e2fab11ad3fa5e0d381b25e800ca998","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-05-16 15:37:08.000000000","message":"Patch Set 8:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 http://logs.openstack.org/09/530509/8/check/openstack-tox-pep8/c7d4035/ : SUCCESS in 3m 39s\n- openstack-tox-py27 http://logs.openstack.org/09/530509/8/check/openstack-tox-py27/f7ac038/ : SUCCESS in 3m 42s\n- openstack-tox-py35 http://logs.openstack.org/09/530509/8/check/openstack-tox-py35/773d2f7/ : SUCCESS in 3m 41s\n- build-openstack-sphinx-docs http://logs.openstack.org/09/530509/8/check/build-openstack-sphinx-docs/93e6133/html/ : SUCCESS in 3m 49s\n- legacy-tempest-dsvm-neutron-src http://logs.openstack.org/09/530509/8/check/legacy-tempest-dsvm-neutron-src/337e482/ : FAILURE in 17m 35s\n- openstack-tox-lower-constraints http://logs.openstack.org/09/530509/8/check/openstack-tox-lower-constraints/2abd87b/ : SUCCESS in 3m 59s","accounts_in_message":[],"_revision_number":8},{"id":"c35051d10678f26de874ec45927e8e4dc1d3e011","author":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"date":"2018-05-16 18:03:30.000000000","message":"Uploaded patch set 9.","accounts_in_message":[],"_revision_number":9},{"id":"fa91b221f626929a0ac56addf8ea74a52ea902f9","author":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"date":"2018-05-16 18:03:35.000000000","message":"Patch Set 8:\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"3973c7fdb39a7b0f3a2c19936e9e54e9710aa796","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-05-16 19:45:16.000000000","message":"Patch Set 9: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/09/530509/9/check/openstack-tox-pep8/28a385f/ : SUCCESS in 4m 10s\n- openstack-tox-py27 http://logs.openstack.org/09/530509/9/check/openstack-tox-py27/b775897/ : SUCCESS in 4m 04s\n- openstack-tox-py35 http://logs.openstack.org/09/530509/9/check/openstack-tox-py35/0c2bb34/ : SUCCESS in 4m 22s\n- build-openstack-sphinx-docs http://logs.openstack.org/09/530509/9/check/build-openstack-sphinx-docs/7cd6202/html/ : SUCCESS in 4m 14s\n- legacy-tempest-dsvm-neutron-src http://logs.openstack.org/09/530509/9/check/legacy-tempest-dsvm-neutron-src/5837af3/ : SUCCESS in 1h 34m 41s\n- openstack-tox-lower-constraints http://logs.openstack.org/09/530509/9/check/openstack-tox-lower-constraints/8fc7b2c/ : SUCCESS in 3m 47s","accounts_in_message":[],"_revision_number":9},{"id":"9309901f7dc0c8bac018fe41877463b229d80110","author":{"_account_id":2472,"name":"Doug Hellmann","email":"dhellmann@redhat.com","username":"doug-hellmann"},"date":"2018-05-30 18:39:38.000000000","message":"Patch Set 9: Code-Review+2","accounts_in_message":[],"_revision_number":9},{"id":"e43fdf251aff44c67600ae8476d103f8f9853c5a","author":{"_account_id":2903,"name":"Morgan Fainberg","email":"morgan.fainberg@gmail.com","username":"mdrnstm"},"date":"2018-05-30 19:33:12.000000000","message":"Patch Set 9: Code-Review+1","accounts_in_message":[],"_revision_number":9},{"id":"84126f1178fb7186c1d0f817678cb5c30f17de9d","author":{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},"date":"2018-05-30 20:35:28.000000000","message":"Patch Set 9: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":9},{"id":"bc840fd83bc5a3508d41106fb1f9639e01a98c98","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-05-30 20:35:40.000000000","message":"Patch Set 9: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":9},{"id":"0fc52923060f784e45b82ac6c19d9b1fe41d36ed","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-05-30 22:26:00.000000000","message":"Patch Set 9: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/09/530509/9/gate/openstack-tox-pep8/1b6d076/ : SUCCESS in 3m 37s\n- openstack-tox-py27 http://logs.openstack.org/09/530509/9/gate/openstack-tox-py27/095b993/ : SUCCESS in 4m 49s\n- openstack-tox-py35 http://logs.openstack.org/09/530509/9/gate/openstack-tox-py35/ad6bb2e/ : SUCCESS in 3m 54s\n- build-openstack-sphinx-docs http://logs.openstack.org/09/530509/9/gate/build-openstack-sphinx-docs/5c6fcdd/html/ : SUCCESS in 3m 43s\n- legacy-tempest-dsvm-neutron-src http://logs.openstack.org/09/530509/9/gate/legacy-tempest-dsvm-neutron-src/5633206/ : SUCCESS in 1h 44m 33s\n- openstack-tox-lower-constraints http://logs.openstack.org/09/530509/9/gate/openstack-tox-lower-constraints/582e207/ : SUCCESS in 3m 25s","accounts_in_message":[],"_revision_number":9},{"id":"ba43d978c5fdf0ad95759daeaf23f7bfb12f29dd","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-05-30 22:26:01.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":9}],"current_revision_number":9,"current_revision":"1a40b3d43bac5244bcba6bdbc4802fb76430d8d3","revisions":{"a7ea625e145906d8b25a81f2c68cb5934ee028b1":{"kind":"REWORK","_number":1,"created":"2017-12-29 21:29:29.000000000","uploader":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"ref":"refs/changes/09/530509/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.context","ref":"refs/changes/09/530509/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.context refs/changes/09/530509/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.context refs/changes/09/530509/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.context refs/changes/09/530509/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.context refs/changes/09/530509/1"}}},"commit":{"parents":[{"commit":"f9970a94e5c70c6fa453061196899cec7a8a5b40","subject":"Merge \"add bandit to pep8 job\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.context/commit/f9970a94e5c70c6fa453061196899cec7a8a5b40"}]}],"author":{"name":"Lance Bragstad","email":"lbragstad@gmail.com","date":"2017-12-29 21:28:17.000000000","tz":0},"committer":{"name":"Lance Bragstad","email":"lbragstad@gmail.com","date":"2017-12-29 21:28:17.000000000","tz":0},"subject":"Implement system-scope","message":"Implement system-scope\n\nThe context should carry some information that all projects will need\nin order to enforce scoping. System scope can be implemented here\nand available for projects when they start adding scope types to\npolicies.\n\nbp system-scope\n\nChange-Id: I02fdaccfdd002d60b0b51c5d3327c783009cf35e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.context/commit/a7ea625e145906d8b25a81f2c68cb5934ee028b1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.context/commit/a7ea625e145906d8b25a81f2c68cb5934ee028b1"}]},"branch":"refs/heads/master"},"304ab1085499290e2ecb307ddfefcf07e5807198":{"kind":"REWORK","_number":2,"created":"2018-03-09 03:17:04.000000000","uploader":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"ref":"refs/changes/09/530509/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.context","ref":"refs/changes/09/530509/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.context refs/changes/09/530509/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.context refs/changes/09/530509/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.context refs/changes/09/530509/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.context refs/changes/09/530509/2"}}},"commit":{"parents":[{"commit":"92b96644eb2db08ec4367161f5a0b08403065348","subject":"Add -W for document build","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.context/commit/92b96644eb2db08ec4367161f5a0b08403065348"}]}],"author":{"name":"Lance Bragstad","email":"lbragstad@gmail.com","date":"2017-12-29 21:28:17.000000000","tz":0},"committer":{"name":"Lance Bragstad","email":"lbragstad@gmail.com","date":"2018-03-09 03:16:41.000000000","tz":0},"subject":"Implement system-scope","message":"Implement system-scope\n\nThe context should carry some information that all services will need\nin order to enforce scoping. System scope can be implemented here\nand available for projects when they start adding scope types to\npolicies.\n\nbp system-scope\n\nChange-Id: I02fdaccfdd002d60b0b51c5d3327c783009cf35e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.context/commit/304ab1085499290e2ecb307ddfefcf07e5807198"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.context/commit/304ab1085499290e2ecb307ddfefcf07e5807198"}]},"branch":"refs/heads/master"},"e752d81b3c934930c6f38dc61491641bf9623e92":{"kind":"REWORK","_number":3,"created":"2018-04-23 14:05:33.000000000","uploader":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"ref":"refs/changes/09/530509/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.context","ref":"refs/changes/09/530509/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.context refs/changes/09/530509/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.context refs/changes/09/530509/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.context refs/changes/09/530509/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.context refs/changes/09/530509/3"}}},"commit":{"parents":[{"commit":"92b96644eb2db08ec4367161f5a0b08403065348","subject":"Add -W for document build","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.context/commit/92b96644eb2db08ec4367161f5a0b08403065348"}]}],"author":{"name":"Lance Bragstad","email":"lbragstad@gmail.com","date":"2017-12-29 21:28:17.000000000","tz":0},"committer":{"name":"Lance Bragstad","email":"lbragstad@gmail.com","date":"2018-04-23 14:05:27.000000000","tz":0},"subject":"Implement system-scope","message":"Implement system-scope\n\nThe context should carry some information that all services will need\nin order to enforce scoping. System scope can be implemented here\nand available for projects when they start adding scope types to\npolicies.\n\nbp system-scope\n\nChange-Id: I02fdaccfdd002d60b0b51c5d3327c783009cf35e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.context/commit/e752d81b3c934930c6f38dc61491641bf9623e92"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.context/commit/e752d81b3c934930c6f38dc61491641bf9623e92"}]},"branch":"refs/heads/master"},"390f66b24bb839651fb2e1ff7345d87192482e5a":{"kind":"REWORK","_number":4,"created":"2018-04-24 15:39:19.000000000","uploader":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"ref":"refs/changes/09/530509/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.context","ref":"refs/changes/09/530509/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.context refs/changes/09/530509/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.context refs/changes/09/530509/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.context refs/changes/09/530509/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.context refs/changes/09/530509/4"}}},"commit":{"parents":[{"commit":"92b96644eb2db08ec4367161f5a0b08403065348","subject":"Add -W for document build","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.context/commit/92b96644eb2db08ec4367161f5a0b08403065348"}]}],"author":{"name":"Lance Bragstad","email":"lbragstad@gmail.com","date":"2017-12-29 21:28:17.000000000","tz":0},"committer":{"name":"Lance Bragstad","email":"lbragstad@gmail.com","date":"2018-04-24 15:39:13.000000000","tz":0},"subject":"Implement system-scope","message":"Implement system-scope\n\nThe context should carry some information that all services will need\nin order to enforce scoping. System scope can be implemented here\nand available for projects when they start adding scope types to\npolicies.\n\nbp system-scope\n\nChange-Id: I02fdaccfdd002d60b0b51c5d3327c783009cf35e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.context/commit/390f66b24bb839651fb2e1ff7345d87192482e5a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.context/commit/390f66b24bb839651fb2e1ff7345d87192482e5a"}]},"branch":"refs/heads/master"},"c4d55eb2b4057b522b184e155121e54d0d6da84c":{"kind":"REWORK","_number":5,"created":"2018-04-24 21:34:42.000000000","uploader":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"ref":"refs/changes/09/530509/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.context","ref":"refs/changes/09/530509/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.context refs/changes/09/530509/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.context refs/changes/09/530509/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.context refs/changes/09/530509/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.context refs/changes/09/530509/5"}}},"commit":{"parents":[{"commit":"92b96644eb2db08ec4367161f5a0b08403065348","subject":"Add -W for document build","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.context/commit/92b96644eb2db08ec4367161f5a0b08403065348"}]}],"author":{"name":"Lance Bragstad","email":"lbragstad@gmail.com","date":"2017-12-29 21:28:17.000000000","tz":0},"committer":{"name":"Lance Bragstad","email":"lbragstad@gmail.com","date":"2018-04-24 21:34:36.000000000","tz":0},"subject":"Implement system-scope","message":"Implement system-scope\n\nThe context should carry some information that all services will need\nin order to enforce scoping. System scope can be implemented here\nand available for projects when they start adding scope types to\npolicies.\n\nbp system-scope\n\nChange-Id: I02fdaccfdd002d60b0b51c5d3327c783009cf35e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.context/commit/c4d55eb2b4057b522b184e155121e54d0d6da84c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.context/commit/c4d55eb2b4057b522b184e155121e54d0d6da84c"}]},"branch":"refs/heads/master"},"34078d7d48a321d00fe1338a8b87f558c2acd12d":{"kind":"REWORK","_number":6,"created":"2018-05-02 19:18:21.000000000","uploader":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"ref":"refs/changes/09/530509/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.context","ref":"refs/changes/09/530509/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.context refs/changes/09/530509/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.context refs/changes/09/530509/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.context refs/changes/09/530509/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.context refs/changes/09/530509/6"}}},"commit":{"parents":[{"commit":"92b96644eb2db08ec4367161f5a0b08403065348","subject":"Add -W for document build","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.context/commit/92b96644eb2db08ec4367161f5a0b08403065348"}]}],"author":{"name":"Lance Bragstad","email":"lbragstad@gmail.com","date":"2017-12-29 21:28:17.000000000","tz":0},"committer":{"name":"Lance Bragstad","email":"lbragstad@gmail.com","date":"2018-05-02 19:18:15.000000000","tz":0},"subject":"Implement system-scope","message":"Implement system-scope\n\nThe context should carry some information that all services will need\nin order to enforce scoping. System scope can be implemented here\nand available for projects when they start adding scope types to\npolicies.\n\nbp system-scope\n\nChange-Id: I02fdaccfdd002d60b0b51c5d3327c783009cf35e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.context/commit/34078d7d48a321d00fe1338a8b87f558c2acd12d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.context/commit/34078d7d48a321d00fe1338a8b87f558c2acd12d"}]},"branch":"refs/heads/master"},"de869a47affb819d39ec3623d7caec536c487d1d":{"kind":"REWORK","_number":7,"created":"2018-05-15 17:56:17.000000000","uploader":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"ref":"refs/changes/09/530509/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.context","ref":"refs/changes/09/530509/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.context refs/changes/09/530509/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.context refs/changes/09/530509/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.context refs/changes/09/530509/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.context refs/changes/09/530509/7"}}},"commit":{"parents":[{"commit":"92b96644eb2db08ec4367161f5a0b08403065348","subject":"Add -W for document build","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.context/commit/92b96644eb2db08ec4367161f5a0b08403065348"}]}],"author":{"name":"Lance Bragstad","email":"lbragstad@gmail.com","date":"2017-12-29 21:28:17.000000000","tz":0},"committer":{"name":"Lance Bragstad","email":"lbragstad@gmail.com","date":"2018-05-15 17:56:10.000000000","tz":0},"subject":"Implement system-scope","message":"Implement system-scope\n\nThe context should carry some information that all services will need\nin order to enforce scoping. System scope can be implemented here\nand available for projects when they start adding scope types to\npolicies.\n\nbp system-scope\n\nChange-Id: I02fdaccfdd002d60b0b51c5d3327c783009cf35e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.context/commit/de869a47affb819d39ec3623d7caec536c487d1d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.context/commit/de869a47affb819d39ec3623d7caec536c487d1d"}]},"branch":"refs/heads/master"},"cf409f11824778c711616e615df4dc61b344fb09":{"kind":"REWORK","_number":8,"created":"2018-05-16 14:40:29.000000000","uploader":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"ref":"refs/changes/09/530509/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.context","ref":"refs/changes/09/530509/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.context refs/changes/09/530509/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.context refs/changes/09/530509/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.context refs/changes/09/530509/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.context refs/changes/09/530509/8"}}},"commit":{"parents":[{"commit":"92b96644eb2db08ec4367161f5a0b08403065348","subject":"Add -W for document build","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.context/commit/92b96644eb2db08ec4367161f5a0b08403065348"}]}],"author":{"name":"Lance Bragstad","email":"lbragstad@gmail.com","date":"2017-12-29 21:28:17.000000000","tz":0},"committer":{"name":"Lance Bragstad","email":"lbragstad@gmail.com","date":"2018-05-16 14:40:21.000000000","tz":0},"subject":"Implement system-scope","message":"Implement system-scope\n\nThe context should carry some information that all services will need\nin order to enforce scoping. System scope can be implemented here\nand available for projects when they start adding scope types to\npolicies.\n\nbp system-scope\n\nChange-Id: I02fdaccfdd002d60b0b51c5d3327c783009cf35e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.context/commit/cf409f11824778c711616e615df4dc61b344fb09"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.context/commit/cf409f11824778c711616e615df4dc61b344fb09"}]},"branch":"refs/heads/master"},"1a40b3d43bac5244bcba6bdbc4802fb76430d8d3":{"kind":"REWORK","_number":9,"created":"2018-05-16 18:03:30.000000000","uploader":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"ref":"refs/changes/09/530509/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.context","ref":"refs/changes/09/530509/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.context refs/changes/09/530509/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.context refs/changes/09/530509/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.context refs/changes/09/530509/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.context refs/changes/09/530509/9"}}},"commit":{"parents":[{"commit":"92b96644eb2db08ec4367161f5a0b08403065348","subject":"Add -W for document build","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.context/commit/92b96644eb2db08ec4367161f5a0b08403065348"}]}],"author":{"name":"Lance Bragstad","email":"lbragstad@gmail.com","date":"2017-12-29 21:28:17.000000000","tz":0},"committer":{"name":"Lance Bragstad","email":"lbragstad@gmail.com","date":"2018-05-16 18:03:24.000000000","tz":0},"subject":"Implement system-scope","message":"Implement system-scope\n\nThe context should carry some information that all services will need\nin order to enforce scoping. System scope can be implemented here\nand available for projects when they start adding scope types to\npolicies.\n\nbp system-scope\n\nChange-Id: I02fdaccfdd002d60b0b51c5d3327c783009cf35e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.context/commit/1a40b3d43bac5244bcba6bdbc4802fb76430d8d3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.context/commit/1a40b3d43bac5244bcba6bdbc4802fb76430d8d3"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
