)]}'
{"id":"openstack%2Foslo.context~896453","triplet_id":"openstack%2Foslo.context~master~I6141842a31379ba6e9b136fe36245ef6b2ab7137","project":"openstack/oslo.context","branch":"master","attention_set":{},"removed_from_attention_set":{"10342":{"account":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},"last_update":"2024-11-06 23:33:12.000000000","reason":"Change was abandoned"},"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2024-11-06 23:33:12.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"I6141842a31379ba6e9b136fe36245ef6b2ab7137","subject":"Add is_admin to default version of redacted_context","status":"ABANDONED","created":"2023-09-25 17:37:16.000000000","updated":"2024-11-06 23:33:12.000000000","total_comment_count":2,"unresolved_comment_count":1,"has_review_started":true,"meta_rev_id":"a3a4624a7538241e28640b0ac4042e22b85f2a5e","_number":896453,"virtual_id_number":896453,"owner":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2024-11-06 23:31:45.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-09-25 19:25:42.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-01-26 11:35:39.000000000","updated_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"reviewer":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"state":"REVIEWER"}],"messages":[{"id":"66c1e677f0e56ff52a6d38ef5e9dbab9f3e21458","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},"date":"2023-09-25 17:37:16.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"ef65d083751249824f2772d1a9d060651a2210a9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},"date":"2023-09-25 18:09:06.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"a5671665f62658fe410efcf991d604e2f62f7b0d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-25 19:25:42.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2e43ad1030464fdd9bd9c6e0f2eabd9c\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/2d4a41f29553402395871b94c6f904f6 : SUCCESS in 1h 15m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b6125ea6bbbf4012b08a2765596d2451 : SUCCESS in 3m 08s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1492141fe64242afbec593c1c0e4ecd1 : SUCCESS in 3m 08s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f285f608deae4073aaa0ffbd4fa4796e : SUCCESS in 3m 14s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/5b61d8efb0944aa4a4aab6099cbaee6a : SUCCESS in 2m 47s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/89efe5fe3b364fc3ba8382052b726fd4 : SUCCESS in 3m 16s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4432ba94fc6342d08bc5787926fc8b3a : SUCCESS in 2m 59s","accounts_in_message":[],"_revision_number":2},{"id":"f0b9ebdf3284655ce68d6bf81f02e944e622683f","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-01-26 11:35:39.000000000","message":"Patch Set 2: Code-Review-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":2},{"id":"a7fcb02ed5db478efadcb1facb923c419f74ab82","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},"date":"2024-11-06 23:02:51.000000000","message":"Patch Set 3: Commit message was updated.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":3},{"id":"9d2e9878a7ced77d212b141fbc5ca3df68b16205","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-06 23:31:45.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/24f84158764e463bbc78f1fa2d1b4e8c\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/ef93a8bef23e4aadbf9debc6dc9b15d6 : FAILURE in 23m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9923880e99924466b8cdeb08b2616039 : SUCCESS in 3m 03s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/81163e295be1474bbde6df1e9f53effc : FAILURE in 2m 35s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/193725c19f67493eb2d8a07e60365d87 : FAILURE in 3m 25s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c54b58c4ba46479bb122884bc719d227 : FAILURE in 2m 48s","accounts_in_message":[],"_revision_number":3},{"id":"a3a4624a7538241e28640b0ac4042e22b85f2a5e","tag":"autogenerated:gerrit:abandon","author":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},"date":"2024-11-06 23:33:12.000000000","message":"Abandoned\n\nthis was fixed in another change, apparently?","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"5b9251db28f571f9656c759d1638f70ee411ea62","revisions":{"d2f54487d77d83440b5a3695d1707ec41e9ea419":{"kind":"REWORK","_number":1,"created":"2023-09-25 17:37:16.000000000","uploader":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},"ref":"refs/changes/53/896453/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.context","ref":"refs/changes/53/896453/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.context refs/changes/53/896453/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.context refs/changes/53/896453/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.context refs/changes/53/896453/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.context refs/changes/53/896453/1"}}},"commit":{"parents":[{"commit":"059f9ba16ef7f49293b62b1505b85ca15bd19624","subject":"Add method for getting redacted copy of context","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.context/commit/059f9ba16ef7f49293b62b1505b85ca15bd19624"}]}],"author":{"name":"Jay Faulkner","email":"jay@jvf.cc","date":"2023-09-25 17:36:36.000000000","tz":-420},"committer":{"name":"Jay Faulkner","email":"jay@jvf.cc","date":"2023-09-25 17:36:36.000000000","tz":-420},"subject":"Add is_admin to default version of redacted_context","message":"Add is_admin to default version of redacted_context\n\nThis is to avoid a version of bug 2037312 during the transition, before\nHeat has had a chance to write their version of redacted_copy.\n\nRelated-big: 2037312\nChange-Id: I6141842a31379ba6e9b136fe36245ef6b2ab7137\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.context/commit/d2f54487d77d83440b5a3695d1707ec41e9ea419"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.context/commit/d2f54487d77d83440b5a3695d1707ec41e9ea419"}]},"branch":"refs/heads/master"},"bac8a1bd68b2ba00b42f8399c53562dbd4f4bdbe":{"kind":"REWORK","_number":2,"created":"2023-09-25 18:09:06.000000000","uploader":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},"ref":"refs/changes/53/896453/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.context","ref":"refs/changes/53/896453/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.context refs/changes/53/896453/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.context refs/changes/53/896453/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.context refs/changes/53/896453/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.context refs/changes/53/896453/2"}}},"commit":{"parents":[{"commit":"059f9ba16ef7f49293b62b1505b85ca15bd19624","subject":"Add method for getting redacted copy of context","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.context/commit/059f9ba16ef7f49293b62b1505b85ca15bd19624"}]}],"author":{"name":"Jay Faulkner","email":"jay@jvf.cc","date":"2023-09-25 17:36:36.000000000","tz":-420},"committer":{"name":"Jay Faulkner","email":"jay@jvf.cc","date":"2023-09-25 18:09:06.000000000","tz":0},"subject":"Add is_admin to default version of redacted_context","message":"Add is_admin to default version of redacted_context\n\nThis is to avoid a version of bug 2037312 during the transition, before\nHeat has had a chance to write their version of redacted_copy.\n\nRelated-big: 2037312\nChange-Id: I6141842a31379ba6e9b136fe36245ef6b2ab7137\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.context/commit/bac8a1bd68b2ba00b42f8399c53562dbd4f4bdbe"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.context/commit/bac8a1bd68b2ba00b42f8399c53562dbd4f4bdbe"}]},"branch":"refs/heads/master"},"5b9251db28f571f9656c759d1638f70ee411ea62":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2024-11-06 23:02:51.000000000","uploader":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},"ref":"refs/changes/53/896453/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.context","ref":"refs/changes/53/896453/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.context refs/changes/53/896453/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.context refs/changes/53/896453/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.context refs/changes/53/896453/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.context refs/changes/53/896453/3"}}},"commit":{"parents":[{"commit":"059f9ba16ef7f49293b62b1505b85ca15bd19624","subject":"Add method for getting redacted copy of context","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.context/commit/059f9ba16ef7f49293b62b1505b85ca15bd19624"}]}],"author":{"name":"Jay Faulkner","email":"jay@jvf.cc","date":"2023-09-25 17:36:36.000000000","tz":-420},"committer":{"name":"Jay Faulkner","email":"jay@jvf.cc","date":"2024-11-06 23:02:51.000000000","tz":0},"subject":"Add is_admin to default version of redacted_context","message":"Add is_admin to default version of redacted_context\n\nThis is to avoid a version of bug 2037312 during the transition, before\nHeat has had a chance to write their version of redacted_copy.\n\nRelated-bug: 2037312\nChange-Id: I6141842a31379ba6e9b136fe36245ef6b2ab7137\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.context/commit/5b9251db28f571f9656c759d1638f70ee411ea62"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.context/commit/5b9251db28f571f9656c759d1638f70ee411ea62"}]},"branch":"refs/heads/master","description":"Edit commit message"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
