)]}'
{"id":"openstack%2Foslo.context~957378","triplet_id":"openstack%2Foslo.context~master~Ifeb27e606959b6a8c7a3ae853aff1e2e74e7b5d4","project":"openstack/oslo.context","branch":"master","topic":"release-health-check-flamingo","attention_set":{},"removed_from_attention_set":{"17685":{"account":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"last_update":"2025-08-15 12:22:36.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"Ifeb27e606959b6a8c7a3ae853aff1e2e74e7b5d4","subject":"DNM: gate health test","status":"ABANDONED","created":"2025-08-14 09:45:53.000000000","updated":"2025-08-15 12:22:36.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"a3f1aef8314d2bfbdf65f664f4048c506f183621","_number":957378,"virtual_id_number":957378,"owner":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2025-08-14 14:31:37.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-08-14 14:31:37.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"eb648d10915aab9129dc60451d08eb36d5a1c807","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2025-08-14 09:45:53.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"0b51df2176254751a8c704513b9eb6c1af6b94f5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-14 14:31:37.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/4c8f7dfb020045c2905a3503aced9b99\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/6d4f088ae2204715a193e5246ba9b5ff : SUCCESS in 1h 44m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a546eb0e922e45d79e757fcf27feec62 : FAILURE in 1m 40s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/4cf96131324649a58f20c44f5b9f6237 : SUCCESS in 2m 08s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/97544e37a0d44e69b2837735fa28a57a : SUCCESS in 2m 57s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/50f0b648f9e14a6bae62b9986035be3e : SUCCESS in 4m 56s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6316904e193c4a7dbe333857c40cf43d : SUCCESS in 3m 19s","accounts_in_message":[],"_revision_number":1},{"id":"a3f1aef8314d2bfbdf65f664f4048c506f183621","tag":"autogenerated:gerrit:abandon","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2025-08-15 12:22:36.000000000","message":"Abandoned\n\npep8 error is due to the inserted comment that is meant to trigger CI jobs. so the gate is functional, this can be abandoned.","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"c8719511dc08430ad1a1e5dad788455aaba8732b","revisions":{"c8719511dc08430ad1a1e5dad788455aaba8732b":{"kind":"REWORK","_number":1,"created":"2025-08-14 09:45:53.000000000","uploader":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"ref":"refs/changes/78/957378/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.context","ref":"refs/changes/78/957378/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.context refs/changes/78/957378/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.context refs/changes/78/957378/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.context refs/changes/78/957378/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.context refs/changes/78/957378/1"}}},"commit":{"parents":[{"commit":"f23bc418577d36f0e6b0f1bc1d58ff9e68b5a001","subject":"Add build-system to pyproject.toml","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.context/commit/f23bc418577d36f0e6b0f1bc1d58ff9e68b5a001"}]}],"author":{"name":"Elod Illes","email":"elod.illes@est.tech","date":"2025-08-14 09:45:44.000000000","tz":120},"committer":{"name":"Elod Illes","email":"elod.illes@est.tech","date":"2025-08-14 09:45:52.000000000","tz":120},"subject":"DNM: gate health test","message":"DNM: gate health test\n\nChange-Id: Ifeb27e606959b6a8c7a3ae853aff1e2e74e7b5d4\nSigned-off-by: Elod Illes \u003celod.illes@est.tech\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.context/commit/c8719511dc08430ad1a1e5dad788455aaba8732b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.context/commit/c8719511dc08430ad1a1e5dad788455aaba8732b"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
