)]}'
{"id":"openstack%2Foslo.context~969048","triplet_id":"openstack%2Foslo.context~master~I91320224dec622b9a39878d84f55b1a55fc23cc9","project":"openstack/oslo.context","branch":"master","topic":"typing","attention_set":{},"removed_from_attention_set":{"15334":{"account":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"last_update":"2025-12-05 17:25:18.000000000","reason":"\u003cGERRIT_ACCOUNT_15334\u003e replied on the change","reason_account":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}}},"hashtags":[],"change_id":"I91320224dec622b9a39878d84f55b1a55fc23cc9","subject":"Run mypy from tox","status":"MERGED","created":"2025-12-01 13:21:09.000000000","updated":"2025-12-05 18:57:49.000000000","submitted":"2025-12-05 18:57:04.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"969048-typing","meta_rev_id":"cdff1a73d27caf20abd117af5b1f8b3876353449","_number":969048,"virtual_id_number":969048,"owner":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-12-05 18:57:03.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":2,"date":"2025-12-05 17:25:18.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":1,"date":"2025-12-05 17:25:18.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-12-01 14:45:22.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"3884da97d51130ab0b59e8c6a34ecfefafd72ea6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2025-12-01 13:21:09.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"67aeb015c4d7a384c41a7321fcf7808f7deb2ba2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-01 14:45:22.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/9b40306467004e99a504c9f22c61b507\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/d7271734b2684101a1dfacf7e11b4b4b : SUCCESS in 2m 33s\n- requirements-check https://zuul.opendev.org/t/openstack/build/e86611c51ecc499397595f0843aeb5b8 : SUCCESS in 1m 54s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/a0f81c87acca417591de9f6598999af5 : SUCCESS in 1h 22m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/62b5142a857749309804511850e60fde : FAILURE in 2m 58s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/9275d812b37f42c5bb0e3ee36cf3feef : SUCCESS in 2m 27s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/534077fe68494c44abe2d8d6f5c9e906 : SUCCESS in 5m 03s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f8b1ea17e7b74204b0184042b3eb08a1 : SUCCESS in 2m 46s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1081526f650845e8b2b4afac02cef0f7 : SUCCESS in 2m 37s","accounts_in_message":[],"_revision_number":1},{"id":"e52e23023b51d6c8541370f518dd622e392a9611","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2025-12-01 16:42:36.000000000","message":"Patch Set 2: Published edit on patch set 1.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"b59645ab56798fcea12d3e0dda547f83f6434c26","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-01 18:19:12.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/89e3518964444fffa82cd8f0167c2826\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/31a6c09475bf404c94dba3aac69e9f39 : SUCCESS in 1m 33s\n- requirements-check https://zuul.opendev.org/t/openstack/build/f2ed184c061f4a72b6e4888bff5020a2 : SUCCESS in 2m 59s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/308fe7fa23ed481babf5743b09e6d60a : SUCCESS in 1h 34m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/541b0f88389f498aa7a6c509bde2275f : SUCCESS in 3m 29s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/e3c2d95dff604c06aa6690bce472263b : SUCCESS in 2m 55s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/2168c9351bc54dee998661c7cc88838a : SUCCESS in 4m 25s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a7787fd63c0f480f94c6b13e6bc29a2e : SUCCESS in 3m 12s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d07a877e24b94652882456730cbe1a21 : SUCCESS in 2m 08s","accounts_in_message":[],"_revision_number":2},{"id":"9c23761bf5ea98e12d60adec55ff95039240fd0a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2025-12-04 18:12:14.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":3},{"id":"e8bcb77a28ee3a9108371ce0bc7c380590e9edd7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-04 20:03:47.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e7159fb4f7de462ba7d6ae7aff409414\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/2fdee2b3de364422bc30e3ed69cb216a : SUCCESS in 2m 13s\n- requirements-check https://zuul.opendev.org/t/openstack/build/868e0167dbb44ed0b5250e5443666981 : SUCCESS in 2m 09s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/65d4301168064c02aea943b5d42362a8 : SUCCESS in 1h 45m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a4a61c92a8a842b3bf4bc76ea06b30e8 : SUCCESS in 3m 38s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/5fb63c4b8d1c4e00b341850ee7832286 : SUCCESS in 3m 07s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/de68faaa671b4362b29377d121e6be0c : SUCCESS in 5m 09s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d170beab7f484fa5b58f372e6f27ccdf : SUCCESS in 4m 13s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/6e0149a31e7643019f31e85cbb191c7c : SUCCESS in 2m 21s","accounts_in_message":[],"_revision_number":3},{"id":"9544702e2e3a220321596fc19f1fdcc5810accf4","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2025-12-05 17:25:18.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"26d546a084a0539fe2c74365cd3d9b92faf5f447","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-05 17:25:45.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"26a32815ec25ebb90c21f34671bd85b81fb6e3da","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-05 18:57:03.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4bdf89e3395b4e719693b10bd7ee12a7\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/3fdfbfeca78f43d1841ffbc91cb492f5 : SUCCESS in 2m 24s\n- requirements-check https://zuul.opendev.org/t/openstack/build/bdbb7fba76454af08d47e963cce9c6a9 : SUCCESS in 3m 02s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/e828d6b8b7fd4b0287db9983001d039f : SUCCESS in 1h 30m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1f49d64328ad41f281ecf44d732c8539 : SUCCESS in 3m 23s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/1e94a6798cad43ee972c84ccdbc32ab5 : SUCCESS in 2m 36s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/803822ce856e43a7b1c32c918d11b9cc : SUCCESS in 3m 27s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/12d011584a4342ddbd6e3c90a3bb9b76 : SUCCESS in 3m 06s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3490bf845c5540859eee480049b53e31 : SUCCESS in 3m 46s","accounts_in_message":[],"_revision_number":3},{"id":"fbb6880e49b25507a7f73a3eff2636cafb0c3ea8","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-05 18:57:04.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"cdff1a73d27caf20abd117af5b1f8b3876353449","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-05 18:57:49.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/afe558790c1d4f7c9483adde9efa435b\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/42868fd5597a4797b3f5bf7e3b1c57df : SUCCESS in 36s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/054adf8f667d48be9a807eb8cd46e74a : SUCCESS in 37s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"2b7e33c416726230d5e4dcddf8eb1985c35d217f","revisions":{"b11600e42ed0f5dbe775c6bf818582446f19dc7b":{"kind":"REWORK","_number":1,"created":"2025-12-01 13:21:09.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/48/969048/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.context","ref":"refs/changes/48/969048/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.context refs/changes/48/969048/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.context refs/changes/48/969048/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.context refs/changes/48/969048/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.context refs/changes/48/969048/1"}}},"commit":{"parents":[{"commit":"f2d9a6fd21d1eb335da75592fe4fc5a661b251f4","subject":"Remove reference to tag framework","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.context/commit/f2d9a6fd21d1eb335da75592fe4fc5a661b251f4"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2025-12-01 13:18:59.000000000","tz":0},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2025-12-01 13:20:39.000000000","tz":0},"subject":"Run mypy from tox","message":"Run mypy from tox\n\nThis avoids the need to duplicate our dependency list in multiple\nplaces and allows us to take advantage of tox\u0027s dependency management\ninfrastructure, to ensure we always get the latest and greatest version\nof a package allowed by upper-constraints.\n\nWhile here, we also update versions of other pre-commit hooks and\nsimplify the mypy configuration, aligning it with that of other typed\noslo projects.\n\nChange-Id: I91320224dec622b9a39878d84f55b1a55fc23cc9\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.context/commit/b11600e42ed0f5dbe775c6bf818582446f19dc7b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.context/commit/b11600e42ed0f5dbe775c6bf818582446f19dc7b"}]},"branch":"refs/heads/master"},"3f9a70a0a08c1f80a43e076cad90046a74943e05":{"kind":"REWORK","_number":2,"created":"2025-12-01 16:42:36.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/48/969048/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.context","ref":"refs/changes/48/969048/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.context refs/changes/48/969048/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.context refs/changes/48/969048/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.context refs/changes/48/969048/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.context refs/changes/48/969048/2"}}},"commit":{"parents":[{"commit":"f2d9a6fd21d1eb335da75592fe4fc5a661b251f4","subject":"Remove reference to tag framework","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.context/commit/f2d9a6fd21d1eb335da75592fe4fc5a661b251f4"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2025-12-01 13:18:59.000000000","tz":0},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2025-12-01 16:42:36.000000000","tz":0},"subject":"Run mypy from tox","message":"Run mypy from tox\n\nThis avoids the need to duplicate our dependency list in multiple\nplaces and allows us to take advantage of tox\u0027s dependency management\ninfrastructure, to ensure we always get the latest and greatest version\nof a package allowed by upper-constraints.\n\nWhile here, we also update versions of other pre-commit hooks and\nsimplify the mypy configuration, aligning it with that of other typed\noslo projects.\n\nChange-Id: I91320224dec622b9a39878d84f55b1a55fc23cc9\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.context/commit/3f9a70a0a08c1f80a43e076cad90046a74943e05"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.context/commit/3f9a70a0a08c1f80a43e076cad90046a74943e05"}]},"branch":"refs/heads/master"},"2b7e33c416726230d5e4dcddf8eb1985c35d217f":{"kind":"REWORK","_number":3,"created":"2025-12-04 18:12:14.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/48/969048/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.context","ref":"refs/changes/48/969048/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.context refs/changes/48/969048/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.context refs/changes/48/969048/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.context refs/changes/48/969048/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.context refs/changes/48/969048/3"}}},"commit":{"parents":[{"commit":"f2d9a6fd21d1eb335da75592fe4fc5a661b251f4","subject":"Remove reference to tag framework","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.context/commit/f2d9a6fd21d1eb335da75592fe4fc5a661b251f4"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2025-12-01 13:18:59.000000000","tz":0},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2025-12-04 18:11:58.000000000","tz":0},"subject":"Run mypy from tox","message":"Run mypy from tox\n\nThis avoids the need to duplicate our dependency list in multiple\nplaces and allows us to take advantage of tox\u0027s dependency management\ninfrastructure, to ensure we always get the latest and greatest version\nof a package allowed by upper-constraints.\n\nWhile here, we also update versions of other pre-commit hooks and\nsimplify the mypy configuration, aligning it with that of other typed\noslo projects.\n\nChange-Id: I91320224dec622b9a39878d84f55b1a55fc23cc9\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.context/commit/2b7e33c416726230d5e4dcddf8eb1985c35d217f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.context/commit/2b7e33c416726230d5e4dcddf8eb1985c35d217f"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
