)]}'
{"id":"openstack%2Foslo.db~409194","triplet_id":"openstack%2Foslo.db~master~Ic43cb0e0946506a4b7ce9f94ee9eba5e025ed3d5","project":"openstack/oslo.db","branch":"master","topic":"bug/1648818","hashtags":[],"change_id":"Ic43cb0e0946506a4b7ce9f94ee9eba5e025ed3d5","subject":"exc_filters: fix deadlock detection for percona xtradb cluster","status":"MERGED","created":"2016-12-09 15:05:49.000000000","updated":"2016-12-12 09:11:43.000000000","submitted":"2016-12-09 16:55:16.000000000","submitter":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"9df21663953a11a06f9927ed27304ee30d068d25","_number":409194,"virtual_id_number":409194,"owner":{"_account_id":6849,"name":"Roman Podoliaka","email":"roman.podoliaka@gmail.com","username":"rpodolyaka"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"all":[{"value":2,"date":"2016-12-09 16:55:16.000000000","post_submit":true,"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":0,"_account_id":11816,"name":"mike bayer","display_name":"Mike Bayer","email":"mike_mp@zzzcomputing.com","username":"zzzeek","status":"Red Hat"},{"value":0,"_account_id":2813,"name":"Mehdi Abaakouk","email":"sileht@sileht.net","username":"sileht"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11816,"name":"mike bayer","display_name":"Mike Bayer","email":"mike_mp@zzzcomputing.com","username":"zzzeek","status":"Red Hat"},"all":[{"value":0,"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":2,"date":"2016-12-09 15:29:02.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11816,"name":"mike bayer","display_name":"Mike Bayer","email":"mike_mp@zzzcomputing.com","username":"zzzeek","status":"Red Hat"},{"value":2,"date":"2016-12-09 15:48:10.000000000","_account_id":2813,"name":"Mehdi Abaakouk","email":"sileht@sileht.net","username":"sileht"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":2813,"name":"Mehdi Abaakouk","email":"sileht@sileht.net","username":"sileht"},"all":[{"value":0,"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":0,"_account_id":11816,"name":"mike bayer","display_name":"Mike Bayer","email":"mike_mp@zzzcomputing.com","username":"zzzeek","status":"Red Hat"},{"value":1,"date":"2016-12-09 15:48:10.000000000","_account_id":2813,"name":"Mehdi Abaakouk","email":"sileht@sileht.net","username":"sileht"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":0,"_account_id":11816,"name":"mike bayer","display_name":"Mike Bayer","email":"mike_mp@zzzcomputing.com","username":"zzzeek","status":"Red Hat"},{"value":0,"_account_id":2813,"name":"Mehdi Abaakouk","email":"sileht@sileht.net","username":"sileht"}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":3,"name":"Jenkins","username":"jenkins"},{"_account_id":2813,"name":"Mehdi Abaakouk","email":"sileht@sileht.net","username":"sileht"},{"_account_id":11816,"name":"mike bayer","display_name":"Mike Bayer","email":"mike_mp@zzzcomputing.com","username":"zzzeek","status":"Red Hat"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2016-12-09 15:29:02.000000000","updated_by":{"_account_id":11816,"name":"mike bayer","display_name":"Mike Bayer","email":"mike_mp@zzzcomputing.com","username":"zzzeek","status":"Red Hat"},"reviewer":{"_account_id":11816,"name":"mike bayer","display_name":"Mike Bayer","email":"mike_mp@zzzcomputing.com","username":"zzzeek","status":"Red Hat"},"state":"REVIEWER"},{"updated":"2016-12-09 15:48:10.000000000","updated_by":{"_account_id":2813,"name":"Mehdi Abaakouk","email":"sileht@sileht.net","username":"sileht"},"reviewer":{"_account_id":2813,"name":"Mehdi Abaakouk","email":"sileht@sileht.net","username":"sileht"},"state":"REVIEWER"},{"updated":"2016-12-09 16:55:16.000000000","updated_by":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"reviewer":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"state":"REVIEWER"}],"messages":[{"id":"f00dd30cd3dda79bc5341f2f4ac7af4d759c3a86","author":{"_account_id":6849,"name":"Roman Podoliaka","email":"roman.podoliaka@gmail.com","username":"rpodolyaka"},"date":"2016-12-09 15:05:49.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"9e5d8663250abea9c0e1aeb069d41872337935a6","author":{"_account_id":11816,"name":"mike bayer","display_name":"Mike Bayer","email":"mike_mp@zzzcomputing.com","username":"zzzeek","status":"Red Hat"},"date":"2016-12-09 15:29:02.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"5473b65366bcdb48ee6958bcc4919e89d16a71db","author":{"_account_id":2813,"name":"Mehdi Abaakouk","email":"sileht@sileht.net","username":"sileht"},"date":"2016-12-09 15:48:10.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"343b0389b2f28ede888231de7053d21831a9ad5d","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-12-09 16:00:05.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- gate-oslo.db-docs-ubuntu-xenial http://docs-draft.openstack.org/94/409194/1/check/gate-oslo.db-docs-ubuntu-xenial/e9f8998//doc/build/html/ : SUCCESS in 1m 51s\n- gate-oslo.db-pep8-ubuntu-xenial http://logs.openstack.org/94/409194/1/check/gate-oslo.db-pep8-ubuntu-xenial/1618afa/ : SUCCESS in 2m 54s\n- gate-oslo.db-python27-db-ubuntu-xenial http://logs.openstack.org/94/409194/1/check/gate-oslo.db-python27-db-ubuntu-xenial/dede773/ : SUCCESS in 3m 32s\n- gate-oslo.db-python34-db http://logs.openstack.org/94/409194/1/check/gate-oslo.db-python34-db/a9c0266/ : SUCCESS in 2m 35s\n- gate-oslo.db-python35-db http://logs.openstack.org/94/409194/1/check/gate-oslo.db-python35-db/cba37e9/ : SUCCESS in 2m 20s\n- gate-tempest-dsvm-neutron-src-oslo.db-ubuntu-xenial http://logs.openstack.org/94/409194/1/check/gate-tempest-dsvm-neutron-src-oslo.db-ubuntu-xenial/7133a65/ : SUCCESS in 52m 00s\n- gate-tempest-dsvm-neutron-src-oslo.db-ubuntu-xenial-newton http://logs.openstack.org/94/409194/1/check/gate-tempest-dsvm-neutron-src-oslo.db-ubuntu-xenial-newton/c60f2b3/ : FAILURE in 15m 52s (non-voting)\n- gate-oslo.db-tox-db-mysql-python-ubuntu-xenial http://logs.openstack.org/94/409194/1/check/gate-oslo.db-tox-db-mysql-python-ubuntu-xenial/3fcf43b/ : SUCCESS in 2m 33s","accounts_in_message":[],"_revision_number":1},{"id":"c6f37e1fb725a6e36670d8a526e36074fd5d3ed9","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-12-09 16:00:16.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.\nhttp://status.openstack.org/zuul/","accounts_in_message":[],"_revision_number":1},{"id":"c279e58def32f74685672c2ccbec409fd5c12a2a","date":"2016-12-09 16:55:16.000000000","message":"Change has been successfully merged into the git repository by Jenkins","accounts_in_message":[],"_revision_number":1},{"id":"499aa19a43fedce28ef1b542885358c62b6309d1","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-12-09 16:55:16.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- gate-oslo.db-docs-ubuntu-xenial http://docs-draft.openstack.org/94/409194/1/gate/gate-oslo.db-docs-ubuntu-xenial/0a1abb8//doc/build/html/ : SUCCESS in 2m 31s\n- gate-oslo.db-pep8-ubuntu-xenial http://logs.openstack.org/94/409194/1/gate/gate-oslo.db-pep8-ubuntu-xenial/8a4b880/ : SUCCESS in 2m 33s\n- gate-oslo.db-python27-db-ubuntu-xenial http://logs.openstack.org/94/409194/1/gate/gate-oslo.db-python27-db-ubuntu-xenial/3d518bc/ : SUCCESS in 2m 31s\n- gate-oslo.db-python34-db http://logs.openstack.org/94/409194/1/gate/gate-oslo.db-python34-db/734c787/ : SUCCESS in 2m 27s\n- gate-oslo.db-python35-db http://logs.openstack.org/94/409194/1/gate/gate-oslo.db-python35-db/2ea61b5/ : SUCCESS in 3m 28s\n- gate-tempest-dsvm-neutron-src-oslo.db-ubuntu-xenial http://logs.openstack.org/94/409194/1/gate/gate-tempest-dsvm-neutron-src-oslo.db-ubuntu-xenial/cc0d256/ : SUCCESS in 54m 14s\n- gate-oslo.db-tox-db-mysql-python-ubuntu-xenial http://logs.openstack.org/94/409194/1/gate/gate-oslo.db-tox-db-mysql-python-ubuntu-xenial/28e1fc1/ : SUCCESS in 3m 36s","accounts_in_message":[],"_revision_number":1},{"id":"d36af4aafa85f6a0327245c68a4577ab6bfef270","author":{"_account_id":6849,"name":"Roman Podoliaka","email":"roman.podoliaka@gmail.com","username":"rpodolyaka"},"date":"2016-12-12 09:11:43.000000000","message":"Patch Set 1: Cherry Picked\n\nThis patchset was cherry picked to branch stable/newton as commit f26d0e09738da6c4faa9ddc08767779289931b42","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"7c8f0f48bfb941dc1147768493bbaf81bb437d58","revisions":{"7c8f0f48bfb941dc1147768493bbaf81bb437d58":{"kind":"REWORK","_number":1,"created":"2016-12-09 15:05:49.000000000","uploader":{"_account_id":6849,"name":"Roman Podoliaka","email":"roman.podoliaka@gmail.com","username":"rpodolyaka"},"ref":"refs/changes/94/409194/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.db","ref":"refs/changes/94/409194/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.db refs/changes/94/409194/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.db refs/changes/94/409194/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.db refs/changes/94/409194/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.db refs/changes/94/409194/1"}}},"commit":{"parents":[{"commit":"6bc62b047b02efd8fd6186ec0935faad975f0ffe","subject":"Merge \"Strip prefix `migrate_` in parameter `migrate_engine`\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.db/commit/6bc62b047b02efd8fd6186ec0935faad975f0ffe"}]}],"author":{"name":"Roman Podoliaka","email":"rpodolyaka@mirantis.com","date":"2016-12-09 13:44:42.000000000","tz":120},"committer":{"name":"Roman Podoliaka","email":"rpodolyaka@mirantis.com","date":"2016-12-09 13:52:08.000000000","tz":120},"subject":"exc_filters: fix deadlock detection for percona xtradb cluster","message":"exc_filters: fix deadlock detection for percona xtradb cluster\n\npercona xtradb cluster server produces an error like:\n\n(1213, u\u0027WSREP detected deadlock/conflict and aborted the transaction.\n         Try restarting the transaction\u0027)\n\nwhich is currently not detected properly and, thus, not retried.\n\nCloses-Bug: #1648818\n\nChange-Id: Ic43cb0e0946506a4b7ce9f94ee9eba5e025ed3d5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.db/commit/7c8f0f48bfb941dc1147768493bbaf81bb437d58"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.db/commit/7c8f0f48bfb941dc1147768493bbaf81bb437d58"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
