)]}'
{"id":"openstack%2Foslo.db~500472","triplet_id":"openstack%2Foslo.db~master~Icb2c88dc18843cbf2f5be08d635d5ba10f36f360","project":"openstack/oslo.db","branch":"master","topic":"oslo-debt-cleanup","hashtags":[],"change_id":"Icb2c88dc18843cbf2f5be08d635d5ba10f36f360","subject":"Remove deprecated class DbMigrationError since we already have DBMigrationError","status":"MERGED","created":"2017-09-04 08:06:33.000000000","updated":"2019-12-17 20:44:50.000000000","submitted":"2019-12-17 20:43:11.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"500472-1576615391058-fead5b33","meta_rev_id":"934c8f618b4b509cef2405264d3ba55f8863cd1d","_number":500472,"virtual_id_number":500472,"owner":{"_account_id":17549,"name":"Yaguo Zhou","email":"zhouyaguo@unionpay.com","username":"zhouyaguo"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":0,"_account_id":9796,"name":"ChangBo Guo","email":"glongwave@gmail.com","username":"gcb"},{"value":0,"_account_id":25254,"name":"Nguyen Van Trung","email":"trungnvfet@outlook.com","username":"trungnv"},{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-12-17 20:43:10.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":17549,"name":"Yaguo Zhou","email":"zhouyaguo@unionpay.com","username":"zhouyaguo"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"recommended":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"disliked":{"_account_id":25254,"name":"Nguyen Van Trung","email":"trungnvfet@outlook.com","username":"trungnv"},"all":[{"value":1,"date":"2019-12-17 17:16:46.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":0,"date":"2019-12-17 17:16:46.000000000","permitted_voting_range":{"min":0,"max":1},"_account_id":9796,"name":"ChangBo Guo","email":"glongwave@gmail.com","username":"gcb"},{"value":-1,"date":"2019-12-17 17:16:46.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":25254,"name":"Nguyen Van Trung","email":"trungnvfet@outlook.com","username":"trungnv"},{"value":2,"date":"2019-12-17 17:17:33.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":2,"date":"2019-12-17 17:48:16.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":17549,"name":"Yaguo Zhou","email":"zhouyaguo@unionpay.com","username":"zhouyaguo"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":-1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":0,"_account_id":9796,"name":"ChangBo Guo","email":"glongwave@gmail.com","username":"gcb"},{"value":0,"_account_id":25254,"name":"Nguyen Van Trung","email":"trungnvfet@outlook.com","username":"trungnv"},{"value":1,"date":"2019-12-17 17:17:33.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":17549,"name":"Yaguo Zhou","email":"zhouyaguo@unionpay.com","username":"zhouyaguo"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":0,"_account_id":9796,"name":"ChangBo Guo","email":"glongwave@gmail.com","username":"gcb"},{"value":0,"_account_id":25254,"name":"Nguyen Van Trung","email":"trungnvfet@outlook.com","username":"trungnv"},{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":17549,"name":"Yaguo Zhou","email":"zhouyaguo@unionpay.com","username":"zhouyaguo"}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":3,"name":"Jenkins","username":"jenkins"},{"_account_id":9796,"name":"ChangBo Guo","email":"glongwave@gmail.com","username":"gcb"},{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":17549,"name":"Yaguo Zhou","email":"zhouyaguo@unionpay.com","username":"zhouyaguo"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25254,"name":"Nguyen Van Trung","email":"trungnvfet@outlook.com","username":"trungnv"},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2017-09-04 09:29:07.000000000","updated_by":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"reviewer":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"state":"REVIEWER"},{"updated":"2017-09-06 01:06:11.000000000","updated_by":{"_account_id":25254,"name":"Nguyen Van Trung","email":"trungnvfet@outlook.com","username":"trungnv"},"reviewer":{"_account_id":25254,"name":"Nguyen Van Trung","email":"trungnvfet@outlook.com","username":"trungnv"},"state":"REVIEWER"},{"updated":"2017-12-13 12:55:36.000000000","updated_by":{"_account_id":9796,"name":"ChangBo Guo","email":"glongwave@gmail.com","username":"gcb"},"reviewer":{"_account_id":9796,"name":"ChangBo Guo","email":"glongwave@gmail.com","username":"gcb"},"state":"REVIEWER"},{"updated":"2017-12-29 15:36:26.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2019-12-17 17:17:33.000000000","updated_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"reviewer":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"state":"REVIEWER"},{"updated":"2019-12-17 17:48:16.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2019-12-17 20:43:10.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"7530da41244745f3a3d1dded5092d4f4ba3f00ff","author":{"_account_id":17549,"name":"Yaguo Zhou","email":"zhouyaguo@unionpay.com","username":"zhouyaguo"},"date":"2017-09-04 08:06:33.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"7ae279a442975ea425d13c4ef4fa70593d892a4d","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2017-09-04 09:29:07.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- gate-oslo.db-docs-ubuntu-xenial http://docs-draft.openstack.org/72/500472/1/check/gate-oslo.db-docs-ubuntu-xenial/9b54562//doc/build/html/ : SUCCESS in 3m 09s\n- gate-oslo.db-pep8-ubuntu-xenial http://logs.openstack.org/72/500472/1/check/gate-oslo.db-pep8-ubuntu-xenial/5eb6fe4/ : SUCCESS in 3m 04s\n- gate-oslo.db-python27-ubuntu-xenial http://logs.openstack.org/72/500472/1/check/gate-oslo.db-python27-ubuntu-xenial/7d7f11c/ : SUCCESS in 3m 23s\n- gate-oslo.db-python35 http://logs.openstack.org/72/500472/1/check/gate-oslo.db-python35/59b7632/ : SUCCESS in 3m 17s\n- gate-tempest-dsvm-neutron-src-oslo.db-ubuntu-xenial http://logs.openstack.org/72/500472/1/check/gate-tempest-dsvm-neutron-src-oslo.db-ubuntu-xenial/bee4714/ : SUCCESS in 1h 21m 32s\n- gate-oslo.db-tox-mysql-python-ubuntu-xenial http://logs.openstack.org/72/500472/1/check/gate-oslo.db-tox-mysql-python-ubuntu-xenial/5592f01/ : SUCCESS in 3m 57s","accounts_in_message":[],"_revision_number":1},{"id":"4d0bd92d15a1c0686d09846817c7aa6d643f523a","author":{"_account_id":9796,"name":"ChangBo Guo","email":"glongwave@gmail.com","username":"gcb"},"date":"2017-09-05 09:13:53.000000000","message":"Patch Set 1: Code-Review-1\n\nNeed remove its usage before removing it, see http://codesearch.openstack.org/?q\u003dDbMigrationError\u0026i\u003dnope\u0026files\u003d\u0026repos\u003d","accounts_in_message":[],"_revision_number":1},{"id":"152233dd9a888a1437f1ae11d95b47809640d060","author":{"_account_id":17549,"name":"Yaguo Zhou","email":"zhouyaguo@unionpay.com","username":"zhouyaguo"},"date":"2017-09-05 12:05:39.000000000","message":"Patch Set 1:\n\n\u003e Need remove its usage before removing it, see http://codesearch.openstack.org/?q\u003dDbMigrationError\u0026i\u003dnope\u0026files\u003d\u0026repos\u003d\n\nOK, got it, thanks, ChangBo.","accounts_in_message":[],"_revision_number":1},{"id":"1f3ab4780486d9954c3f7ed679cea79bd45fc43d","author":{"_account_id":25254,"name":"Nguyen Van Trung","email":"trungnvfet@outlook.com","username":"trungnv"},"date":"2017-09-06 01:06:11.000000000","message":"Patch Set 1: Code-Review-1\n\nAgree with ChangBo, This change need to some notification into openstack project before delete them.","accounts_in_message":[],"_revision_number":1},{"id":"dce6db85a7e0a697de0a23a3292bfdd1e3a1e4c7","author":{"_account_id":17549,"name":"Yaguo Zhou","email":"zhouyaguo@unionpay.com","username":"zhouyaguo"},"date":"2017-09-06 15:13:52.000000000","message":"Patch Set 1:\n\n\u003e Agree with ChangBo, This change need to some notification into\n \u003e openstack project before delete them.\n\nAgreed, thanks. I\u0027ll recheck this after making some commits to the consuming projects.","accounts_in_message":[],"_revision_number":1},{"id":"8436f5bc91e407e134eb62cfd9cf703fab7b0253","author":{"_account_id":17549,"name":"Yaguo Zhou","email":"zhouyaguo@unionpay.com","username":"zhouyaguo"},"date":"2017-09-08 05:02:37.000000000","message":"Patch Set 1:\n\nDepends on https://review.openstack.org/#/c/501941, I\u0027ll give a new patch set after the merge of https://review.openstack.org/#/c/501941","accounts_in_message":[],"_revision_number":1},{"id":"7f0b10938ee247a9595956a45d3144fadba853a6","author":{"_account_id":9796,"name":"ChangBo Guo","email":"glongwave@gmail.com","username":"gcb"},"date":"2017-12-13 12:55:36.000000000","message":"Patch Set 1: -Code-Review\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"7f197eca9aec2234372f50faf65eb0007c6cc474","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2017-12-13 14:04:27.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/72/500472/1/check/openstack-tox-pep8/1089094/ : SUCCESS in 4m 39s\n- openstack-tox-py27 http://logs.openstack.org/72/500472/1/check/openstack-tox-py27/87321d1/ : SUCCESS in 4m 53s\n- openstack-tox-py35 http://logs.openstack.org/72/500472/1/check/openstack-tox-py35/5c80428/ : SUCCESS in 6m 31s\n- build-openstack-sphinx-docs http://logs.openstack.org/72/500472/1/check/build-openstack-sphinx-docs/e5086e8/html/ : SUCCESS in 4m 17s\n- legacy-tempest-dsvm-neutron-src http://logs.openstack.org/72/500472/1/check/legacy-tempest-dsvm-neutron-src/c31f9fe/ : SUCCESS in 1h 03m 55s\n- oslo.db-tox-mysql-python http://logs.openstack.org/72/500472/1/check/oslo.db-tox-mysql-python/c55b86d/ : SUCCESS in 5m 23s","accounts_in_message":[],"_revision_number":1},{"id":"5e61239ffc3a734d1788d4589bcef9bfe56fac3b","author":{"_account_id":9796,"name":"ChangBo Guo","email":"glongwave@gmail.com","username":"gcb"},"date":"2017-12-28 05:58:05.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"84f24a1a56d8e08889fec74ecf061b9326a726dc","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2017-12-28 07:22:26.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/72/500472/2/check/openstack-tox-pep8/7886732/ : SUCCESS in 5m 39s\n- openstack-tox-py27 http://logs.openstack.org/72/500472/2/check/openstack-tox-py27/fc123ea/ : SUCCESS in 3m 53s\n- openstack-tox-py35 http://logs.openstack.org/72/500472/2/check/openstack-tox-py35/3aae9d6/ : SUCCESS in 3m 44s\n- build-openstack-sphinx-docs http://logs.openstack.org/72/500472/2/check/build-openstack-sphinx-docs/b6af18c/html/ : SUCCESS in 3m 42s\n- legacy-tempest-dsvm-neutron-src http://logs.openstack.org/72/500472/2/check/legacy-tempest-dsvm-neutron-src/79900ff/ : SUCCESS in 1h 21m 59s","accounts_in_message":[],"_revision_number":2},{"id":"b9f872abbb266dfea7cacd6150b09eddce6ec139","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2017-12-29 15:36:26.000000000","message":"Patch Set 2: Code-Review+1\n\nLooks like usage of the old exception have been cleaned up in active projects.","accounts_in_message":[],"_revision_number":2},{"id":"69ef5bab93fe7aff0f78df608e2836f81634b3d7","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2019-12-17 17:16:46.000000000","message":"Patch Set 3: Patch Set 2 was rebased","accounts_in_message":[],"_revision_number":3},{"id":"7116b2ca2a08431082a396a83afbdcd5094f0813","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2019-12-17 17:17:33.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1\n\ncodesearch.o.o doesn\u0027t turn up any relevant references to this now. Time to merge.","accounts_in_message":[],"_revision_number":3},{"id":"f5967ae335589162ddf58ffd7a764b775bfd77ed","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2019-12-17 17:48:16.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"6f5c192be533e8ddaec1dc33f72c146938507e9c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-17 18:55:11.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tempest-full https://zuul.opendev.org/t/openstack/build/300aa8cf695040aca4b17475e22daeac : SUCCESS in 1h 30m 15s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/b763a790fa8944ef8e43ce761b8339ea : SUCCESS in 1h 30m 13s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/08cdecb0626b4fbba4c99140ee34b724 : SUCCESS in 6m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d7c46ea2f46242feb17290d1b45b8c73 : SUCCESS in 7m 56s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/78487d94f6dd45da88eda7d6cc7dc116 : SUCCESS in 8m 11s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/df4fab3555c54bbead56f797e9d5704e : SUCCESS in 5m 47s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/029a48f5b4d7446ca736da8d111684cb : SUCCESS in 6m 54s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cb4b84c22fef455f90d229ee0eb21d3b : SUCCESS in 6m 14s","accounts_in_message":[],"_revision_number":3},{"id":"8912f2157e7fac2d444063fd0dffeb30234cb3e7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-17 18:55:23.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"97f264379785c1c051bc2825c7a99225078d06e5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-17 20:43:10.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- tempest-full https://zuul.opendev.org/t/openstack/build/ffae41f1620b4e67bf3529dd4a86c41b : SUCCESS in 1h 40m 29s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/525d283825e44c6db0fedf5eaeaba197 : SUCCESS in 1h 22m 02s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/83f52cc933744f51947673deda12beb9 : SUCCESS in 5m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e07debe6ea4948769b21422eaa09e938 : SUCCESS in 5m 29s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/15d9c28d065b41dea45fad588f588991 : SUCCESS in 5m 45s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b1a208a5e6a74967a33ae1e3a041fa0a : SUCCESS in 10m 31s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/cc0a5acf4a5f4d83abd6a312ec489fe6 : SUCCESS in 13m 33s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3235a963ef0046e09705c5f257eaf375 : SUCCESS in 6m 51s","accounts_in_message":[],"_revision_number":3},{"id":"c0d06c2729295f509bc1b456e0e474bc6b2e651c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-17 20:43:11.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3},{"id":"8eff5dcc5ce54796d741888b3585bec5a53127d3","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-17 20:44:50.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e2676091f8744094a97b4513e45cbfe5 : SUCCESS in 1m 17s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"b4b377b713faf6109605987b80a17a327bd8de0c","revisions":{"dfd76e2e315732a71810f405d863c022bdd7878c":{"kind":"REWORK","_number":1,"created":"2017-09-04 08:06:33.000000000","uploader":{"_account_id":17549,"name":"Yaguo Zhou","email":"zhouyaguo@unionpay.com","username":"zhouyaguo"},"ref":"refs/changes/72/500472/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.db","ref":"refs/changes/72/500472/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.db refs/changes/72/500472/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.db refs/changes/72/500472/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.db refs/changes/72/500472/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.db refs/changes/72/500472/1"}}},"commit":{"parents":[{"commit":"e9a9701d54ea1b708e7fc3ef68e75eb288059632","subject":"Merge \"Replace six.iteritems() with .items()\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.db/commit/e9a9701d54ea1b708e7fc3ef68e75eb288059632"}]}],"author":{"name":"Yaguo Zhou","email":"ygzhou@sysnew.com","date":"2017-09-04 08:01:07.000000000","tz":480},"committer":{"name":"zhouyaguo","email":"zhouyaguo@unionpay.com","date":"2017-09-04 08:06:10.000000000","tz":480},"subject":"Remove deprecated class DbMigrationError since we already have DBMigrationError","message":"Remove deprecated class DbMigrationError since we already have DBMigrationError\n\nChange-Id: Icb2c88dc18843cbf2f5be08d635d5ba10f36f360\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.db/commit/dfd76e2e315732a71810f405d863c022bdd7878c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.db/commit/dfd76e2e315732a71810f405d863c022bdd7878c"}]},"branch":"refs/heads/master"},"b47a4bf5b209d1ea9cac859ae9bd298a1764206e":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2017-12-28 05:58:05.000000000","uploader":{"_account_id":9796,"name":"ChangBo Guo","email":"glongwave@gmail.com","username":"gcb"},"ref":"refs/changes/72/500472/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.db","ref":"refs/changes/72/500472/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.db refs/changes/72/500472/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.db refs/changes/72/500472/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.db refs/changes/72/500472/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.db refs/changes/72/500472/2"}}},"commit":{"parents":[{"commit":"7c2eb8ad9eede513eda721aa2e53601c05136274","subject":"Merge \"Updated from global requirements\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.db/commit/7c2eb8ad9eede513eda721aa2e53601c05136274"}]}],"author":{"name":"Yaguo Zhou","email":"ygzhou@sysnew.com","date":"2017-09-04 08:01:07.000000000","tz":480},"committer":{"name":"ChangBo Guo(gcb)","email":"glongwave@gmail.com","date":"2017-12-28 05:58:05.000000000","tz":0},"subject":"Remove deprecated class DbMigrationError since we already have DBMigrationError","message":"Remove deprecated class DbMigrationError since we already have DBMigrationError\n\nChange-Id: Icb2c88dc18843cbf2f5be08d635d5ba10f36f360\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.db/commit/b47a4bf5b209d1ea9cac859ae9bd298a1764206e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.db/commit/b47a4bf5b209d1ea9cac859ae9bd298a1764206e"}]},"branch":"refs/heads/master"},"b4b377b713faf6109605987b80a17a327bd8de0c":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2019-12-17 17:16:46.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/72/500472/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.db","ref":"refs/changes/72/500472/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.db refs/changes/72/500472/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.db refs/changes/72/500472/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.db refs/changes/72/500472/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.db refs/changes/72/500472/3"}}},"commit":{"parents":[{"commit":"fce052b2af28001f4269c0790e0d1c4464813675","subject":"Use regex to compare SQL strings with IN","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.db/commit/fce052b2af28001f4269c0790e0d1c4464813675"}]}],"author":{"name":"Yaguo Zhou","email":"ygzhou@sysnew.com","date":"2017-09-04 08:01:07.000000000","tz":480},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2019-12-17 17:16:46.000000000","tz":0},"subject":"Remove deprecated class DbMigrationError since we already have DBMigrationError","message":"Remove deprecated class DbMigrationError since we already have DBMigrationError\n\nChange-Id: Icb2c88dc18843cbf2f5be08d635d5ba10f36f360\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.db/commit/b4b377b713faf6109605987b80a17a327bd8de0c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.db/commit/b4b377b713faf6109605987b80a17a327bd8de0c"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
