)]}'
{"id":"openstack%2Foslo.db~650520","triplet_id":"openstack%2Foslo.db~master~I89754e9e4b42c45259729b0b0bb0a4bf464a981c","project":"openstack/oslo.db","branch":"master","hashtags":[],"change_id":"I89754e9e4b42c45259729b0b0bb0a4bf464a981c","subject":"Removing deprecated min_pool_size","status":"MERGED","created":"2019-04-06 19:01:27.000000000","updated":"2019-04-24 14:35:10.000000000","submitted":"2019-04-24 14:35:09.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"650520-1556116510278-11ac1cc1","meta_rev_id":"e96561ba246d40b9e50fc4ca09e48593b277bd92","_number":650520,"virtual_id_number":650520,"owner":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},{"value":0,"_account_id":9796,"name":"ChangBo Guo","email":"glongwave@gmail.com","username":"gcb"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":2,"date":"2019-04-24 14:35:09.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},"all":[{"value":2,"date":"2019-04-23 20:39:29.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},{"value":2,"date":"2019-04-24 12:40:51.000000000","_account_id":9796,"name":"ChangBo Guo","email":"glongwave@gmail.com","username":"gcb"},{"value":1,"date":"2019-04-23 14:00:37.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9796,"name":"ChangBo Guo","email":"glongwave@gmail.com","username":"gcb"},"all":[{"value":0,"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},{"value":1,"date":"2019-04-24 12:40:51.000000000","_account_id":9796,"name":"ChangBo Guo","email":"glongwave@gmail.com","username":"gcb"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},{"value":0,"_account_id":9796,"name":"ChangBo Guo","email":"glongwave@gmail.com","username":"gcb"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},{"_account_id":9796,"name":"ChangBo Guo","email":"glongwave@gmail.com","username":"gcb"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-04-23 14:00:37.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2019-04-23 20:39:29.000000000","updated_by":{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},"reviewer":{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},"state":"REVIEWER"},{"updated":"2019-04-24 12:40:51.000000000","updated_by":{"_account_id":9796,"name":"ChangBo Guo","email":"glongwave@gmail.com","username":"gcb"},"reviewer":{"_account_id":9796,"name":"ChangBo Guo","email":"glongwave@gmail.com","username":"gcb"},"state":"REVIEWER"},{"updated":"2019-04-24 14:35:09.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"48bea1f3693d2606cd4d6bf459ed65bec67d2091","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2019-04-06 19:01:27.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"af88e7a0d54bec99c4577a805a931cf14b8485bf","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2019-04-06 19:55:51.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"788635edf172debfa06622de8466de8592f91afa","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-06 21:39:49.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tempest-full http://logs.openstack.org/20/650520/2/check/tempest-full/f6c1161/ : SUCCESS in 1h 40m 26s\n- tempest-full-py3 http://logs.openstack.org/20/650520/2/check/tempest-full-py3/4087657/ : SUCCESS in 1h 39m 27s\n- openstack-tox-lower-constraints http://logs.openstack.org/20/650520/2/check/openstack-tox-lower-constraints/64e2112/ : SUCCESS in 5m 35s\n- openstack-tox-pep8 http://logs.openstack.org/20/650520/2/check/openstack-tox-pep8/9926232/ : SUCCESS in 4m 38s\n- openstack-tox-py27 http://logs.openstack.org/20/650520/2/check/openstack-tox-py27/b2c4d89/ : SUCCESS in 9m 51s\n- openstack-tox-py35 http://logs.openstack.org/20/650520/2/check/openstack-tox-py35/863397c/ : SUCCESS in 6m 12s\n- openstack-tox-py36 http://logs.openstack.org/20/650520/2/check/openstack-tox-py36/f10d986/ : SUCCESS in 6m 10s\n- openstack-tox-docs http://logs.openstack.org/20/650520/2/check/openstack-tox-docs/6231f98/html/ : SUCCESS in 4m 16s\n- build-openstack-releasenotes http://logs.openstack.org/20/650520/2/check/build-openstack-releasenotes/7057fc8/html/ : SUCCESS in 5m 14s","accounts_in_message":[],"_revision_number":2},{"id":"1f800709ea4cb81064434dbe1da9c1f455f89465","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2019-04-23 14:00:37.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"de19db386c07a085515c684593117738983e80aa","author":{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},"date":"2019-04-23 20:39:29.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"9cc92b93c0c53a097bfb78d8c2c2e940865f3404","author":{"_account_id":9796,"name":"ChangBo Guo","email":"glongwave@gmail.com","username":"gcb"},"date":"2019-04-24 12:40:51.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"693b0eb5f6bd2231ed02111bd2fdd11ea4ad4b18","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-24 12:41:03.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"be7cb8b35f6309ba4abecf7d21073a4d0532f927","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-24 14:35:09.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- tempest-full http://logs.openstack.org/20/650520/2/gate/tempest-full/5e9c09e/ : SUCCESS in 1h 46m 16s\n- tempest-full-py3 http://logs.openstack.org/20/650520/2/gate/tempest-full-py3/a47b9db/ : SUCCESS in 1h 23m 50s\n- openstack-tox-lower-constraints http://logs.openstack.org/20/650520/2/gate/openstack-tox-lower-constraints/d7159ff/ : SUCCESS in 5m 21s\n- openstack-tox-pep8 http://logs.openstack.org/20/650520/2/gate/openstack-tox-pep8/fed7ae8/ : SUCCESS in 4m 41s\n- openstack-tox-py27 http://logs.openstack.org/20/650520/2/gate/openstack-tox-py27/a0cee0d/ : SUCCESS in 9m 07s\n- openstack-tox-py35 http://logs.openstack.org/20/650520/2/gate/openstack-tox-py35/367768a/ : SUCCESS in 4m 58s\n- openstack-tox-py36 http://logs.openstack.org/20/650520/2/gate/openstack-tox-py36/b50d900/ : SUCCESS in 4m 48s\n- openstack-tox-docs http://logs.openstack.org/20/650520/2/gate/openstack-tox-docs/4d68e54/html/ : SUCCESS in 4m 58s\n- build-openstack-releasenotes http://logs.openstack.org/20/650520/2/gate/build-openstack-releasenotes/f1277eb/html/ : SUCCESS in 4m 44s","accounts_in_message":[],"_revision_number":2},{"id":"e96561ba246d40b9e50fc4ca09e48593b277bd92","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-24 14:35:10.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"6cfdc11ea1ed7d8873f91683d6ea1be95fdff89e","revisions":{"e1baec0f79c147443b0cdad189022c7ebeabd8f4":{"kind":"REWORK","_number":1,"created":"2019-04-06 19:01:27.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/20/650520/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.db","ref":"refs/changes/20/650520/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.db refs/changes/20/650520/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.db refs/changes/20/650520/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.db refs/changes/20/650520/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.db refs/changes/20/650520/1"}}},"commit":{"parents":[{"commit":"57333e902a3747715b2f13ea63477cc4aedc2129","subject":"Bump psycopg lower-constraint to 2.7","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.db/commit/57333e902a3747715b2f13ea63477cc4aedc2129"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2019-04-06 18:59:42.000000000","tz":-420},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2019-04-06 19:01:20.000000000","tz":-420},"subject":"Removing deprecated min_pool_size","message":"Removing deprecated min_pool_size\n\nThis is a folllow up to\nIb07b7b4200b5cf3a34b8eed441e6754da7d446cf\n\nRemoving min_pool_size as this setting has not\ndone anything for a very long time. Removing\nthis variable helps make the code cleaner,\nas well removes something that can add confusion\nfor operators.\n\nChange-Id: I89754e9e4b42c45259729b0b0bb0a4bf464a981c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.db/commit/e1baec0f79c147443b0cdad189022c7ebeabd8f4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.db/commit/e1baec0f79c147443b0cdad189022c7ebeabd8f4"}]},"branch":"refs/heads/master"},"6cfdc11ea1ed7d8873f91683d6ea1be95fdff89e":{"kind":"REWORK","_number":2,"created":"2019-04-06 19:55:51.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/20/650520/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.db","ref":"refs/changes/20/650520/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.db refs/changes/20/650520/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.db refs/changes/20/650520/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.db refs/changes/20/650520/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.db refs/changes/20/650520/2"}}},"commit":{"parents":[{"commit":"57333e902a3747715b2f13ea63477cc4aedc2129","subject":"Bump psycopg lower-constraint to 2.7","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.db/commit/57333e902a3747715b2f13ea63477cc4aedc2129"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2019-04-06 18:59:42.000000000","tz":-420},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2019-04-06 19:54:28.000000000","tz":-420},"subject":"Removing deprecated min_pool_size","message":"Removing deprecated min_pool_size\n\nThe database option min_pool_size is not actually\nused in oslo.db.\n\nThis is a folllow up to\nIb07b7b4200b5cf3a34b8eed441e6754da7d446cf\n\nChange-Id: I89754e9e4b42c45259729b0b0bb0a4bf464a981c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.db/commit/6cfdc11ea1ed7d8873f91683d6ea1be95fdff89e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.db/commit/6cfdc11ea1ed7d8873f91683d6ea1be95fdff89e"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
