)]}'
{"id":"openstack%2Foslo.db~652440","triplet_id":"openstack%2Foslo.db~master~Iba00d9fadaf3a682f3e4bb2dea335dfc74cf030d","project":"openstack/oslo.db","branch":"master","topic":"dropping-py35-testing","hashtags":[],"change_id":"Iba00d9fadaf3a682f3e4bb2dea335dfc74cf030d","subject":"Dropping the py35 testing","status":"MERGED","created":"2019-04-15 01:02:39.000000000","updated":"2019-05-30 20:36:19.000000000","submitted":"2019-05-30 20:36:19.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"652440-1559248579778-efcb9f90","meta_rev_id":"15c1d1ae17c8c6ec187a05fb22a6c4531595f294","_number":652440,"virtual_id_number":652440,"owner":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":2,"date":"2019-05-30 20:36:19.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},"all":[{"value":2,"date":"2019-05-30 14:01:37.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},"all":[{"value":1,"date":"2019-05-30 18:40:27.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-05-21 08:33:29.000000000","updated_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"reviewer":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"state":"REVIEWER"},{"updated":"2019-05-30 18:40:27.000000000","updated_by":{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},"reviewer":{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},"state":"REVIEWER"},{"updated":"2019-05-30 20:36:19.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"01b445f19db3cf6f7b33fa5e00de7d2deaafb8c8","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2019-04-15 01:02:39.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"95c028af8d9cb23cbf7dcd4aa0f0776b1e1a5bbb","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-15 02:44:11.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- test-release-openstack http://logs.openstack.org/40/652440/1/check/test-release-openstack/0d04072/ : SUCCESS in 3m 20s\n- tempest-full http://logs.openstack.org/40/652440/1/check/tempest-full/77ea5ed/ : SUCCESS in 1h 36m 13s\n- tempest-full-py3 http://logs.openstack.org/40/652440/1/check/tempest-full-py3/fd0a9c2/ : SUCCESS in 1h 36m 09s\n- openstack-tox-lower-constraints http://logs.openstack.org/40/652440/1/check/openstack-tox-lower-constraints/43e2866/ : SUCCESS in 4m 46s\n- openstack-tox-pep8 http://logs.openstack.org/40/652440/1/check/openstack-tox-pep8/bed2535/ : SUCCESS in 4m 39s\n- openstack-tox-py27 http://logs.openstack.org/40/652440/1/check/openstack-tox-py27/28ed0d4/ : SUCCESS in 5m 00s\n- openstack-tox-py36 http://logs.openstack.org/40/652440/1/check/openstack-tox-py36/fcb27fd/ : SUCCESS in 4m 49s\n- openstack-tox-docs http://logs.openstack.org/40/652440/1/check/openstack-tox-docs/a9c336b/html/ : SUCCESS in 4m 16s\n- build-openstack-releasenotes http://logs.openstack.org/40/652440/1/check/build-openstack-releasenotes/fe0ef55/html/ : SUCCESS in 4m 42s","accounts_in_message":[],"_revision_number":1},{"id":"1b92ef0b37baf67fc476b23fde26ba696d17c0dd","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2019-04-15 18:58:53.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"c95a40f138ec883157377664ac490837db1247b2","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-15 20:47:44.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- test-release-openstack http://logs.openstack.org/40/652440/2/check/test-release-openstack/ea38af7/ : SUCCESS in 4m 05s\n- tempest-full http://logs.openstack.org/40/652440/2/check/tempest-full/6c77d47/ : FAILURE in 1h 33m 41s\n- tempest-full-py3 http://logs.openstack.org/40/652440/2/check/tempest-full-py3/eab9cc3/ : SUCCESS in 1h 42m 08s\n- openstack-tox-lower-constraints http://logs.openstack.org/40/652440/2/check/openstack-tox-lower-constraints/6aa5e23/ : SUCCESS in 4m 58s\n- openstack-tox-pep8 http://logs.openstack.org/40/652440/2/check/openstack-tox-pep8/4e66448/ : SUCCESS in 5m 22s\n- openstack-tox-py27 http://logs.openstack.org/40/652440/2/check/openstack-tox-py27/0baf770/ : SUCCESS in 5m 38s\n- openstack-tox-py36 http://logs.openstack.org/40/652440/2/check/openstack-tox-py36/79c248c/ : SUCCESS in 5m 29s\n- openstack-tox-docs http://logs.openstack.org/40/652440/2/check/openstack-tox-docs/12d7ba4/html/ : SUCCESS in 4m 48s\n- build-openstack-releasenotes http://logs.openstack.org/40/652440/2/check/build-openstack-releasenotes/14318b7/html/ : SUCCESS in 4m 44s","accounts_in_message":[],"_revision_number":2},{"id":"b54c7e56187a343c45c6cd3b529bab5f73600b90","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2019-05-21 08:26:53.000000000","message":"Patch Set 3: Patch Set 2 was rebased","accounts_in_message":[],"_revision_number":3},{"id":"bed6baaa6e1f23bc3337611598a51f6183b620cb","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2019-05-21 08:33:29.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"2b9a77c0c0339c3ed83a249b09836b89919f1afd","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2019-05-21 08:34:04.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"ffcb4de94eb4685ba2fa0f966559c168f973f947","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-21 10:17:45.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- test-release-openstack http://logs.openstack.org/40/652440/3/check/test-release-openstack/d4002d2/ : SUCCESS in 3m 04s\n- tempest-full http://logs.openstack.org/40/652440/3/check/tempest-full/8c9b350/ : SUCCESS in 1h 43m 49s\n- tempest-full-py3 http://logs.openstack.org/40/652440/3/check/tempest-full-py3/59ae599/ : SUCCESS in 1h 40m 58s\n- openstack-tox-lower-constraints http://logs.openstack.org/40/652440/3/check/openstack-tox-lower-constraints/b11aa6a/ : SUCCESS in 4m 53s\n- openstack-tox-pep8 http://logs.openstack.org/40/652440/3/check/openstack-tox-pep8/48cca0d/ : SUCCESS in 4m 51s\n- openstack-tox-py27 http://logs.openstack.org/40/652440/3/check/openstack-tox-py27/de0587e/ : SUCCESS in 4m 53s\n- openstack-tox-py36 http://logs.openstack.org/40/652440/3/check/openstack-tox-py36/b218d0f/ : SUCCESS in 4m 54s\n- openstack-tox-docs http://logs.openstack.org/40/652440/3/check/openstack-tox-docs/13bf288/html/ : SUCCESS in 4m 14s\n- build-openstack-releasenotes http://logs.openstack.org/40/652440/3/check/build-openstack-releasenotes/a43d53d/html/ : SUCCESS in 4m 21s","accounts_in_message":[],"_revision_number":3},{"id":"af58c504b5f2e898505fc6d643b05935f397f5d3","author":{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},"date":"2019-05-30 13:12:38.000000000","message":"Patch Set 3: Workflow-1\n\nLooks like this has a merge conflict. Otherwise lgtm. py37 job is added in the followup.","accounts_in_message":[],"_revision_number":3},{"id":"cc22fe43ef2408fc4a55e94c39eaa72e391bb9bb","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2019-05-30 13:37:17.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"7edad8121ccc9ee94830cf05b925878c99328a2d","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2019-05-30 13:39:26.000000000","message":"Patch Set 5: Published edit on patch set 4.","accounts_in_message":[],"_revision_number":5},{"id":"08381e6a8f1a7831ca1f3390f0bd34e848c658d8","author":{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},"date":"2019-05-30 14:01:37.000000000","message":"Patch Set 5: Code-Review+2","accounts_in_message":[],"_revision_number":5},{"id":"e5c14b2fb662b5382df7d99b1500855f50ee154c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-30 15:45:22.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- test-release-openstack http://logs.openstack.org/40/652440/5/check/test-release-openstack/b4c9d56/ : SUCCESS in 3m 17s\n- tempest-full http://logs.openstack.org/40/652440/5/check/tempest-full/2918161/ : TIMED_OUT in 2h 04m 43s\n- tempest-full-py3 http://logs.openstack.org/40/652440/5/check/tempest-full-py3/e4adbca/ : TIMED_OUT in 2h 04m 19s\n- openstack-tox-lower-constraints http://logs.openstack.org/40/652440/5/check/openstack-tox-lower-constraints/4226441/ : SUCCESS in 5m 00s\n- openstack-tox-pep8 http://logs.openstack.org/40/652440/5/check/openstack-tox-pep8/878e122/ : SUCCESS in 4m 26s\n- openstack-tox-py27 http://logs.openstack.org/40/652440/5/check/openstack-tox-py27/334f421/ : SUCCESS in 5m 19s\n- openstack-tox-py36 http://logs.openstack.org/40/652440/5/check/openstack-tox-py36/46c9e41/ : SUCCESS in 5m 26s\n- openstack-tox-docs http://logs.openstack.org/40/652440/5/check/openstack-tox-docs/4dd3da1/html/ : SUCCESS in 5m 04s\n- build-openstack-releasenotes http://logs.openstack.org/40/652440/5/check/build-openstack-releasenotes/525f465/html/ : SUCCESS in 4m 37s","accounts_in_message":[],"_revision_number":5},{"id":"7e0a6c61ecfe77936ac381ddf579e8f8d5cb23f3","author":{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},"date":"2019-05-30 16:04:07.000000000","message":"Patch Set 5:\n\nrecheck","accounts_in_message":[],"_revision_number":5},{"id":"ef39f7b3ce1ddaea5540e2e2f5b693f18da902e2","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-30 17:58:38.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- test-release-openstack http://logs.openstack.org/40/652440/5/check/test-release-openstack/0f70e95/ : SUCCESS in 3m 08s\n- tempest-full http://logs.openstack.org/40/652440/5/check/tempest-full/4fd8225/ : SUCCESS in 1h 46m 46s\n- tempest-full-py3 http://logs.openstack.org/40/652440/5/check/tempest-full-py3/e41bb90/ : SUCCESS in 1h 47m 18s\n- openstack-tox-lower-constraints http://logs.openstack.org/40/652440/5/check/openstack-tox-lower-constraints/88ecf0d/ : SUCCESS in 7m 20s\n- openstack-tox-pep8 http://logs.openstack.org/40/652440/5/check/openstack-tox-pep8/1ad08a1/ : SUCCESS in 6m 14s\n- openstack-tox-py27 http://logs.openstack.org/40/652440/5/check/openstack-tox-py27/ae36824/ : SUCCESS in 7m 26s\n- openstack-tox-py36 http://logs.openstack.org/40/652440/5/check/openstack-tox-py36/a500b7d/ : SUCCESS in 7m 14s\n- openstack-tox-docs http://logs.openstack.org/40/652440/5/check/openstack-tox-docs/55a9e17/html/ : SUCCESS in 3m 52s\n- build-openstack-releasenotes http://logs.openstack.org/40/652440/5/check/build-openstack-releasenotes/953bd35/html/ : SUCCESS in 4m 37s","accounts_in_message":[],"_revision_number":5},{"id":"48cdc19833ece737b8093bd38274bc2674dd4965","author":{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},"date":"2019-05-30 18:40:27.000000000","message":"Patch Set 5: Workflow+1\n\nAlready had a +2 from Stephen pre-rebase.","accounts_in_message":[],"_revision_number":5},{"id":"363e008747f15fe880c04a504f98c43e48bd41dc","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-30 18:40:42.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"29ea05719a65fd70f88aeafeab7be4ec347827b0","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-30 20:36:19.000000000","message":"Patch Set 5: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- test-release-openstack http://logs.openstack.org/40/652440/5/gate/test-release-openstack/33fd2ca/ : SUCCESS in 4m 46s\n- tempest-full http://logs.openstack.org/40/652440/5/gate/tempest-full/4686baf/ : SUCCESS in 1h 49m 35s\n- tempest-full-py3 http://logs.openstack.org/40/652440/5/gate/tempest-full-py3/7c98f52/ : SUCCESS in 1h 40m 14s\n- openstack-tox-lower-constraints http://logs.openstack.org/40/652440/5/gate/openstack-tox-lower-constraints/0cb8842/ : SUCCESS in 5m 19s\n- openstack-tox-pep8 http://logs.openstack.org/40/652440/5/gate/openstack-tox-pep8/e294a0e/ : SUCCESS in 4m 25s\n- openstack-tox-py27 http://logs.openstack.org/40/652440/5/gate/openstack-tox-py27/5633329/ : SUCCESS in 5m 14s\n- openstack-tox-py36 http://logs.openstack.org/40/652440/5/gate/openstack-tox-py36/753f240/ : SUCCESS in 4m 23s\n- openstack-tox-docs http://logs.openstack.org/40/652440/5/gate/openstack-tox-docs/5c7b562/html/ : SUCCESS in 4m 08s\n- build-openstack-releasenotes http://logs.openstack.org/40/652440/5/gate/build-openstack-releasenotes/b371406/html/ : SUCCESS in 6m 14s","accounts_in_message":[],"_revision_number":5},{"id":"b2754bf01fcaa59ae08012f26998e6dccc8f3406","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-30 20:36:19.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"2f8250f3f747473c8f4258b148f2f7ded0a95084","revisions":{"50f4d47ac4831b03fc81d226f07bbd650cbba8d2":{"kind":"REWORK","_number":1,"created":"2019-04-15 01:02:39.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/40/652440/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.db","ref":"refs/changes/40/652440/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.db refs/changes/40/652440/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.db refs/changes/40/652440/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.db refs/changes/40/652440/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.db refs/changes/40/652440/1"}}},"commit":{"parents":[{"commit":"3c50cb5cce82101a7be491ea24a713d70ca064e3","subject":"Fix deprecation warnings under py36","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.db/commit/3c50cb5cce82101a7be491ea24a713d70ca064e3"}]}],"author":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2019-04-15 01:02:17.000000000","tz":0},"committer":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2019-04-15 01:02:37.000000000","tz":0},"subject":"Dropping the py35 testing","message":"Dropping the py35 testing\n\nAll the integration testing has been moved to\nBionic now[1] and py3.5 is not tested runtime for\nTrain or stable/stein[2].\n\nAs per below ML thread, we are good to drop the py35\ntesting now:\nhttp://lists.openstack.org/pipermail/openstack-discuss/2019-April/005097.html\n\n[1] http://lists.openstack.org/pipermail/openstack-discuss/2019-April/004647.html\n[2]\nhttps://governance.openstack.org/tc/reference/runtimes/stein.html\nhttps://governance.openstack.org/tc/reference/runtimes/train.html\n\nChange-Id: Iba00d9fadaf3a682f3e4bb2dea335dfc74cf030d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.db/commit/50f4d47ac4831b03fc81d226f07bbd650cbba8d2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.db/commit/50f4d47ac4831b03fc81d226f07bbd650cbba8d2"}]},"branch":"refs/heads/master"},"991c4dbcb6b71db5b4cd5de78a0bb4b7af120815":{"kind":"REWORK","_number":2,"created":"2019-04-15 18:58:53.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/40/652440/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.db","ref":"refs/changes/40/652440/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.db refs/changes/40/652440/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.db refs/changes/40/652440/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.db refs/changes/40/652440/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.db refs/changes/40/652440/2"}}},"commit":{"parents":[{"commit":"3c50cb5cce82101a7be491ea24a713d70ca064e3","subject":"Fix deprecation warnings under py36","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.db/commit/3c50cb5cce82101a7be491ea24a713d70ca064e3"}]}],"author":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2019-04-15 01:02:17.000000000","tz":0},"committer":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2019-04-15 18:58:50.000000000","tz":0},"subject":"Dropping the py35 testing","message":"Dropping the py35 testing\n\nAll the integration testing has been moved to\nBionic now[1] and py3.5 is not tested runtime for\nTrain or stable/stein[2].\n\nAs per below ML thread, we are good to drop the py35\ntesting now:\nhttp://lists.openstack.org/pipermail/openstack-discuss/2019-April/005097.html\n\n[1] http://lists.openstack.org/pipermail/openstack-discuss/2019-April/004647.html\n[2]\nhttps://governance.openstack.org/tc/reference/runtimes/stein.html\nhttps://governance.openstack.org/tc/reference/runtimes/train.html\n\nChange-Id: Iba00d9fadaf3a682f3e4bb2dea335dfc74cf030d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.db/commit/991c4dbcb6b71db5b4cd5de78a0bb4b7af120815"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.db/commit/991c4dbcb6b71db5b4cd5de78a0bb4b7af120815"}]},"branch":"refs/heads/master"},"4316ea1761d93765e1f0ce6e5153b3d2b9e6e82d":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2019-05-21 08:26:53.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/40/652440/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.db","ref":"refs/changes/40/652440/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.db refs/changes/40/652440/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.db refs/changes/40/652440/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.db refs/changes/40/652440/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.db refs/changes/40/652440/3"}}},"commit":{"parents":[{"commit":"b9c51e295b81cedbc747e662628bc69d1f60e6db","subject":"Merge \"Support context function argument as keyword\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.db/commit/b9c51e295b81cedbc747e662628bc69d1f60e6db"}]}],"author":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2019-04-15 01:02:17.000000000","tz":0},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2019-05-21 08:26:53.000000000","tz":0},"subject":"Dropping the py35 testing","message":"Dropping the py35 testing\n\nAll the integration testing has been moved to\nBionic now[1] and py3.5 is not tested runtime for\nTrain or stable/stein[2].\n\nAs per below ML thread, we are good to drop the py35\ntesting now:\nhttp://lists.openstack.org/pipermail/openstack-discuss/2019-April/005097.html\n\n[1] http://lists.openstack.org/pipermail/openstack-discuss/2019-April/004647.html\n[2]\nhttps://governance.openstack.org/tc/reference/runtimes/stein.html\nhttps://governance.openstack.org/tc/reference/runtimes/train.html\n\nChange-Id: Iba00d9fadaf3a682f3e4bb2dea335dfc74cf030d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.db/commit/4316ea1761d93765e1f0ce6e5153b3d2b9e6e82d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.db/commit/4316ea1761d93765e1f0ce6e5153b3d2b9e6e82d"}]},"branch":"refs/heads/master"},"a00e9bac52bb1c9bf865cdbaccb8e70fe602df8e":{"kind":"REWORK","_number":4,"created":"2019-05-30 13:37:17.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/40/652440/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.db","ref":"refs/changes/40/652440/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.db refs/changes/40/652440/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.db refs/changes/40/652440/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.db refs/changes/40/652440/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.db refs/changes/40/652440/4"}}},"commit":{"parents":[{"commit":"05ddd9fc9e814b8a949ee65b512b35f2432ccec6","subject":"docs: Use sphinxcontrib.apidoc for building API docs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.db/commit/05ddd9fc9e814b8a949ee65b512b35f2432ccec6"}]}],"author":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2019-04-15 01:02:17.000000000","tz":0},"committer":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2019-05-30 13:37:01.000000000","tz":0},"subject":"Dropping the py35 testing","message":"Dropping the py35 testing\n\nAll the integration testing has been moved to\nBionic now[1] and py3.5 is not tested runtime for\nTrain or stable/stein[2].\n\nAs per below ML thread, we are good to drop the py35\ntesting now:\nhttp://lists.openstack.org/pipermail/openstack-discuss/2019-April/005097.html\n\n[1] http://lists.openstack.org/pipermail/openstack-discuss/2019-April/004647.html\n[2]\nhttps://governance.openstack.org/tc/reference/runtimes/stein.html\nhttps://governance.openstack.org/tc/reference/runtimes/train.html\n\nChange-Id: Iba00d9fadaf3a682f3e4bb2dea335dfc74cf030d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.db/commit/a00e9bac52bb1c9bf865cdbaccb8e70fe602df8e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.db/commit/a00e9bac52bb1c9bf865cdbaccb8e70fe602df8e"}]},"branch":"refs/heads/master"},"2f8250f3f747473c8f4258b148f2f7ded0a95084":{"kind":"REWORK","_number":5,"created":"2019-05-30 13:39:26.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/40/652440/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.db","ref":"refs/changes/40/652440/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.db refs/changes/40/652440/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.db refs/changes/40/652440/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.db refs/changes/40/652440/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.db refs/changes/40/652440/5"}}},"commit":{"parents":[{"commit":"05ddd9fc9e814b8a949ee65b512b35f2432ccec6","subject":"docs: Use sphinxcontrib.apidoc for building API docs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.db/commit/05ddd9fc9e814b8a949ee65b512b35f2432ccec6"}]}],"author":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2019-04-15 01:02:17.000000000","tz":0},"committer":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2019-05-30 13:39:24.000000000","tz":0},"subject":"Dropping the py35 testing","message":"Dropping the py35 testing\n\nAll the integration testing has been moved to\nBionic now[1] and py3.5 is not tested runtime for\nTrain or stable/stein[2].\n\nAs per below ML thread, we are good to drop the py35\ntesting now:\nhttp://lists.openstack.org/pipermail/openstack-discuss/2019-April/005097.html\n\n[1] http://lists.openstack.org/pipermail/openstack-discuss/2019-April/004647.html\n[2]\nhttps://governance.openstack.org/tc/reference/runtimes/stein.html\nhttps://governance.openstack.org/tc/reference/runtimes/train.html\n\nChange-Id: Iba00d9fadaf3a682f3e4bb2dea335dfc74cf030d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.db/commit/2f8250f3f747473c8f4258b148f2f7ded0a95084"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.db/commit/2f8250f3f747473c8f4258b148f2f7ded0a95084"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
