)]}'
{"id":"openstack%2Foslo.db~654725","triplet_id":"openstack%2Foslo.db~master~Iff5086fee8c94325aa16a3b5727dac66a64478ec","project":"openstack/oslo.db","branch":"master","hashtags":[],"change_id":"Iff5086fee8c94325aa16a3b5727dac66a64478ec","subject":"Replace git.openstack.org URLs with opendev.org URLs","status":"MERGED","created":"2019-04-23 05:25:32.000000000","updated":"2019-05-03 20:20:47.000000000","submitted":"2019-05-03 20:20:47.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"654725-1556914847496-eabd120a","meta_rev_id":"2bf647859930e8da1c49d2db7bb84ffd413e2f1b","_number":654725,"virtual_id_number":654725,"owner":{"_account_id":28614,"name":"jacky06","email":"zhang.min@99cloud.net","username":"jacky06"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},{"value":0,"_account_id":27781,"name":"zhulingjie","email":"easyzlj@gmail.com","username":"zhulingjie"},{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":2,"date":"2019-05-03 20:20:47.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},"all":[{"value":2,"date":"2019-05-03 18:30:47.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},{"value":1,"date":"2019-04-30 05:51:04.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":27781,"name":"zhulingjie","email":"easyzlj@gmail.com","username":"zhulingjie"},{"value":2,"date":"2019-05-03 16:40:28.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},"all":[{"value":1,"date":"2019-05-03 18:30:47.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},{"value":0,"_account_id":27781,"name":"zhulingjie","email":"easyzlj@gmail.com","username":"zhulingjie"},{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},{"value":0,"_account_id":27781,"name":"zhulingjie","email":"easyzlj@gmail.com","username":"zhulingjie"},{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":27781,"name":"zhulingjie","email":"easyzlj@gmail.com","username":"zhulingjie"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-04-30 05:51:04.000000000","updated_by":{"_account_id":27781,"name":"zhulingjie","email":"easyzlj@gmail.com","username":"zhulingjie"},"reviewer":{"_account_id":27781,"name":"zhulingjie","email":"easyzlj@gmail.com","username":"zhulingjie"},"state":"REVIEWER"},{"updated":"2019-05-03 16:40:28.000000000","updated_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"reviewer":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"state":"REVIEWER"},{"updated":"2019-05-03 18:30:47.000000000","updated_by":{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},"reviewer":{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},"state":"REVIEWER"},{"updated":"2019-05-03 20:20:47.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"4c6a4b60a38eecbce3ed9ebcfd96c4b57913c019","author":{"_account_id":28614,"name":"jacky06","email":"zhang.min@99cloud.net","username":"jacky06"},"date":"2019-04-23 05:25:32.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"297825fb477e94fd7d3bd141d59117c9f9b543b3","author":{"_account_id":28614,"name":"jacky06","email":"zhang.min@99cloud.net","username":"jacky06"},"date":"2019-04-23 05:29:27.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"97403086309a516516012eddc1a0801f83ac8434","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-23 08:55:01.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- test-release-openstack http://logs.openstack.org/25/654725/2/check/test-release-openstack/f27d0d8/ : SUCCESS in 5m 39s\n- tempest-full http://logs.openstack.org/25/654725/2/check/tempest-full/52dc1d3/ : SUCCESS in 1h 38m 54s\n- tempest-full-py3 http://logs.openstack.org/25/654725/2/check/tempest-full-py3/a68b68a/ : SUCCESS in 1h 27m 08s\n- openstack-tox-lower-constraints http://logs.openstack.org/25/654725/2/check/openstack-tox-lower-constraints/fafb521/ : SUCCESS in 6m 11s\n- openstack-tox-pep8 http://logs.openstack.org/25/654725/2/check/openstack-tox-pep8/6472105/ : SUCCESS in 4m 45s\n- openstack-tox-py27 http://logs.openstack.org/25/654725/2/check/openstack-tox-py27/065e885/ : SUCCESS in 5m 22s\n- openstack-tox-py35 http://logs.openstack.org/25/654725/2/check/openstack-tox-py35/9ff0f0c/ : SUCCESS in 5m 33s\n- openstack-tox-py36 http://logs.openstack.org/25/654725/2/check/openstack-tox-py36/ed1a3d9/ : SUCCESS in 5m 32s\n- openstack-tox-docs http://logs.openstack.org/25/654725/2/check/openstack-tox-docs/f680d4c/html/ : SUCCESS in 4m 04s\n- build-openstack-releasenotes http://logs.openstack.org/25/654725/2/check/build-openstack-releasenotes/9268f6a/html/ : SUCCESS in 4m 30s","accounts_in_message":[],"_revision_number":2},{"id":"8fe197655096fce66c2d28606dfc1159728986e0","author":{"_account_id":27781,"name":"zhulingjie","email":"easyzlj@gmail.com","username":"zhulingjie"},"date":"2019-04-30 05:51:04.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"a1eec5717d71d465dbd8516d4d8b3e85f68bd26a","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2019-05-03 16:40:28.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"fa789caf0bb7fa4416da64a13fde45a8d8512044","author":{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},"date":"2019-05-03 18:30:47.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"bca2643afe8f5001b2a091c06121838d14c65159","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-03 18:31:26.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"03901a439e1b4102dd39fbf9f631a0ca66825fde","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-03 20:20:47.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"2bf647859930e8da1c49d2db7bb84ffd413e2f1b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-03 20:20:47.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- test-release-openstack http://logs.openstack.org/25/654725/2/gate/test-release-openstack/f00a619/ : SUCCESS in 3m 15s\n- tempest-full http://logs.openstack.org/25/654725/2/gate/tempest-full/97aa0fc/ : SUCCESS in 1h 45m 27s\n- tempest-full-py3 http://logs.openstack.org/25/654725/2/gate/tempest-full-py3/b0e3b5e/ : SUCCESS in 1h 30m 50s\n- openstack-tox-lower-constraints http://logs.openstack.org/25/654725/2/gate/openstack-tox-lower-constraints/a90a446/ : SUCCESS in 5m 08s\n- openstack-tox-pep8 http://logs.openstack.org/25/654725/2/gate/openstack-tox-pep8/c676007/ : SUCCESS in 4m 14s\n- openstack-tox-py27 http://logs.openstack.org/25/654725/2/gate/openstack-tox-py27/983107a/ : SUCCESS in 5m 40s\n- openstack-tox-py35 http://logs.openstack.org/25/654725/2/gate/openstack-tox-py35/4d7c8dd/ : SUCCESS in 5m 28s\n- openstack-tox-py36 http://logs.openstack.org/25/654725/2/gate/openstack-tox-py36/76715f4/ : SUCCESS in 4m 49s\n- openstack-tox-docs http://logs.openstack.org/25/654725/2/gate/openstack-tox-docs/e8d945c/html/ : SUCCESS in 4m 42s\n- build-openstack-releasenotes http://logs.openstack.org/25/654725/2/gate/build-openstack-releasenotes/9ed8cbe/html/ : SUCCESS in 4m 27s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"cc5c0e36cd7cf43bac4468bf4dadddb479fabd6a","revisions":{"f06280b0c2756dcfd084901fe4f1f173054dfd39":{"kind":"REWORK","_number":1,"created":"2019-04-23 05:25:32.000000000","uploader":{"_account_id":28614,"name":"jacky06","email":"zhang.min@99cloud.net","username":"jacky06"},"ref":"refs/changes/25/654725/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.db","ref":"refs/changes/25/654725/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.db refs/changes/25/654725/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.db refs/changes/25/654725/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.db refs/changes/25/654725/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.db refs/changes/25/654725/1"}}},"commit":{"parents":[{"commit":"0de3a705e845cc3d16dc89b3b86b072b3785ea5d","subject":"OpenDev Migration Patch","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.db/commit/0de3a705e845cc3d16dc89b3b86b072b3785ea5d"}]}],"author":{"name":"jacky06","email":"zhang.min@99cloud.net","date":"2019-04-23 05:25:13.000000000","tz":480},"committer":{"name":"jacky06","email":"zhang.min@99cloud.net","date":"2019-04-23 05:25:28.000000000","tz":480},"subject":"Replace git.openstack.org URLs with opendev.org URLs","message":"Replace git.openstack.org URLs with opendev.org URLs\n\nChange-Id: Iff5086fee8c94325aa16a3b5727dac66a64478ec\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.db/commit/f06280b0c2756dcfd084901fe4f1f173054dfd39"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.db/commit/f06280b0c2756dcfd084901fe4f1f173054dfd39"}]},"branch":"refs/heads/master"},"cc5c0e36cd7cf43bac4468bf4dadddb479fabd6a":{"kind":"REWORK","_number":2,"created":"2019-04-23 05:29:27.000000000","uploader":{"_account_id":28614,"name":"jacky06","email":"zhang.min@99cloud.net","username":"jacky06"},"ref":"refs/changes/25/654725/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.db","ref":"refs/changes/25/654725/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.db refs/changes/25/654725/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.db refs/changes/25/654725/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.db refs/changes/25/654725/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.db refs/changes/25/654725/2"}}},"commit":{"parents":[{"commit":"0de3a705e845cc3d16dc89b3b86b072b3785ea5d","subject":"OpenDev Migration Patch","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.db/commit/0de3a705e845cc3d16dc89b3b86b072b3785ea5d"}]}],"author":{"name":"jacky06","email":"zhang.min@99cloud.net","date":"2019-04-23 05:25:13.000000000","tz":480},"committer":{"name":"jacky06","email":"zhang.min@99cloud.net","date":"2019-04-23 05:29:23.000000000","tz":0},"subject":"Replace git.openstack.org URLs with opendev.org URLs","message":"Replace git.openstack.org URLs with opendev.org URLs\n\nChange-Id: Iff5086fee8c94325aa16a3b5727dac66a64478ec\nCloses-Bug: #1825933\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.db/commit/cc5c0e36cd7cf43bac4468bf4dadddb479fabd6a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.db/commit/cc5c0e36cd7cf43bac4468bf4dadddb479fabd6a"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
