)]}'
{"id":"openstack%2Foslo.db~658829","triplet_id":"openstack%2Foslo.db~master~I1543836873fe065d58c85217b627f2d2c569a219","project":"openstack/oslo.db","branch":"master","hashtags":[],"change_id":"I1543836873fe065d58c85217b627f2d2c569a219","subject":"Blacklist bandit 1.6.0 in test-requirements and update sphinx versions","status":"ABANDONED","created":"2019-05-13 14:22:23.000000000","updated":"2019-05-15 16:06:42.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"a01a80e89c2b8497dedf5bde42be5a6da199f040","_number":658829,"virtual_id_number":658829,"owner":{"_account_id":28714,"name":"Adrian Chiris","email":"adrianc@nvidia.com","username":"adrianc"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":11816,"name":"mike bayer","display_name":"Mike Bayer","email":"mike_mp@zzzcomputing.com","username":"zzzeek","status":"Red Hat"},{"value":1,"date":"2019-05-14 14:59:27.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28543,"name":"shupeng","email":"15050873171@163.com","username":"Vieri"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":11816,"name":"mike bayer","display_name":"Mike Bayer","email":"mike_mp@zzzcomputing.com","username":"zzzeek","status":"Red Hat"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":28543,"name":"shupeng","email":"15050873171@163.com","username":"Vieri"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"date":"2019-05-14 18:51:16.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":11816,"name":"mike bayer","display_name":"Mike Bayer","email":"mike_mp@zzzcomputing.com","username":"zzzeek","status":"Red Hat"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28543,"name":"shupeng","email":"15050873171@163.com","username":"Vieri"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":11816,"name":"mike bayer","display_name":"Mike Bayer","email":"mike_mp@zzzcomputing.com","username":"zzzeek","status":"Red Hat"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":28543,"name":"shupeng","email":"15050873171@163.com","username":"Vieri"}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},{"_account_id":11816,"name":"mike bayer","display_name":"Mike Bayer","email":"mike_mp@zzzcomputing.com","username":"zzzeek","status":"Red Hat"},{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28543,"name":"shupeng","email":"15050873171@163.com","username":"Vieri"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-05-13 14:47:01.000000000","updated_by":{"_account_id":11816,"name":"mike bayer","display_name":"Mike Bayer","email":"mike_mp@zzzcomputing.com","username":"zzzeek","status":"Red Hat"},"reviewer":{"_account_id":11816,"name":"mike bayer","display_name":"Mike Bayer","email":"mike_mp@zzzcomputing.com","username":"zzzeek","status":"Red Hat"},"state":"REVIEWER"},{"updated":"2019-05-13 15:44:39.000000000","updated_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"reviewer":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"state":"REVIEWER"},{"updated":"2019-05-14 09:44:14.000000000","updated_by":{"_account_id":28543,"name":"shupeng","email":"15050873171@163.com","username":"Vieri"},"reviewer":{"_account_id":28543,"name":"shupeng","email":"15050873171@163.com","username":"Vieri"},"state":"REVIEWER"},{"updated":"2019-05-14 14:59:27.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-05-14 18:51:16.000000000","updated_by":{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},"reviewer":{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},"state":"REVIEWER"}],"messages":[{"id":"6606579f300f165a95e8d29f0f65e9b115c22859","author":{"_account_id":28714,"name":"Adrian Chiris","email":"adrianc@nvidia.com","username":"adrianc"},"date":"2019-05-13 14:22:23.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"8ec93c3d5ad5c28e922d8d542fe79476d7a9c491","author":{"_account_id":11816,"name":"mike bayer","display_name":"Mike Bayer","email":"mike_mp@zzzcomputing.com","username":"zzzeek","status":"Red Hat"},"date":"2019-05-13 14:47:01.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"06aeb83c62875604f45f075742c5948c865aab90","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-13 16:18:30.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- requirements-check http://logs.openstack.org/29/658829/1/check/requirements-check/9f42ba3/ : FAILURE in 3m 31s\n- tempest-full http://logs.openstack.org/29/658829/1/check/tempest-full/d2c0ddc/ : SUCCESS in 1h 50m 59s\n- tempest-full-py3 http://logs.openstack.org/29/658829/1/check/tempest-full-py3/d99dff6/ : SUCCESS in 1h 43m 26s\n- openstack-tox-lower-constraints http://logs.openstack.org/29/658829/1/check/openstack-tox-lower-constraints/ead99f6/ : SUCCESS in 5m 22s\n- openstack-tox-pep8 http://logs.openstack.org/29/658829/1/check/openstack-tox-pep8/5f61271/ : SUCCESS in 4m 20s\n- openstack-tox-py27 http://logs.openstack.org/29/658829/1/check/openstack-tox-py27/8ed9b45/ : SUCCESS in 4m 50s\n- openstack-tox-py35 http://logs.openstack.org/29/658829/1/check/openstack-tox-py35/1be1930/ : SUCCESS in 4m 35s\n- openstack-tox-py36 http://logs.openstack.org/29/658829/1/check/openstack-tox-py36/55db6a0/ : SUCCESS in 4m 47s\n- openstack-tox-docs http://logs.openstack.org/29/658829/1/check/openstack-tox-docs/7bd7cda/html/ : SUCCESS in 4m 30s\n- build-openstack-releasenotes http://logs.openstack.org/29/658829/1/check/build-openstack-releasenotes/a6bc8ba/html/ : SUCCESS in 4m 13s","accounts_in_message":[],"_revision_number":1},{"id":"97a7e975a0500e3bef5ec0542b9fdb2f1de77c85","author":{"_account_id":28543,"name":"shupeng","email":"15050873171@163.com","username":"Vieri"},"date":"2019-05-14 09:44:14.000000000","message":"Patch Set 1: Code-Review+1\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"41550aab449e5314998235aa374f0b1a4b040904","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-14 11:25:15.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- requirements-check http://logs.openstack.org/29/658829/1/check/requirements-check/f31fb47/ : FAILURE in 3m 23s\n- tempest-full http://logs.openstack.org/29/658829/1/check/tempest-full/68cd6e5/ : SUCCESS in 1h 35m 36s\n- tempest-full-py3 http://logs.openstack.org/29/658829/1/check/tempest-full-py3/0fb6f59/ : SUCCESS in 1h 30m 25s\n- openstack-tox-lower-constraints http://logs.openstack.org/29/658829/1/check/openstack-tox-lower-constraints/208bbd5/ : SUCCESS in 4m 42s\n- openstack-tox-pep8 http://logs.openstack.org/29/658829/1/check/openstack-tox-pep8/dbfc65b/ : SUCCESS in 3m 52s\n- openstack-tox-py27 http://logs.openstack.org/29/658829/1/check/openstack-tox-py27/4ddbee3/ : SUCCESS in 5m 04s\n- openstack-tox-py35 http://logs.openstack.org/29/658829/1/check/openstack-tox-py35/e1555ba/ : SUCCESS in 4m 41s\n- openstack-tox-py36 http://logs.openstack.org/29/658829/1/check/openstack-tox-py36/56ba8d9/ : SUCCESS in 4m 01s\n- openstack-tox-docs http://logs.openstack.org/29/658829/1/check/openstack-tox-docs/879920c/html/ : SUCCESS in 4m 23s\n- build-openstack-releasenotes http://logs.openstack.org/29/658829/1/check/build-openstack-releasenotes/6595a1f/html/ : SUCCESS in 4m 04s","accounts_in_message":[],"_revision_number":1},{"id":"04a377cff20deb5fbb99bdf0a1da8036b3df0e82","author":{"_account_id":28714,"name":"Adrian Chiris","email":"adrianc@nvidia.com","username":"adrianc"},"date":"2019-05-14 13:10:51.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"c30bea6973502139ffcc3820dac1e3583ab0955c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-14 14:59:27.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- requirements-check http://logs.openstack.org/29/658829/2/check/requirements-check/4979d5d/ : SUCCESS in 3m 10s\n- tempest-full http://logs.openstack.org/29/658829/2/check/tempest-full/10eb8b3/ : SUCCESS in 1h 48m 05s\n- tempest-full-py3 http://logs.openstack.org/29/658829/2/check/tempest-full-py3/0d65e46/ : SUCCESS in 1h 40m 29s\n- openstack-tox-lower-constraints http://logs.openstack.org/29/658829/2/check/openstack-tox-lower-constraints/d2651da/ : SUCCESS in 7m 21s\n- openstack-tox-pep8 http://logs.openstack.org/29/658829/2/check/openstack-tox-pep8/bb916e3/ : SUCCESS in 5m 43s\n- openstack-tox-py27 http://logs.openstack.org/29/658829/2/check/openstack-tox-py27/4a8808a/ : SUCCESS in 5m 16s\n- openstack-tox-py35 http://logs.openstack.org/29/658829/2/check/openstack-tox-py35/2c27c2e/ : SUCCESS in 4m 59s\n- openstack-tox-py36 http://logs.openstack.org/29/658829/2/check/openstack-tox-py36/ba8cbaf/ : SUCCESS in 6m 52s\n- openstack-tox-docs http://logs.openstack.org/29/658829/2/check/openstack-tox-docs/772180e/html/ : SUCCESS in 4m 24s\n- build-openstack-releasenotes http://logs.openstack.org/29/658829/2/check/build-openstack-releasenotes/e449a3d/html/ : SUCCESS in 6m 23s","accounts_in_message":[],"_revision_number":2},{"id":"1813231c77eb1850a388e28224a5174f67f4b8d3","author":{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},"date":"2019-05-14 18:51:16.000000000","message":"Patch Set 2:\n\nThanks for this. I\u0027m afraid I didn\u0027t see it in time and submitted https://review.opendev.org/#/c/658910/ to do essentially the same thing. That should unblock this repo.","accounts_in_message":[],"_revision_number":2},{"id":"246f91b7253cf0fdeeb1a34eac0260740e6cb646","author":{"_account_id":28714,"name":"Adrian Chiris","email":"adrianc@nvidia.com","username":"adrianc"},"date":"2019-05-15 16:06:15.000000000","message":"Abandoned\n\na similar patch already pushed:\nhttps://review.opendev.org/#/c/658910/","accounts_in_message":[],"_revision_number":2},{"id":"a01a80e89c2b8497dedf5bde42be5a6da199f040","author":{"_account_id":28714,"name":"Adrian Chiris","email":"adrianc@nvidia.com","username":"adrianc"},"date":"2019-05-15 16:06:42.000000000","message":"Patch Set 2:\n\nno worries, glad the issue was sorted.","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"98d11fc8a2afcb8a1d05091b03e3c9f1b08d264f","revisions":{"de6f268d01d34c961b0348183ce6c0b307f97302":{"kind":"REWORK","_number":1,"created":"2019-05-13 14:22:23.000000000","uploader":{"_account_id":28714,"name":"Adrian Chiris","email":"adrianc@nvidia.com","username":"adrianc"},"ref":"refs/changes/29/658829/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.db","ref":"refs/changes/29/658829/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.db refs/changes/29/658829/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.db refs/changes/29/658829/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.db refs/changes/29/658829/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.db refs/changes/29/658829/1"}}},"commit":{"parents":[{"commit":"aef64ea8aea03caa68e9006570f0e8943553bcfc","subject":"Merge \"Replace git.openstack.org URLs with opendev.org URLs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.db/commit/aef64ea8aea03caa68e9006570f0e8943553bcfc"}]}],"author":{"name":"Adrian Chiris","email":"adrianc@mellanox.com","date":"2019-05-13 14:15:06.000000000","tz":180},"committer":{"name":"Adrian Chiris","email":"adrianc@mellanox.com","date":"2019-05-13 14:20:24.000000000","tz":180},"subject":"Blacklist bandit 1.6.0 in test-requirements","message":"Blacklist bandit 1.6.0 in test-requirements\n\nThere is a bug in bandit[1] that causes tox pep8 test to fail\nas it does not respect the excluded directories.\n\nThis change excludes this version from being used during tests\nas oslo.db tests cause several violations.\n\n[1] https://github.com/PyCQA/bandit/issues/488\n\nChange-Id: I1543836873fe065d58c85217b627f2d2c569a219\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.db/commit/de6f268d01d34c961b0348183ce6c0b307f97302"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.db/commit/de6f268d01d34c961b0348183ce6c0b307f97302"}]},"branch":"refs/heads/master"},"98d11fc8a2afcb8a1d05091b03e3c9f1b08d264f":{"kind":"REWORK","_number":2,"created":"2019-05-14 13:10:51.000000000","uploader":{"_account_id":28714,"name":"Adrian Chiris","email":"adrianc@nvidia.com","username":"adrianc"},"ref":"refs/changes/29/658829/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.db","ref":"refs/changes/29/658829/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.db refs/changes/29/658829/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.db refs/changes/29/658829/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.db refs/changes/29/658829/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.db refs/changes/29/658829/2"}}},"commit":{"parents":[{"commit":"aef64ea8aea03caa68e9006570f0e8943553bcfc","subject":"Merge \"Replace git.openstack.org URLs with opendev.org URLs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.db/commit/aef64ea8aea03caa68e9006570f0e8943553bcfc"}]}],"author":{"name":"Adrian Chiris","email":"adrianc@mellanox.com","date":"2019-05-13 14:15:06.000000000","tz":180},"committer":{"name":"Adrian Chiris","email":"adrianc@mellanox.com","date":"2019-05-14 12:17:01.000000000","tz":180},"subject":"Blacklist bandit 1.6.0 in test-requirements and update sphinx versions","message":"Blacklist bandit 1.6.0 in test-requirements and update sphinx versions\n\nThere is a bug in bandit[1] that causes tox pep8 test to fail\nas it does not respect the excluded directories.\n\nThis change excludes this version from being used during tests\nas oslo.db tests cause several violations.\n\nThe second change aligns sphinx requirements to OpenStack global\nrequirements as sphinx 2.0.0 onwards does not support python 2.7.\nThis change is required to be submitted in the same patch as unit\ntests fails.\n\n[1] https://github.com/PyCQA/bandit/issues/488\n\nChange-Id: I1543836873fe065d58c85217b627f2d2c569a219\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.db/commit/98d11fc8a2afcb8a1d05091b03e3c9f1b08d264f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.db/commit/98d11fc8a2afcb8a1d05091b03e3c9f1b08d264f"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
