)]}'
{"id":"openstack%2Foslo.db~658910","triplet_id":"openstack%2Foslo.db~master~If86c82e0f4a519baca664af79352846c4af9a01c","project":"openstack/oslo.db","branch":"master","topic":"cap-bandit","hashtags":[],"change_id":"If86c82e0f4a519baca664af79352846c4af9a01c","subject":"Cap Bandit below 1.6.0 and update Sphinx requirement","status":"MERGED","created":"2019-05-13 20:10:13.000000000","updated":"2019-05-14 18:30:36.000000000","submitted":"2019-05-14 18:30:36.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"658910-1557858636174-0de1d12d","meta_rev_id":"19d98c5941731efe15d853d8e50ab6e1dc99bb5a","_number":658910,"virtual_id_number":658910,"owner":{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"date":"2019-05-14 14:33:17.000000000","_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},{"value":0,"_account_id":28935,"name":"98k","email":"18552437190@163.com","username":"98k"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":2,"date":"2019-05-14 18:30:35.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"recommended":{"_account_id":28935,"name":"98k","email":"18552437190@163.com","username":"98k"},"all":[{"value":0,"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},{"value":1,"date":"2019-05-14 14:37:39.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28935,"name":"98k","email":"18552437190@163.com","username":"98k"},{"value":2,"date":"2019-05-14 12:46:18.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"all":[{"value":0,"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},{"value":0,"_account_id":28935,"name":"98k","email":"18552437190@163.com","username":"98k"},{"value":1,"date":"2019-05-14 12:46:18.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},{"value":0,"_account_id":28935,"name":"98k","email":"18552437190@163.com","username":"98k"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"_account_id":28935,"name":"98k","email":"18552437190@163.com","username":"98k"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-05-14 12:46:18.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2019-05-14 14:37:39.000000000","updated_by":{"_account_id":28935,"name":"98k","email":"18552437190@163.com","username":"98k"},"reviewer":{"_account_id":28935,"name":"98k","email":"18552437190@163.com","username":"98k"},"state":"REVIEWER"},{"updated":"2019-05-14 18:30:35.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"c132d57ed4a49e88adeb274e2cdbcb2cbac77ec4","author":{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},"date":"2019-05-13 20:10:13.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"4db0b5bbbca6888c8906848e439fcc9639212db9","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-13 22:22:56.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- requirements-check http://logs.openstack.org/10/658910/1/check/requirements-check/f95d9d7/ : SUCCESS in 3m 26s\n- tempest-full http://logs.openstack.org/10/658910/1/check/tempest-full/174b413/ : FAILURE in 1h 59m 09s\n- tempest-full-py3 http://logs.openstack.org/10/658910/1/check/tempest-full-py3/16b4472/ : SUCCESS in 1h 46m 09s\n- openstack-tox-lower-constraints http://logs.openstack.org/10/658910/1/check/openstack-tox-lower-constraints/e4fb478/ : SUCCESS in 7m 35s\n- openstack-tox-pep8 http://logs.openstack.org/10/658910/1/check/openstack-tox-pep8/ccb302b/ : SUCCESS in 4m 41s\n- openstack-tox-py27 http://logs.openstack.org/10/658910/1/check/openstack-tox-py27/5ea3b16/ : SUCCESS in 4m 54s\n- openstack-tox-py35 http://logs.openstack.org/10/658910/1/check/openstack-tox-py35/b2604a0/ : SUCCESS in 4m 39s\n- openstack-tox-py36 http://logs.openstack.org/10/658910/1/check/openstack-tox-py36/6f2d80c/ : SUCCESS in 4m 44s\n- openstack-tox-docs http://logs.openstack.org/10/658910/1/check/openstack-tox-docs/efa4693/html/ : SUCCESS in 4m 10s\n- build-openstack-releasenotes http://logs.openstack.org/10/658910/1/check/build-openstack-releasenotes/48dd21f/html/ : SUCCESS in 7m 07s","accounts_in_message":[],"_revision_number":1},{"id":"c9822b8f8822be45787b23a9dcd5fcfa7510b8f7","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2019-05-14 07:53:35.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"ccdc4a1501bd42ba70c8932442888e6aa4c1dd4c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-14 09:49:31.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- requirements-check http://logs.openstack.org/10/658910/1/check/requirements-check/28202e1/ : SUCCESS in 3m 21s\n- tempest-full http://logs.openstack.org/10/658910/1/check/tempest-full/73d903e/ : SUCCESS in 1h 52m 37s\n- tempest-full-py3 http://logs.openstack.org/10/658910/1/check/tempest-full-py3/868addc/ : SUCCESS in 1h 40m 48s\n- openstack-tox-lower-constraints http://logs.openstack.org/10/658910/1/check/openstack-tox-lower-constraints/e24b2f7/ : SUCCESS in 5m 37s\n- openstack-tox-pep8 http://logs.openstack.org/10/658910/1/check/openstack-tox-pep8/e0358b9/ : SUCCESS in 4m 23s\n- openstack-tox-py27 http://logs.openstack.org/10/658910/1/check/openstack-tox-py27/a320770/ : SUCCESS in 5m 43s\n- openstack-tox-py35 http://logs.openstack.org/10/658910/1/check/openstack-tox-py35/154ca99/ : SUCCESS in 6m 09s\n- openstack-tox-py36 http://logs.openstack.org/10/658910/1/check/openstack-tox-py36/430b36e/ : SUCCESS in 4m 55s\n- openstack-tox-docs http://logs.openstack.org/10/658910/1/check/openstack-tox-docs/bed3b98/html/ : SUCCESS in 4m 57s\n- build-openstack-releasenotes http://logs.openstack.org/10/658910/1/check/build-openstack-releasenotes/1886d97/html/ : SUCCESS in 5m 23s","accounts_in_message":[],"_revision_number":1},{"id":"5f603bf8e7a4de2b847b985d6a988c7ba8aa68bd","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2019-05-14 12:46:18.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"b863ba5f088c98d1b110e26de75b0a001521bf27","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-14 12:46:29.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"122642627519bc416725c46e8d45058b4e13f015","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-14 14:29:34.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- requirements-check http://logs.openstack.org/10/658910/1/gate/requirements-check/0f5a286/ : SUCCESS in 3m 37s\n- tempest-full http://logs.openstack.org/10/658910/1/gate/tempest-full/7cacd19/ : FAILURE in 1h 32m 16s\n- tempest-full-py3 http://logs.openstack.org/10/658910/1/gate/tempest-full-py3/16d6f5e/ : SUCCESS in 1h 38m 15s\n- openstack-tox-lower-constraints http://logs.openstack.org/10/658910/1/gate/openstack-tox-lower-constraints/5070def/ : SUCCESS in 5m 09s\n- openstack-tox-pep8 http://logs.openstack.org/10/658910/1/gate/openstack-tox-pep8/955028b/ : SUCCESS in 4m 09s\n- openstack-tox-py27 http://logs.openstack.org/10/658910/1/gate/openstack-tox-py27/d259aec/ : SUCCESS in 4m 44s\n- openstack-tox-py35 http://logs.openstack.org/10/658910/1/gate/openstack-tox-py35/1e3e876/ : SUCCESS in 4m 58s\n- openstack-tox-py36 http://logs.openstack.org/10/658910/1/gate/openstack-tox-py36/b09da2f/ : SUCCESS in 4m 09s\n- openstack-tox-docs http://logs.openstack.org/10/658910/1/gate/openstack-tox-docs/023581c/html/ : SUCCESS in 4m 03s\n- build-openstack-releasenotes http://logs.openstack.org/10/658910/1/gate/build-openstack-releasenotes/eb57abb/html/ : SUCCESS in 4m 42s","accounts_in_message":[],"_revision_number":1},{"id":"f9ecf17a9675dd2a192757e9402798cff607e97d","author":{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},"date":"2019-05-14 14:33:17.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"1acecf1c44311ada70b83ea6adecf89680465c7f","author":{"_account_id":28935,"name":"98k","email":"18552437190@163.com","username":"98k"},"date":"2019-05-14 14:37:39.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"09658fc217e75300fe66e794fbc149566a849a16","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-14 16:36:00.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- requirements-check http://logs.openstack.org/10/658910/1/check/requirements-check/7e77546/ : SUCCESS in 3m 36s\n- tempest-full http://logs.openstack.org/10/658910/1/check/tempest-full/907a4df/ : SUCCESS in 2h 00m 14s\n- tempest-full-py3 http://logs.openstack.org/10/658910/1/check/tempest-full-py3/3ec3065/ : SUCCESS in 1h 57m 23s\n- openstack-tox-lower-constraints http://logs.openstack.org/10/658910/1/check/openstack-tox-lower-constraints/620e36c/ : SUCCESS in 5m 51s\n- openstack-tox-pep8 http://logs.openstack.org/10/658910/1/check/openstack-tox-pep8/fbec58e/ : SUCCESS in 5m 22s\n- openstack-tox-py27 http://logs.openstack.org/10/658910/1/check/openstack-tox-py27/ee9d303/ : SUCCESS in 5m 57s\n- openstack-tox-py35 http://logs.openstack.org/10/658910/1/check/openstack-tox-py35/a8a24e4/ : SUCCESS in 7m 05s\n- openstack-tox-py36 http://logs.openstack.org/10/658910/1/check/openstack-tox-py36/55276cc/ : SUCCESS in 6m 56s\n- openstack-tox-docs http://logs.openstack.org/10/658910/1/check/openstack-tox-docs/0dab963/html/ : SUCCESS in 4m 41s\n- build-openstack-releasenotes http://logs.openstack.org/10/658910/1/check/build-openstack-releasenotes/a59998c/html/ : SUCCESS in 5m 09s","accounts_in_message":[],"_revision_number":1},{"id":"f797ed05b4d88a27cff295e23ef69a280c841857","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-14 16:36:12.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"fe6da0220ba004adada30b2f4004558b3a94253f","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-14 18:30:35.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- requirements-check http://logs.openstack.org/10/658910/1/gate/requirements-check/de9405d/ : SUCCESS in 5m 35s\n- tempest-full http://logs.openstack.org/10/658910/1/gate/tempest-full/3490fac/ : SUCCESS in 1h 48m 43s\n- tempest-full-py3 http://logs.openstack.org/10/658910/1/gate/tempest-full-py3/9671c0d/ : SUCCESS in 1h 32m 41s\n- openstack-tox-lower-constraints http://logs.openstack.org/10/658910/1/gate/openstack-tox-lower-constraints/4031c5d/ : SUCCESS in 5m 14s\n- openstack-tox-pep8 http://logs.openstack.org/10/658910/1/gate/openstack-tox-pep8/a42866c/ : SUCCESS in 4m 19s\n- openstack-tox-py27 http://logs.openstack.org/10/658910/1/gate/openstack-tox-py27/835aaad/ : SUCCESS in 6m 35s\n- openstack-tox-py35 http://logs.openstack.org/10/658910/1/gate/openstack-tox-py35/9b5255f/ : SUCCESS in 7m 37s\n- openstack-tox-py36 http://logs.openstack.org/10/658910/1/gate/openstack-tox-py36/da25b4c/ : SUCCESS in 4m 17s\n- openstack-tox-docs http://logs.openstack.org/10/658910/1/gate/openstack-tox-docs/07a5c15/html/ : SUCCESS in 3m 53s\n- build-openstack-releasenotes http://logs.openstack.org/10/658910/1/gate/build-openstack-releasenotes/3f629f5/html/ : SUCCESS in 4m 24s","accounts_in_message":[],"_revision_number":1},{"id":"9b259eeec12cf17e4169be2dfa93c4a7c16e6bac","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-14 18:30:36.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"925bb81c7f79f71050461e06b8cd3d9f0eb47042","revisions":{"925bb81c7f79f71050461e06b8cd3d9f0eb47042":{"kind":"REWORK","_number":1,"created":"2019-05-13 20:10:13.000000000","uploader":{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},"ref":"refs/changes/10/658910/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.db","ref":"refs/changes/10/658910/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.db refs/changes/10/658910/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.db refs/changes/10/658910/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.db refs/changes/10/658910/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.db refs/changes/10/658910/1"}}},"commit":{"parents":[{"commit":"aef64ea8aea03caa68e9006570f0e8943553bcfc","subject":"Merge \"Replace git.openstack.org URLs with opendev.org URLs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.db/commit/aef64ea8aea03caa68e9006570f0e8943553bcfc"}]}],"author":{"name":"Ben Nemec","email":"bnemec@redhat.com","date":"2019-05-13 20:09:00.000000000","tz":0},"committer":{"name":"Ben Nemec","email":"bnemec@redhat.com","date":"2019-05-13 20:09:00.000000000","tz":0},"subject":"Cap Bandit below 1.6.0 and update Sphinx requirement","message":"Cap Bandit below 1.6.0 and update Sphinx requirement\n\nBandit 1.6.0 accidentally changed how the exclusion list option is\nhandled and breaks our use of it. Cap to the previous version until\nBandit has fixed the problem.\n\nSphinx 2.0 no longer works on python 2.7, so we need to start capping\nit there as well.\n\nChange-Id: If86c82e0f4a519baca664af79352846c4af9a01c\nReference: https://github.com/PyCQA/bandit/pull/489\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.db/commit/925bb81c7f79f71050461e06b8cd3d9f0eb47042"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.db/commit/925bb81c7f79f71050461e06b8cd3d9f0eb47042"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
