)]}'
{"id":"openstack%2Foslo.db~658911","triplet_id":"openstack%2Foslo.db~master~Ieb2e0dd6e53cf6a3a169d75cc493d8d2ee20d5af","project":"openstack/oslo.db","branch":"master","topic":"cap-bandit","hashtags":[],"change_id":"Ieb2e0dd6e53cf6a3a169d75cc493d8d2ee20d5af","subject":"Uncap Bandit","status":"ABANDONED","created":"2019-05-13 20:10:13.000000000","updated":"2019-05-22 14:17:44.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"021dcc85069f79417e0e1c6e8719c3cb0b27b3ad","_number":658911,"virtual_id_number":658911,"owner":{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":1,"date":"2019-05-13 22:07:49.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-05-13 22:07:49.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"60cfc52b000fc82ef9fffbe02dc729404e134b1b","author":{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},"date":"2019-05-13 20:10:13.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"096ff4f7880acc207fbccec12a0188e308338baf","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-13 22:07:49.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- requirements-check http://logs.openstack.org/11/658911/1/check/requirements-check/e4a54de/ : SUCCESS in 3m 25s\n- tempest-full http://logs.openstack.org/11/658911/1/check/tempest-full/ad73dc4/ : SUCCESS in 1h 48m 36s\n- tempest-full-py3 http://logs.openstack.org/11/658911/1/check/tempest-full-py3/d490a49/ : SUCCESS in 1h 39m 35s\n- openstack-tox-lower-constraints http://logs.openstack.org/11/658911/1/check/openstack-tox-lower-constraints/1092f98/ : SUCCESS in 5m 26s\n- openstack-tox-pep8 http://logs.openstack.org/11/658911/1/check/openstack-tox-pep8/8389c17/ : SUCCESS in 4m 28s\n- openstack-tox-py27 http://logs.openstack.org/11/658911/1/check/openstack-tox-py27/3fd3cd1/ : SUCCESS in 12m 41s\n- openstack-tox-py35 http://logs.openstack.org/11/658911/1/check/openstack-tox-py35/de9e66b/ : SUCCESS in 13m 18s\n- openstack-tox-py36 http://logs.openstack.org/11/658911/1/check/openstack-tox-py36/8e73d30/ : SUCCESS in 13m 02s\n- openstack-tox-docs http://logs.openstack.org/11/658911/1/check/openstack-tox-docs/25ccf86/html/ : SUCCESS in 14m 50s\n- build-openstack-releasenotes http://logs.openstack.org/11/658911/1/check/build-openstack-releasenotes/a026621/html/ : SUCCESS in 13m 42s","accounts_in_message":[],"_revision_number":1},{"id":"0376156854e436ecaa6677d4fd104fa14e59559b","author":{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},"date":"2019-05-22 14:17:44.000000000","message":"Abandoned\n\nAbandoning per http://lists.openstack.org/pipermail/openstack-discuss/2019-May/006432.html","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"cd8e7a8a72e20a909dd5bb7f04d1cbbd112d0420","revisions":{"cd8e7a8a72e20a909dd5bb7f04d1cbbd112d0420":{"kind":"REWORK","_number":1,"created":"2019-05-13 20:10:13.000000000","uploader":{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},"ref":"refs/changes/11/658911/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.db","ref":"refs/changes/11/658911/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.db refs/changes/11/658911/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.db refs/changes/11/658911/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.db refs/changes/11/658911/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.db refs/changes/11/658911/1"}}},"commit":{"parents":[{"commit":"925bb81c7f79f71050461e06b8cd3d9f0eb47042","subject":"Cap Bandit below 1.6.0 and update Sphinx requirement","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.db/commit/925bb81c7f79f71050461e06b8cd3d9f0eb47042"}]}],"author":{"name":"Ben Nemec","email":"bnemec@redhat.com","date":"2019-05-13 20:09:55.000000000","tz":0},"committer":{"name":"Ben Nemec","email":"bnemec@redhat.com","date":"2019-05-13 20:09:55.000000000","tz":0},"subject":"Uncap Bandit","message":"Uncap Bandit\n\nThis was capped due to a problem with how it processed exclusions.\nOnce that problem is fixed we want to uncap it and exlude only the\nspecific versions that are broken.\n\nChange-Id: Ieb2e0dd6e53cf6a3a169d75cc493d8d2ee20d5af\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.db/commit/cd8e7a8a72e20a909dd5bb7f04d1cbbd112d0420"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.db/commit/cd8e7a8a72e20a909dd5bb7f04d1cbbd112d0420"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
