)]}'
{"id":"openstack%2Foslo.db~659170","triplet_id":"openstack%2Foslo.db~stable%2Fpike~Ifb4a81d50e490e734e12986e51efd6bd3c339c31","project":"openstack/oslo.db","branch":"stable/pike","topic":"bug/1815210-stable/pike","hashtags":[],"change_id":"Ifb4a81d50e490e734e12986e51efd6bd3c339c31","subject":"exc_filters: fix deadlock detection for MariaDB/Galera cluster","status":"MERGED","created":"2019-05-14 21:19:53.000000000","updated":"2019-05-20 09:36:32.000000000","submitted":"2019-05-20 09:36:32.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"659170-1558344992333-38c90c1d","meta_rev_id":"72849e7ba7b3868e3dae3b72fe3102988f386531","_number":659170,"virtual_id_number":659170,"owner":{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9796,"name":"ChangBo Guo","email":"glongwave@gmail.com","username":"gcb"},{"value":0,"_account_id":11816,"name":"mike bayer","display_name":"Mike Bayer","email":"mike_mp@zzzcomputing.com","username":"zzzeek","status":"Red Hat"},{"value":2,"date":"2019-05-20 09:36:32.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":29037,"name":"Pierre-Samuel Le Stang","email":"pierre-samuel.le-stang@ovhcloud.com","username":"pslestang"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9796,"name":"ChangBo Guo","email":"glongwave@gmail.com","username":"gcb"},"all":[{"value":2,"date":"2019-05-20 07:44:32.000000000","_account_id":9796,"name":"ChangBo Guo","email":"glongwave@gmail.com","username":"gcb"},{"value":2,"date":"2019-05-14 22:33:53.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11816,"name":"mike bayer","display_name":"Mike Bayer","email":"mike_mp@zzzcomputing.com","username":"zzzeek","status":"Red Hat"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2019-05-15 07:18:23.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":29037,"name":"Pierre-Samuel Le Stang","email":"pierre-samuel.le-stang@ovhcloud.com","username":"pslestang"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9796,"name":"ChangBo Guo","email":"glongwave@gmail.com","username":"gcb"},"all":[{"value":1,"date":"2019-05-20 07:44:32.000000000","_account_id":9796,"name":"ChangBo Guo","email":"glongwave@gmail.com","username":"gcb"},{"value":0,"_account_id":11816,"name":"mike bayer","display_name":"Mike Bayer","email":"mike_mp@zzzcomputing.com","username":"zzzeek","status":"Red Hat"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":29037,"name":"Pierre-Samuel Le Stang","email":"pierre-samuel.le-stang@ovhcloud.com","username":"pslestang"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":9796,"name":"ChangBo Guo","email":"glongwave@gmail.com","username":"gcb"},{"value":0,"_account_id":11816,"name":"mike bayer","display_name":"Mike Bayer","email":"mike_mp@zzzcomputing.com","username":"zzzeek","status":"Red Hat"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":29037,"name":"Pierre-Samuel Le Stang","email":"pierre-samuel.le-stang@ovhcloud.com","username":"pslestang"}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9796,"name":"ChangBo Guo","email":"glongwave@gmail.com","username":"gcb"},{"_account_id":11816,"name":"mike bayer","display_name":"Mike Bayer","email":"mike_mp@zzzcomputing.com","username":"zzzeek","status":"Red Hat"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29037,"name":"Pierre-Samuel Le Stang","email":"pierre-samuel.le-stang@ovhcloud.com","username":"pslestang"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-05-14 22:33:53.000000000","updated_by":{"_account_id":11816,"name":"mike bayer","display_name":"Mike Bayer","email":"mike_mp@zzzcomputing.com","username":"zzzeek","status":"Red Hat"},"reviewer":{"_account_id":11816,"name":"mike bayer","display_name":"Mike Bayer","email":"mike_mp@zzzcomputing.com","username":"zzzeek","status":"Red Hat"},"state":"REVIEWER"},{"updated":"2019-05-15 07:18:23.000000000","updated_by":{"_account_id":29037,"name":"Pierre-Samuel Le Stang","email":"pierre-samuel.le-stang@ovhcloud.com","username":"pslestang"},"reviewer":{"_account_id":29037,"name":"Pierre-Samuel Le Stang","email":"pierre-samuel.le-stang@ovhcloud.com","username":"pslestang"},"state":"REVIEWER"},{"updated":"2019-05-20 07:44:32.000000000","updated_by":{"_account_id":9796,"name":"ChangBo Guo","email":"glongwave@gmail.com","username":"gcb"},"reviewer":{"_account_id":9796,"name":"ChangBo Guo","email":"glongwave@gmail.com","username":"gcb"},"state":"REVIEWER"},{"updated":"2019-05-20 09:36:32.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"ef6af0bcf61db1c9af421223e004adcd0c523d90","author":{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},"date":"2019-05-14 21:19:53.000000000","message":"Patch Set 1: Cherry Picked from branch stable/queens.","accounts_in_message":[],"_revision_number":1},{"id":"40b09bb266555014207679bf607ccf44597d9c1b","author":{"_account_id":11816,"name":"mike bayer","display_name":"Mike Bayer","email":"mike_mp@zzzcomputing.com","username":"zzzeek","status":"Red Hat"},"date":"2019-05-14 22:33:53.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"9135df2def2a7145978ed0cc53c120a70556fee3","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-14 23:00:59.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/70/659170/1/check/openstack-tox-pep8/424865c/ : SUCCESS in 4m 26s\n- openstack-tox-py27 http://logs.openstack.org/70/659170/1/check/openstack-tox-py27/d901fa1/ : SUCCESS in 5m 00s\n- openstack-tox-py35 http://logs.openstack.org/70/659170/1/check/openstack-tox-py35/c954ca8/ : SUCCESS in 5m 41s\n- openstack-tox-docs http://logs.openstack.org/70/659170/1/check/openstack-tox-docs/47572c7/html/ : SUCCESS in 4m 07s\n- tempest-full http://logs.openstack.org/70/659170/1/check/tempest-full/b56248c/ : SUCCESS in 1h 32m 43s","accounts_in_message":[],"_revision_number":1},{"id":"ad20202891b09b064a8ba60643f833ca9aa6a1d8","author":{"_account_id":29037,"name":"Pierre-Samuel Le Stang","email":"pierre-samuel.le-stang@ovhcloud.com","username":"pslestang"},"date":"2019-05-15 07:18:23.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"62f0ca1fac80e0672278ba14bae5a778f1a1cc10","author":{"_account_id":9796,"name":"ChangBo Guo","email":"glongwave@gmail.com","username":"gcb"},"date":"2019-05-20 07:44:32.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"ca1be57052522e95e9c6e74fe4ecdfcf2d43498a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-20 07:44:43.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"9888aefee8438a1daf113b052b3367761f750bbb","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-20 09:36:32.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"8b1f9ddc241b06a416cb84d4511429cdab008012","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-20 09:36:32.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/70/659170/1/gate/openstack-tox-pep8/ff721df/ : SUCCESS in 5m 34s\n- openstack-tox-py27 http://logs.openstack.org/70/659170/1/gate/openstack-tox-py27/d87deca/ : SUCCESS in 7m 59s\n- openstack-tox-py35 http://logs.openstack.org/70/659170/1/gate/openstack-tox-py35/a3d6c99/ : SUCCESS in 5m 53s\n- openstack-tox-docs http://logs.openstack.org/70/659170/1/gate/openstack-tox-docs/7e7d433/html/ : SUCCESS in 5m 13s\n- tempest-full http://logs.openstack.org/70/659170/1/gate/tempest-full/b8d5c94/ : SUCCESS in 1h 39m 07s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"30787bb61706f353b312917d87a784cdf1d0ee75","revisions":{"30787bb61706f353b312917d87a784cdf1d0ee75":{"kind":"REWORK","_number":1,"created":"2019-05-14 21:19:53.000000000","uploader":{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},"ref":"refs/changes/70/659170/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.db","ref":"refs/changes/70/659170/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.db refs/changes/70/659170/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.db refs/changes/70/659170/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.db refs/changes/70/659170/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.db refs/changes/70/659170/1"}}},"commit":{"parents":[{"commit":"2f040e34594ff4d20eb7e55a5cec12561bdf13e4","subject":"OpenDev Migration Patch","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.db/commit/2f040e34594ff4d20eb7e55a5cec12561bdf13e4"}]}],"author":{"name":"Pierre-Samuel Le Stang","email":"pierre-samuel.le-stang@corp.ovh.com","date":"2019-02-08 15:46:21.000000000","tz":60},"committer":{"name":"Dirk Mueller","email":"dirk@dmllr.de","date":"2019-05-14 21:19:53.000000000","tz":0},"subject":"exc_filters: fix deadlock detection for MariaDB/Galera cluster","message":"exc_filters: fix deadlock detection for MariaDB/Galera cluster\n\nMariasDB/Galera cluster server produces an error like:\n\n(1213, u\u0027Deadlock: wsrep aborted transaction\u0027)\n\nwhich is currently not detected properly and, thus, not retried.\n\nChange-Id: Ifb4a81d50e490e734e12986e51efd6bd3c339c31\nSigned-off-by: Pierre-Samuel Le Stang \u003cpierre-samuel.le-stang@corp.ovh.com\u003e\n(cherry picked from commit ae227eb3b9adb430cf6aa5a4f46f846e964fff91)\n(cherry picked from commit c114b932dbc7a2fe446b57bd08308fa85007c59f)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.db/commit/30787bb61706f353b312917d87a784cdf1d0ee75"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.db/commit/30787bb61706f353b312917d87a784cdf1d0ee75"}]},"branch":"refs/heads/stable/pike"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
