)]}'
{"id":"openstack%2Foslo.db~663984","triplet_id":"openstack%2Foslo.db~master~Idcb616fd3add4d58f22b91865cec8a54fe492093","project":"openstack/oslo.db","branch":"master","topic":"reset_nested_trans","hashtags":[],"change_id":"Idcb616fd3add4d58f22b91865cec8a54fe492093","subject":"Rollback existing nested transacvtion before restarting","status":"MERGED","created":"2019-06-07 15:54:07.000000000","updated":"2019-06-11 21:17:36.000000000","submitted":"2019-06-11 21:17:36.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":4,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"663984-1560287856125-216c3fa0","meta_rev_id":"0751ff6eeef13c8aee2f7238d85eb82344b91ccc","_number":663984,"virtual_id_number":663984,"owner":{"_account_id":11816,"name":"mike bayer","display_name":"Mike Bayer","email":"mike_mp@zzzcomputing.com","username":"zzzeek","status":"Red Hat"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},{"value":0,"_account_id":9796,"name":"ChangBo Guo","email":"glongwave@gmail.com","username":"gcb"},{"value":0,"date":"2019-06-10 17:02:05.000000000","_account_id":11816,"name":"mike bayer","display_name":"Mike Bayer","email":"mike_mp@zzzcomputing.com","username":"zzzeek","status":"Red Hat"},{"value":2,"date":"2019-06-11 21:17:35.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},"all":[{"value":2,"date":"2019-06-10 20:06:16.000000000","_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},{"value":2,"date":"2019-06-11 06:57:32.000000000","_account_id":9796,"name":"ChangBo Guo","email":"glongwave@gmail.com","username":"gcb"},{"value":0,"_account_id":11816,"name":"mike bayer","display_name":"Mike Bayer","email":"mike_mp@zzzcomputing.com","username":"zzzeek","status":"Red Hat"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9796,"name":"ChangBo Guo","email":"glongwave@gmail.com","username":"gcb"},"all":[{"value":0,"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},{"value":1,"date":"2019-06-11 06:57:32.000000000","_account_id":9796,"name":"ChangBo Guo","email":"glongwave@gmail.com","username":"gcb"},{"value":0,"_account_id":11816,"name":"mike bayer","display_name":"Mike Bayer","email":"mike_mp@zzzcomputing.com","username":"zzzeek","status":"Red Hat"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},{"value":0,"_account_id":9796,"name":"ChangBo Guo","email":"glongwave@gmail.com","username":"gcb"},{"value":0,"_account_id":11816,"name":"mike bayer","display_name":"Mike Bayer","email":"mike_mp@zzzcomputing.com","username":"zzzeek","status":"Red Hat"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},{"_account_id":9796,"name":"ChangBo Guo","email":"glongwave@gmail.com","username":"gcb"},{"_account_id":11816,"name":"mike bayer","display_name":"Mike Bayer","email":"mike_mp@zzzcomputing.com","username":"zzzeek","status":"Red Hat"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-06-10 20:06:16.000000000","updated_by":{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},"reviewer":{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},"state":"REVIEWER"},{"updated":"2019-06-11 06:57:32.000000000","updated_by":{"_account_id":9796,"name":"ChangBo Guo","email":"glongwave@gmail.com","username":"gcb"},"reviewer":{"_account_id":9796,"name":"ChangBo Guo","email":"glongwave@gmail.com","username":"gcb"},"state":"REVIEWER"},{"updated":"2019-06-11 21:17:35.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"d508b92adb159bde65a2da484d61f666421f43ba","author":{"_account_id":11816,"name":"mike bayer","display_name":"Mike Bayer","email":"mike_mp@zzzcomputing.com","username":"zzzeek","status":"Red Hat"},"date":"2019-06-07 15:54:07.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"948c7cd838d15834adbf280bcb99aefd77c22a72","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-07 17:54:23.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tempest-full http://logs.openstack.org/84/663984/1/check/tempest-full/2418c02/ : SUCCESS in 1h 47m 40s\n- tempest-full-py3 http://logs.openstack.org/84/663984/1/check/tempest-full-py3/0a379e1/ : SUCCESS in 1h 54m 43s\n- openstack-tox-lower-constraints http://logs.openstack.org/84/663984/1/check/openstack-tox-lower-constraints/6bcd328/ : SUCCESS in 5m 06s\n- openstack-tox-pep8 http://logs.openstack.org/84/663984/1/check/openstack-tox-pep8/9507b30/ : SUCCESS in 6m 08s\n- openstack-tox-py27 http://logs.openstack.org/84/663984/1/check/openstack-tox-py27/4023cbf/ : SUCCESS in 6m 04s\n- openstack-tox-py36 http://logs.openstack.org/84/663984/1/check/openstack-tox-py36/0167311/ : SUCCESS in 6m 19s\n- openstack-tox-docs http://logs.openstack.org/84/663984/1/check/openstack-tox-docs/8cb9844/html/ : SUCCESS in 8m 20s","accounts_in_message":[],"_revision_number":1},{"id":"b1f1c143bf1e93dede3241ef1de8a8a8036a44ff","author":{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},"date":"2019-06-10 15:46:14.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"f34319abb2f596f439406a3bebac9f575e604394","author":{"_account_id":11816,"name":"mike bayer","display_name":"Mike Bayer","email":"mike_mp@zzzcomputing.com","username":"zzzeek","status":"Red Hat"},"date":"2019-06-10 17:02:05.000000000","message":"Patch Set 1:\n\n(3 comments)","accounts_in_message":[],"_revision_number":1},{"id":"8c0db998a94bd1d4876d9146bfdd2cdbb485c602","author":{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},"date":"2019-06-10 20:06:16.000000000","message":"Patch Set 1: Code-Review+2\n\nOkay, sounds good. Thanks for the explanation.","accounts_in_message":[],"_revision_number":1},{"id":"8a36031c283c1395d0bbbdc417e9dfb330c3cd56","author":{"_account_id":9796,"name":"ChangBo Guo","email":"glongwave@gmail.com","username":"gcb"},"date":"2019-06-11 06:57:32.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"81515aeb050eb0eb437d828a448be03e5d9a86f6","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-11 06:57:44.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"58e0723796088ba70d8297c423f96f09133e1da4","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-11 09:04:37.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tempest-full http://logs.openstack.org/84/663984/1/gate/tempest-full/887f160/ : TIMED_OUT in 2h 05m 54s\n- tempest-full-py3 http://logs.openstack.org/84/663984/1/gate/tempest-full-py3/2690988/ : SUCCESS in 1h 42m 30s\n- openstack-tox-lower-constraints http://logs.openstack.org/84/663984/1/gate/openstack-tox-lower-constraints/91aec77/ : SUCCESS in 5m 13s\n- openstack-tox-pep8 http://logs.openstack.org/84/663984/1/gate/openstack-tox-pep8/f19f27f/ : SUCCESS in 4m 30s\n- openstack-tox-py27 http://logs.openstack.org/84/663984/1/gate/openstack-tox-py27/06cf1f6/ : SUCCESS in 6m 49s\n- openstack-tox-py36 http://logs.openstack.org/84/663984/1/gate/openstack-tox-py36/fff2c93/ : SUCCESS in 5m 30s\n- openstack-tox-docs http://logs.openstack.org/84/663984/1/gate/openstack-tox-docs/ec477af/html/ : SUCCESS in 4m 28s","accounts_in_message":[],"_revision_number":1},{"id":"faaf4fda7545b5fd5ca1e2181bfc914a8d418158","author":{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},"date":"2019-06-11 17:24:38.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"4bf184b4a0ac62684a2d49d1f7daf8c2ff8e5f90","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-11 19:26:52.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tempest-full http://logs.openstack.org/84/663984/1/check/tempest-full/6b9eb12/ : SUCCESS in 1h 53m 50s\n- tempest-full-py3 http://logs.openstack.org/84/663984/1/check/tempest-full-py3/5e989f4/ : SUCCESS in 1h 47m 04s\n- openstack-tox-lower-constraints http://logs.openstack.org/84/663984/1/check/openstack-tox-lower-constraints/3fc7b13/ : SUCCESS in 5m 53s\n- openstack-tox-pep8 http://logs.openstack.org/84/663984/1/check/openstack-tox-pep8/e41704c/ : SUCCESS in 4m 08s\n- openstack-tox-py27 http://logs.openstack.org/84/663984/1/check/openstack-tox-py27/caf81e4/ : SUCCESS in 5m 41s\n- openstack-tox-py36 http://logs.openstack.org/84/663984/1/check/openstack-tox-py36/5dc1f72/ : SUCCESS in 6m 19s\n- openstack-tox-docs http://logs.openstack.org/84/663984/1/check/openstack-tox-docs/45cb072/html/ : SUCCESS in 6m 45s","accounts_in_message":[],"_revision_number":1},{"id":"a71f5215bea09c76eeb123886f7639fbe9d43d0d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-11 19:27:06.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"c19fd2992d029119437fa983e3dab6bd527c5781","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-11 21:17:35.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- tempest-full http://logs.openstack.org/84/663984/1/gate/tempest-full/4f9b1ef/ : SUCCESS in 1h 44m 41s\n- tempest-full-py3 http://logs.openstack.org/84/663984/1/gate/tempest-full-py3/2110478/ : SUCCESS in 1h 36m 46s\n- openstack-tox-lower-constraints http://logs.openstack.org/84/663984/1/gate/openstack-tox-lower-constraints/a9ccda1/ : SUCCESS in 5m 10s\n- openstack-tox-pep8 http://logs.openstack.org/84/663984/1/gate/openstack-tox-pep8/88f473f/ : SUCCESS in 5m 30s\n- openstack-tox-py27 http://logs.openstack.org/84/663984/1/gate/openstack-tox-py27/9288afc/ : SUCCESS in 5m 13s\n- openstack-tox-py36 http://logs.openstack.org/84/663984/1/gate/openstack-tox-py36/6ce67ab/ : SUCCESS in 4m 44s\n- openstack-tox-docs http://logs.openstack.org/84/663984/1/gate/openstack-tox-docs/c4230d6/html/ : SUCCESS in 6m 02s","accounts_in_message":[],"_revision_number":1},{"id":"7be0d5a20906bd75ac012287624fedefb71c0a1c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-11 21:17:36.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"459de143578367c00fb508edeea68eb96531724c","revisions":{"459de143578367c00fb508edeea68eb96531724c":{"kind":"REWORK","_number":1,"created":"2019-06-07 15:54:07.000000000","uploader":{"_account_id":11816,"name":"mike bayer","display_name":"Mike Bayer","email":"mike_mp@zzzcomputing.com","username":"zzzeek","status":"Red Hat"},"ref":"refs/changes/84/663984/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.db","ref":"refs/changes/84/663984/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.db refs/changes/84/663984/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.db refs/changes/84/663984/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.db refs/changes/84/663984/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.db refs/changes/84/663984/1"}}},"commit":{"parents":[{"commit":"2f8250f3f747473c8f4258b148f2f7ded0a95084","subject":"Dropping the py35 testing","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.db/commit/2f8250f3f747473c8f4258b148f2f7ded0a95084"}]}],"author":{"name":"Mike Bayer","email":"mike_mp@zzzcomputing.com","date":"2019-06-07 15:52:40.000000000","tz":-240},"committer":{"name":"Mike Bayer","email":"mike_mp@zzzcomputing.com","date":"2019-06-07 15:53:46.000000000","tz":-240},"subject":"Rollback existing nested transacvtion before restarting","message":"Rollback existing nested transacvtion before restarting\n\nSQLAlchemy 1.4 will be adding stricter rules to Connection objects\nsuch that a new transaction cannot be started while an old one hasn\u0027t\nbeen cleared.   Ensure this particular test fixture clears\nthe \"savepoint\" transaction first before starting a new one.\n\nReferences: https://github.com/sqlalchemy/sqlalchemy/issues/4712\nChange-Id: Idcb616fd3add4d58f22b91865cec8a54fe492093\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.db/commit/459de143578367c00fb508edeea68eb96531724c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.db/commit/459de143578367c00fb508edeea68eb96531724c"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
