)]}'
{"id":"openstack%2Foslo.db~667623","triplet_id":"openstack%2Foslo.db~master~I6b237ecf55ff440b5f5c956ee849a1927bb15441","project":"openstack/oslo.db","branch":"master","topic":"bindep-fallback","hashtags":[],"change_id":"I6b237ecf55ff440b5f5c956ee849a1927bb15441","subject":"Add local bindep.txt","status":"MERGED","created":"2019-06-26 14:04:55.000000000","updated":"2019-07-04 10:48:43.000000000","submitted":"2019-07-04 10:48:43.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"667623-1562237323172-5870c8eb","meta_rev_id":"98756e78ae55bc60fc47ac1180b55dde5bd7521f","_number":667623,"virtual_id_number":667623,"owner":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"date":"2019-07-01 12:09:06.000000000","_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"date":"2019-06-28 15:15:13.000000000","_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":2,"date":"2019-07-04 10:48:42.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},"all":[{"value":2,"date":"2019-06-26 18:35:48.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},{"value":1,"date":"2019-06-27 12:59:21.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":1,"date":"2019-07-04 00:58:48.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},{"value":2,"date":"2019-07-04 09:07:00.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":0,"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},{"value":1,"date":"2019-07-04 09:07:00.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-06-26 18:35:48.000000000","updated_by":{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},"reviewer":{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},"state":"REVIEWER"},{"updated":"2019-06-27 12:59:21.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2019-06-28 15:15:13.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2019-07-04 00:58:48.000000000","updated_by":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"reviewer":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"state":"REVIEWER"},{"updated":"2019-07-04 09:07:00.000000000","updated_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"reviewer":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"state":"REVIEWER"},{"updated":"2019-07-04 10:48:42.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"4f3c43ca1fad6a31cf90f6711b971435bbcc0a4d","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2019-06-26 14:04:55.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"5f1fd3b0c7d5d2813bad6e0ec31e15b537e99c7c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-26 15:55:43.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tempest-full http://logs.openstack.org/23/667623/1/check/tempest-full/cbab076/ : SUCCESS in 1h 41m 18s\n- tempest-full-py3 http://logs.openstack.org/23/667623/1/check/tempest-full-py3/c9e995a/ : SUCCESS in 1h 29m 35s\n- openstack-tox-lower-constraints http://logs.openstack.org/23/667623/1/check/openstack-tox-lower-constraints/ce7ed45/ : SUCCESS in 4m 33s\n- openstack-tox-pep8 http://logs.openstack.org/23/667623/1/check/openstack-tox-pep8/050ea03/ : SUCCESS in 3m 43s\n- openstack-tox-py27 http://logs.openstack.org/23/667623/1/check/openstack-tox-py27/7455a72/ : SUCCESS in 4m 23s\n- openstack-tox-py36 http://logs.openstack.org/23/667623/1/check/openstack-tox-py36/54450fc/ : SUCCESS in 5m 25s\n- openstack-tox-docs http://logs.openstack.org/23/667623/1/check/openstack-tox-docs/46e7e6a/html/ : SUCCESS in 3m 35s\n- build-openstack-releasenotes http://logs.openstack.org/23/667623/1/check/build-openstack-releasenotes/7dc0c0d/html/ : SUCCESS in 4m 06s","accounts_in_message":[],"_revision_number":1},{"id":"7b767c3cd7b93a0aec48941b4f1d569bd2b000c0","author":{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},"date":"2019-06-26 18:35:48.000000000","message":"Patch Set 1: Code-Review+2\n\nWorking for me, thanks.","accounts_in_message":[],"_revision_number":1},{"id":"8049ead36a76a4c74d05754d0eed6535497c87cb","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2019-06-27 12:59:21.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"7e9294c9429854d7a947fe19809ad30bdcfbbaa4","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2019-06-28 15:15:13.000000000","message":"Patch Set 1:\n\nI facing an issue locally, maybe it missing something on my side...\n```\n$ bindep --brief                      \nTraceback (most recent call last):\n  File \"/home/hberaud/.local/bin/bindep\", line 10, in \u003cmodule\u003e\n    sys.exit(main())\n  File \"/home/hberaud/.local/lib/python3.7/site-packages/bindep/main.py\", line 77, in main\n    errors \u003d depends.check_rules(rules)\n  File \"/home/hberaud/.local/lib/python3.7/site-packages/bindep/depends.py\", line 258, in check_rules\n    installed \u003d self.platform.get_pkg_version(rule[0])\n  File \"/home/hberaud/.local/lib/python3.7/site-packages/bindep/depends.py\", line 481, in get_pkg_version\n    stderr\u003dsubprocess.STDOUT).decode(getpreferredencoding(False))\n  File \"/usr/lib64/python3.7/subprocess.py\", line 395, in check_output\n    **kwargs).stdout\n  File \"/usr/lib64/python3.7/subprocess.py\", line 487, in run\n    output\u003dstdout, stderr\u003dstderr)\nsubprocess.CalledProcessError: Command \u0027[\u0027rpm\u0027, \u0027--qf\u0027, \u0027%{NAME} %|EPOCH?{%{EPOCH}:}|%{VERSION}-%{RELEASE}\\n\u0027, \u0027--whatprovides\u0027, \u0027-q\u0027, \u0027postgresql\u0027]\u0027 returned non-zero exit status 1.\n```\n\nWhen I try to play:\n```\n$ rpm --qf --whatprovides -q postgresql \npackage not installed\n```\n\nDo you have an idea?\n\nI guess it\u0027s not really an issue because we mostly want to use it for the Zuul/CI side.","accounts_in_message":[],"_revision_number":1},{"id":"9674ca4e0ae4458ebb71ed571354ee094295d24b","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2019-07-01 12:09:06.000000000","message":"Patch Set 1:\n\n@Hervé Beraud\n\nI haven\u0027t tested with RH (I guess you have that), but with other distros what I get is this:\n\n```\n# rpm --qf --whatprovides -q postgresql\npackage postgresql is not installed\n# echo $?\n1\n# pip3.7 install bindep\n[..]\nSuccessfully installed Parsley-1.3 bindep-2.8.1 distro-1.4.0 pbr-5.3.1\n# bindep --brief\npostgresql\n# echo $?\n1\n#\n```","accounts_in_message":[],"_revision_number":1},{"id":"1d675010848401710cf7c080ccac6285048c22f1","author":{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},"date":"2019-07-03 18:53:15.000000000","message":"Patch Set 1:\n\nNote that oslo.db\u0027s gate is currently blocked on this. Also, when I tested the new tox target on my fedora system it worked as expected (correctly identified that I was missing the postgres client).\n\nGiven that previously we had no way to automatically install distro deps on any platform, I\u0027m inclined to go with this and figure out any issues on fedora as a followup.","accounts_in_message":[],"_revision_number":1},{"id":"ecaa268403effa5a143bda28fbf881d333141241","author":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"date":"2019-07-04 00:58:48.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"267b7ecbf9e55a40e384278d65f7f14b353f265b","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2019-07-04 09:07:00.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"375cf12d06f26650d144b6d1acd25b8780f4f926","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-04 09:07:14.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"abe2054af22dc5f2a533e93656bb1900e780f92c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-04 10:48:42.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- tempest-full http://logs.openstack.org/23/667623/1/gate/tempest-full/128012f/ : SUCCESS in 1h 39m 55s\n- tempest-full-py3 http://logs.openstack.org/23/667623/1/gate/tempest-full-py3/037c64e/ : SUCCESS in 1h 32m 42s\n- openstack-tox-lower-constraints http://logs.openstack.org/23/667623/1/gate/openstack-tox-lower-constraints/818c400/ : SUCCESS in 4m 27s\n- openstack-tox-pep8 http://logs.openstack.org/23/667623/1/gate/openstack-tox-pep8/0b3f4ea/ : SUCCESS in 4m 46s\n- openstack-tox-py27 http://logs.openstack.org/23/667623/1/gate/openstack-tox-py27/22ffb51/ : SUCCESS in 6m 08s\n- openstack-tox-py36 http://logs.openstack.org/23/667623/1/gate/openstack-tox-py36/aa13c85/ : SUCCESS in 4m 30s\n- openstack-tox-docs http://logs.openstack.org/23/667623/1/gate/openstack-tox-docs/791c9a8/html/ : SUCCESS in 5m 25s\n- build-openstack-releasenotes http://logs.openstack.org/23/667623/1/gate/build-openstack-releasenotes/bc79afa/html/ : SUCCESS in 4m 53s","accounts_in_message":[],"_revision_number":1},{"id":"c8ea042c221fb09fe803f0e6af70cfc29a3c8d36","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-04 10:48:43.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"f34e53ebf7670f51f16d5d96412a42fa7f3a8232","revisions":{"f34e53ebf7670f51f16d5d96412a42fa7f3a8232":{"kind":"REWORK","_number":1,"created":"2019-06-26 14:04:55.000000000","uploader":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"ref":"refs/changes/23/667623/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.db","ref":"refs/changes/23/667623/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.db refs/changes/23/667623/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.db refs/changes/23/667623/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.db refs/changes/23/667623/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.db refs/changes/23/667623/1"}}},"commit":{"parents":[{"commit":"459de143578367c00fb508edeea68eb96531724c","subject":"Rollback existing nested transacvtion before restarting","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.db/commit/459de143578367c00fb508edeea68eb96531724c"}]}],"author":{"name":"Elod Illes","email":"elod.illes@est.tech","date":"2019-06-26 14:03:39.000000000","tz":120},"committer":{"name":"Elod Illes","email":"elod.illes@est.tech","date":"2019-06-26 14:03:39.000000000","tz":120},"subject":"Add local bindep.txt","message":"Add local bindep.txt\n\nAs it was announced [1] global bindep-fallback.txt was removed and now\nprojects need to have a local bindep.txt to be able to install binary\ndependencies for testing.\n\nIn test jobs the script tools/test-setup.sh is called which requires\nmysql and postgres servers and clients to be installed.\n\n[1] http://lists.openstack.org/pipermail/openstack-discuss/2019-June/007272.html\n\nChange-Id: I6b237ecf55ff440b5f5c956ee849a1927bb15441\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.db/commit/f34e53ebf7670f51f16d5d96412a42fa7f3a8232"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.db/commit/f34e53ebf7670f51f16d5d96412a42fa7f3a8232"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
