)]}'
{"id":"openstack%2Foslo.db~668863","triplet_id":"openstack%2Foslo.db~stable%2Frocky~I6b237ecf55ff440b5f5c956ee849a1927bb15441","project":"openstack/oslo.db","branch":"stable/rocky","topic":"bindep-fallback","hashtags":[],"change_id":"I6b237ecf55ff440b5f5c956ee849a1927bb15441","subject":"Add local bindep.txt","status":"MERGED","created":"2019-07-03 11:08:31.000000000","updated":"2019-07-08 05:04:06.000000000","submitted":"2019-07-08 05:04:05.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"668863-1562562246171-6efaa91b","meta_rev_id":"893424c522d7635c703070b4f570b013df50b232","_number":668863,"virtual_id_number":668863,"owner":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9796,"name":"ChangBo Guo","email":"glongwave@gmail.com","username":"gcb"},{"value":0,"date":"2019-07-03 11:43:02.000000000","_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":2,"date":"2019-07-08 05:04:05.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9796,"name":"ChangBo Guo","email":"glongwave@gmail.com","username":"gcb"},"all":[{"value":2,"date":"2019-07-08 03:08:36.000000000","_account_id":9796,"name":"ChangBo Guo","email":"glongwave@gmail.com","username":"gcb"},{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":2,"date":"2019-07-04 12:40:04.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9796,"name":"ChangBo Guo","email":"glongwave@gmail.com","username":"gcb"},"all":[{"value":1,"date":"2019-07-08 03:08:36.000000000","_account_id":9796,"name":"ChangBo Guo","email":"glongwave@gmail.com","username":"gcb"},{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":9796,"name":"ChangBo Guo","email":"glongwave@gmail.com","username":"gcb"},{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9796,"name":"ChangBo Guo","email":"glongwave@gmail.com","username":"gcb"},{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-07-04 12:40:04.000000000","updated_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"reviewer":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"state":"REVIEWER"},{"updated":"2019-07-08 03:08:36.000000000","updated_by":{"_account_id":9796,"name":"ChangBo Guo","email":"glongwave@gmail.com","username":"gcb"},"reviewer":{"_account_id":9796,"name":"ChangBo Guo","email":"glongwave@gmail.com","username":"gcb"},"state":"REVIEWER"},{"updated":"2019-07-08 05:04:05.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"08558a8a934f2b3a9ed2d32c7f864c9dfc26d016","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2019-07-03 11:08:31.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"40c8d51fb5f8f9aa0209bc5eef95c0bd3f0cd420","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2019-07-03 11:43:02.000000000","message":"Patch Set 1:\n\nThis patch is needed as periodic job fails:\n\nhttp://logs.openstack.org/periodic-stable/opendev.org/openstack/oslo.db/stable/rocky/openstack-tox-py27/d9b2525/job-output.txt.gz#_2019-07-03_07_27_49_609604","accounts_in_message":[],"_revision_number":1},{"id":"4d95736ac16628e82d1e9c9e424b057f101dd6c9","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-03 13:00:39.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/63/668863/1/check/openstack-tox-pep8/2ce59b4/ : SUCCESS in 3m 52s\n- openstack-tox-py27 http://logs.openstack.org/63/668863/1/check/openstack-tox-py27/7e319d2/ : SUCCESS in 4m 33s\n- openstack-tox-py35 http://logs.openstack.org/63/668863/1/check/openstack-tox-py35/36d8908/ : SUCCESS in 6m 10s\n- openstack-tox-docs http://logs.openstack.org/63/668863/1/check/openstack-tox-docs/9679abf/html/ : SUCCESS in 5m 01s\n- tempest-full http://logs.openstack.org/63/668863/1/check/tempest-full/2a6af33/ : SUCCESS in 1h 46m 11s\n- build-openstack-releasenotes http://logs.openstack.org/63/668863/1/check/build-openstack-releasenotes/b8e3702/html/ : SUCCESS in 4m 20s\n- openstack-tox-lower-constraints http://logs.openstack.org/63/668863/1/check/openstack-tox-lower-constraints/132eb79/ : SUCCESS in 7m 02s","accounts_in_message":[],"_revision_number":1},{"id":"8c6fbbd2ecea4da52bdac97eeb4f265816f21304","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2019-07-04 12:40:04.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"309e759d4cb25a7d09bcf4cb75225d09935ebe2a","author":{"_account_id":9796,"name":"ChangBo Guo","email":"glongwave@gmail.com","username":"gcb"},"date":"2019-07-08 03:08:36.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"9209c731d4c7262ba4526bc1a781603b4960e24c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-08 03:08:47.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"af673844530b091e75dd0a2097db3d4e19c01b23","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-08 05:04:05.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/63/668863/1/gate/openstack-tox-pep8/8b155b7/ : SUCCESS in 4m 43s\n- openstack-tox-py27 http://logs.openstack.org/63/668863/1/gate/openstack-tox-py27/deeecca/ : SUCCESS in 5m 07s\n- openstack-tox-py35 http://logs.openstack.org/63/668863/1/gate/openstack-tox-py35/2786df7/ : SUCCESS in 4m 29s\n- openstack-tox-docs http://logs.openstack.org/63/668863/1/gate/openstack-tox-docs/b9cd73b/html/ : SUCCESS in 4m 16s\n- tempest-full http://logs.openstack.org/63/668863/1/gate/tempest-full/cc806ef/ : SUCCESS in 1h 32m 53s\n- build-openstack-releasenotes http://logs.openstack.org/63/668863/1/gate/build-openstack-releasenotes/a44d65d/html/ : SUCCESS in 3m 49s\n- openstack-tox-lower-constraints http://logs.openstack.org/63/668863/1/gate/openstack-tox-lower-constraints/6f676f9/ : SUCCESS in 5m 06s","accounts_in_message":[],"_revision_number":1},{"id":"d09ea0d866deef6b06a931f2bb19f5c9bc6a24db","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-08 05:04:06.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"d37a0eb3d5b20234592a6a514f5a4d7c75b5c763","revisions":{"d37a0eb3d5b20234592a6a514f5a4d7c75b5c763":{"kind":"REWORK","_number":1,"created":"2019-07-03 11:08:31.000000000","uploader":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"ref":"refs/changes/63/668863/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.db","ref":"refs/changes/63/668863/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.db refs/changes/63/668863/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.db refs/changes/63/668863/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.db refs/changes/63/668863/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.db refs/changes/63/668863/1"}}},"commit":{"parents":[{"commit":"249c65b5010941855d77d642c508177443a5c371","subject":"OpenDev Migration Patch","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.db/commit/249c65b5010941855d77d642c508177443a5c371"}]}],"author":{"name":"Elod Illes","email":"elod.illes@est.tech","date":"2019-06-26 14:03:39.000000000","tz":120},"committer":{"name":"Elod Illes","email":"elod.illes@est.tech","date":"2019-07-03 11:08:02.000000000","tz":120},"subject":"Add local bindep.txt","message":"Add local bindep.txt\n\nAs it was announced [1] global bindep-fallback.txt was removed and now\nprojects need to have a local bindep.txt to be able to install binary\ndependencies for testing.\n\nIn test jobs the script tools/test-setup.sh is called which requires\nmysql and postgres servers and clients to be installed.\n\n[1] http://lists.openstack.org/pipermail/openstack-discuss/2019-June/007272.html\n\nChange-Id: I6b237ecf55ff440b5f5c956ee849a1927bb15441\n(cherry picked from commit f34e53ebf7670f51f16d5d96412a42fa7f3a8232)\n(cherry picked from commit 67904312ab7fff2939f5115125994fc3753be334)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.db/commit/d37a0eb3d5b20234592a6a514f5a4d7c75b5c763"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.db/commit/d37a0eb3d5b20234592a6a514f5a4d7c75b5c763"}]},"branch":"refs/heads/stable/rocky"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
