)]}'
{"id":"openstack%2Foslo.db~714833","triplet_id":"openstack%2Foslo.db~stable%2Ftrain~Ie5d7b2b5f60efe9612cef396ed58f9d0378b9cf7","project":"openstack/oslo.db","branch":"stable/train","topic":"bug/1844929","hashtags":[],"change_id":"Ie5d7b2b5f60efe9612cef396ed58f9d0378b9cf7","subject":"DNM: investigating bug 1844929","status":"ABANDONED","created":"2020-03-25 04:49:57.000000000","updated":"2020-04-09 23:37:38.000000000","total_comment_count":25,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"bde99574db53b1ad1f8826ad250c10d4a0caf824","_number":714833,"virtual_id_number":714833,"owner":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2020-04-05 01:31:42.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-04-05 01:31:42.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"72ee6dec5116efbc3d4f2c899e94612c81c230c9","author":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},"date":"2020-03-25 04:49:57.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"cad51f3e36d65334588001d4af904a34444e598d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-25 06:53:09.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- tempest-full https://zuul.opendev.org/t/openstack/build/0010799cc93645c787b7d5c8d1e9012a : SUCCESS in 1h 25m 18s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/373457f783d045d091aa085800d13bcf : POST_FAILURE in 2h 02m 22s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/e73832a4a41c4038b6187f4f25e69a32 : SUCCESS in 4m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4837c1d6e6ed4a95a9be152aa7764668 : SUCCESS in 3m 39s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/eac071e878894eb9b469ddfbdd6e3b1f : SUCCESS in 5m 07s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b3f70af4deef445db24668d65d58154f : SUCCESS in 3m 56s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/f0bac9a7c8bf4e4eb2815399fd5f5942 : SUCCESS in 7m 01s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/95b8f5dac4f1486ea17f76d4a10366de : SUCCESS in 4m 19s","accounts_in_message":[],"_revision_number":1},{"id":"17c916cc0d3776dfd631b4e32a5086bd8a1ed5e2","author":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},"date":"2020-03-26 21:34:12.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"a015c2824b00ee90258c8cd4d6fff37c6b58e9f0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-26 23:16:38.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- tempest-full https://zuul.opendev.org/t/openstack/build/246ed66184044859928258a8d3dd6c4e : FAILURE in 12m 54s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/fc9387bc872643da885582607d9daf00 : SUCCESS in 1h 40m 05s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/51aa3eca9a3047269b7b8b4a6b74c1da : SUCCESS in 4m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5a3465bc7d444089b75adce9d574f974 : SUCCESS in 4m 11s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/37cdfc9ef0424e29a8dced8d885e6784 : SUCCESS in 5m 02s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/34461d9211e0426f9ea2f11ac34f9775 : SUCCESS in 4m 49s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/6d005151ed0347a0bf5bb9afb69c53b0 : SUCCESS in 5m 35s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3df1d5e49c544b6da3c61e4da5bb3a19 : SUCCESS in 4m 46s","accounts_in_message":[],"_revision_number":2},{"id":"d754ac21278f4818450955d9cc5d5a538d20b3ce","author":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},"date":"2020-03-27 16:13:30.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"9a1872500fc6afb14de412e5bdfb189107d94eca","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-27 18:20:28.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tempest-full https://zuul.opendev.org/t/openstack/build/3c190503048d4ee5a4e2cef63fdbda4c : SUCCESS in 1h 59m 04s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/f34ba528f7b644af8aded6c2d3d239fe : SUCCESS in 1h 31m 11s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/3f374f14eeef4bfe92cce74ac54c5d6e : SUCCESS in 6m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/afe0f6bc7e04441ba28a0411054f8b3d : SUCCESS in 7m 44s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/f502988b76784b98b514d9878f39658e : SUCCESS in 5m 40s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d95f2561058548c5b6b19c2166367fc3 : SUCCESS in 5m 10s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/904a3c7a970449a28187c6afcdbfbdc4 : SUCCESS in 6m 04s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7655574cd82a45d69a6d2dff8f0d29e0 : SUCCESS in 5m 52s","accounts_in_message":[],"_revision_number":3},{"id":"cbcf1c448d056bc93c700318741dbf8de679129c","author":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},"date":"2020-03-31 20:19:18.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"a1aad348caf953abc03aa8f55dcc71a62162acc7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-31 22:32:27.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tempest-full https://zuul.opendev.org/t/openstack/build/6c64707a19434a08b4bbc97b2e2e3051 : SUCCESS in 2h 08m 54s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/44871a15c8c140a99134bdd0bba2c33a : SUCCESS in 1h 40m 02s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/56a321b7e7de4337af89a424f1a42b05 : SUCCESS in 5m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5b1b2481990b4d95866597377ea47a24 : SUCCESS in 4m 57s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/a478520a27bc4898bcf6fd36fd8068be : SUCCESS in 6m 24s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/97af4bdfd4c245bfafe5892ec52fef73 : SUCCESS in 5m 24s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/e690f2dd94a64771bf254afd06991b81 : SUCCESS in 8m 30s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5eced663c26a4e20829205aa48487459 : SUCCESS in 6m 04s","accounts_in_message":[],"_revision_number":4},{"id":"58438f4785363738a42527411db53c4ec6172265","author":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},"date":"2020-04-04 00:56:18.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"2a004b84fc3e4bf4cc17cf4c0598af03b1239fdf","author":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},"date":"2020-04-04 00:56:29.000000000","message":"Topic changed from bug/1844929-stable/train to bug/1844929","accounts_in_message":[],"_revision_number":5},{"id":"5734745c2cb401bf77dd4cddae457044e3c4461b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-04 02:32:12.000000000","message":"Patch Set 5: Verified-1\n\n(13 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- tempest-full https://zuul.opendev.org/t/openstack/build/0c2a788649b347e09a5ab8a033cb2686 : FAILURE in 29m 29s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/bb75e22a212d4e359cae082b1521278c : SUCCESS in 1h 34m 07s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d4e886ab7cfc4ffe9f87fd223770fa78 : SUCCESS in 5m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7f026586d92b46f1a6eecf7bbad45f5a : SUCCESS in 3m 54s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/afb87f5f85c9411d845cfd1a0a4fb8bf : SUCCESS in 6m 05s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9f1fbedda00a421397e6b3c9dde4c447 : SUCCESS in 4m 55s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/2e8a9919b9b94c9cbdffd1f5a1118d9b : SUCCESS in 4m 54s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f021557628cc4c67bcb0ba2478e974eb : SUCCESS in 4m 35s\n\nWarning:\n  Job openstack-tox-py36: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path .tox/py36/lib/python3.6/site-packages/pymysql/cursors.py in the commit\u0027\n  Job openstack-tox-py36: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path .tox/py36/lib/python3.6/site-packages/sqlalchemy/orm/mapper.py in the commit\u0027\n  Job openstack-tox-lower-constraints: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path .tox/lower-constraints/lib/python3.6/site-packages/sqlalchemy/orm/mapper.py in the commit\u0027\n  Job openstack-tox-lower-constraints: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path .tox/lower-constraints/lib/python3.6/site-packages/sqlalchemy/orm/session.py in the commit\u0027\n  Job openstack-tox-lower-constraints: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path .tox/lower-constraints/lib/python3.6/site-packages/pymysql/cursors.py in the commit\u0027\n  Job openstack-tox-py27: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path .tox/py27/lib/python2.7/site-packages/pymysql/cursors.py in the commit\u0027\n  Job openstack-tox-py27: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path .tox/py27/lib/python2.7/site-packages/sqlalchemy/util/langhelpers.py in the commit\u0027\n  Job openstack-tox-py27: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path .tox/py27/lib/python2.7/site-packages/sqlalchemy/orm/mapper.py in the commit\u0027\n  Job openstack-tox-py37: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path .tox/py37/lib/python3.7/site-packages/pymysql/cursors.py in the commit\u0027\n  Job openstack-tox-py37: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path .tox/py37/lib/python3.7/site-packages/sqlalchemy/orm/mapper.py in the commit\u0027\n  Comments left for invalid file .tox/lower-constraints/lib/python3.6/site-packages/sqlalchemy/orm/mapper.py\n  Comments left for invalid file .tox/py36/lib/python3.6/site-packages/sqlalchemy/orm/mapper.py\n  Comments left for invalid file .tox/py27/lib/python2.7/site-packages/pymysql/cursors.py\n  Comments left for invalid file .tox/py27/lib/python2.7/site-packages/sqlalchemy/util/langhelpers.py\n  Comments left for invalid file .tox/py36/lib/python3.6/site-packages/pymysql/cursors.py\n  Comments left for invalid file .tox/py27/lib/python2.7/site-packages/sqlalchemy/orm/mapper.py\n  Comments left for invalid file .tox/lower-constraints/lib/python3.6/site-packages/sqlalchemy/orm/session.py\n  Comments left for invalid file .tox/lower-constraints/lib/python3.6/site-packages/pymysql/cursors.py\n  Comments left for invalid file .tox/py37/lib/python3.7/site-packages/sqlalchemy/orm/mapper.py\n  Comments left for invalid file .tox/py37/lib/python3.7/site-packages/pymysql/cursors.py","accounts_in_message":[],"_revision_number":5},{"id":"f803ca8c324e25c856d03212533697de433ff75f","author":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},"date":"2020-04-04 23:57:56.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"dea04892ef98e1f2f82f2a5065e72abdf7d1e40c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-05 01:31:42.000000000","message":"Patch Set 6: Verified-1\n\n(12 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- tempest-full https://zuul.opendev.org/t/openstack/build/380750014e83438f84976ad9518c748e : FAILURE in 30m 15s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/0c2b8dcab4e3478fab02b996a5a2ecef : SUCCESS in 1h 32m 12s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/55f9d9a00df44992a25fce02df6218c3 : SUCCESS in 4m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/af482166257847759931a9f23646feb0 : SUCCESS in 4m 12s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/18a7524bb0384c399388203819de5380 : SUCCESS in 5m 09s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/37996c1380be4fc699ed2e7761e07d23 : SUCCESS in 4m 32s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/dd72959c720c4f03a5b2ae367531cae2 : SUCCESS in 5m 35s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0752e1f87ab94890bae523127fc4b22d : SUCCESS in 4m 13s","accounts_in_message":[],"_revision_number":6},{"id":"bde99574db53b1ad1f8826ad250c10d4a0caf824","author":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},"date":"2020-04-09 23:37:38.000000000","message":"Abandoned\n\nNo longer needed.","accounts_in_message":[],"_revision_number":6}],"current_revision_number":6,"current_revision":"f1c0e3f788f0fdc08fddf36795a7d52b65db9c8e","revisions":{"832b41092ebb1a97cb8d578fed5e7f5cacf566e0":{"kind":"REWORK","_number":1,"created":"2020-03-25 04:49:57.000000000","uploader":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},"ref":"refs/changes/33/714833/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.db","ref":"refs/changes/33/714833/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.db refs/changes/33/714833/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.db refs/changes/33/714833/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.db refs/changes/33/714833/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.db refs/changes/33/714833/1"}}},"commit":{"parents":[{"commit":"d0b6a78cd55814b0a7d9cfc01aac319badc9df85","subject":"Update TOX/UPPER_CONSTRAINTS_FILE for stable/train","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.db/commit/d0b6a78cd55814b0a7d9cfc01aac319badc9df85"}]}],"author":{"name":"melanie witt","email":"melwittt@gmail.com","date":"2020-03-25 01:40:20.000000000","tz":0},"committer":{"name":"melanie witt","email":"melwittt@gmail.com","date":"2020-03-25 04:49:57.000000000","tz":0},"subject":"DNM: investigating bug 1844929","message":"DNM: investigating bug 1844929\n\nChange-Id: Ie5d7b2b5f60efe9612cef396ed58f9d0378b9cf7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.db/commit/832b41092ebb1a97cb8d578fed5e7f5cacf566e0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.db/commit/832b41092ebb1a97cb8d578fed5e7f5cacf566e0"}]},"branch":"refs/heads/stable/train"},"7c6049a176022d283e32f2d66639830326db5083":{"kind":"REWORK","_number":2,"created":"2020-03-26 21:34:12.000000000","uploader":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},"ref":"refs/changes/33/714833/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.db","ref":"refs/changes/33/714833/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.db refs/changes/33/714833/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.db refs/changes/33/714833/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.db refs/changes/33/714833/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.db refs/changes/33/714833/2"}}},"commit":{"parents":[{"commit":"d0b6a78cd55814b0a7d9cfc01aac319badc9df85","subject":"Update TOX/UPPER_CONSTRAINTS_FILE for stable/train","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.db/commit/d0b6a78cd55814b0a7d9cfc01aac319badc9df85"}]}],"author":{"name":"melanie witt","email":"melwittt@gmail.com","date":"2020-03-25 01:40:20.000000000","tz":0},"committer":{"name":"melanie witt","email":"melwittt@gmail.com","date":"2020-03-26 21:34:12.000000000","tz":0},"subject":"DNM: investigating bug 1844929","message":"DNM: investigating bug 1844929\n\nChange-Id: Ie5d7b2b5f60efe9612cef396ed58f9d0378b9cf7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.db/commit/7c6049a176022d283e32f2d66639830326db5083"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.db/commit/7c6049a176022d283e32f2d66639830326db5083"}]},"branch":"refs/heads/stable/train"},"644e1ad1d06cae63659dc7802ffabcc14bb9adf8":{"kind":"REWORK","_number":3,"created":"2020-03-27 16:13:30.000000000","uploader":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},"ref":"refs/changes/33/714833/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.db","ref":"refs/changes/33/714833/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.db refs/changes/33/714833/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.db refs/changes/33/714833/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.db refs/changes/33/714833/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.db refs/changes/33/714833/3"}}},"commit":{"parents":[{"commit":"d0b6a78cd55814b0a7d9cfc01aac319badc9df85","subject":"Update TOX/UPPER_CONSTRAINTS_FILE for stable/train","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.db/commit/d0b6a78cd55814b0a7d9cfc01aac319badc9df85"}]}],"author":{"name":"melanie witt","email":"melwittt@gmail.com","date":"2020-03-25 01:40:20.000000000","tz":0},"committer":{"name":"melanie witt","email":"melwittt@gmail.com","date":"2020-03-27 16:13:30.000000000","tz":0},"subject":"DNM: investigating bug 1844929","message":"DNM: investigating bug 1844929\n\nChange-Id: Ie5d7b2b5f60efe9612cef396ed58f9d0378b9cf7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.db/commit/644e1ad1d06cae63659dc7802ffabcc14bb9adf8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.db/commit/644e1ad1d06cae63659dc7802ffabcc14bb9adf8"}]},"branch":"refs/heads/stable/train"},"97c1001d470654b5151208cba6a470621e75aee4":{"kind":"REWORK","_number":4,"created":"2020-03-31 20:19:18.000000000","uploader":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},"ref":"refs/changes/33/714833/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.db","ref":"refs/changes/33/714833/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.db refs/changes/33/714833/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.db refs/changes/33/714833/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.db refs/changes/33/714833/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.db refs/changes/33/714833/4"}}},"commit":{"parents":[{"commit":"d0b6a78cd55814b0a7d9cfc01aac319badc9df85","subject":"Update TOX/UPPER_CONSTRAINTS_FILE for stable/train","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.db/commit/d0b6a78cd55814b0a7d9cfc01aac319badc9df85"}]}],"author":{"name":"melanie witt","email":"melwittt@gmail.com","date":"2020-03-25 01:40:20.000000000","tz":0},"committer":{"name":"melanie witt","email":"melwittt@gmail.com","date":"2020-03-31 20:19:18.000000000","tz":0},"subject":"DNM: investigating bug 1844929","message":"DNM: investigating bug 1844929\n\nChange-Id: Ie5d7b2b5f60efe9612cef396ed58f9d0378b9cf7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.db/commit/97c1001d470654b5151208cba6a470621e75aee4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.db/commit/97c1001d470654b5151208cba6a470621e75aee4"}]},"branch":"refs/heads/stable/train"},"86630a7133fbee21631b0e9b3633ba9f3dfc7860":{"kind":"REWORK","_number":5,"created":"2020-04-04 00:56:18.000000000","uploader":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},"ref":"refs/changes/33/714833/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.db","ref":"refs/changes/33/714833/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.db refs/changes/33/714833/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.db refs/changes/33/714833/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.db refs/changes/33/714833/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.db refs/changes/33/714833/5"}}},"commit":{"parents":[{"commit":"d0b6a78cd55814b0a7d9cfc01aac319badc9df85","subject":"Update TOX/UPPER_CONSTRAINTS_FILE for stable/train","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.db/commit/d0b6a78cd55814b0a7d9cfc01aac319badc9df85"}]}],"author":{"name":"melanie witt","email":"melwittt@gmail.com","date":"2020-03-25 01:40:20.000000000","tz":0},"committer":{"name":"melanie witt","email":"melwittt@gmail.com","date":"2020-04-04 00:56:18.000000000","tz":0},"subject":"DNM: investigating bug 1844929","message":"DNM: investigating bug 1844929\n\nChange-Id: Ie5d7b2b5f60efe9612cef396ed58f9d0378b9cf7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.db/commit/86630a7133fbee21631b0e9b3633ba9f3dfc7860"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.db/commit/86630a7133fbee21631b0e9b3633ba9f3dfc7860"}]},"branch":"refs/heads/stable/train"},"f1c0e3f788f0fdc08fddf36795a7d52b65db9c8e":{"kind":"REWORK","_number":6,"created":"2020-04-04 23:57:56.000000000","uploader":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},"ref":"refs/changes/33/714833/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.db","ref":"refs/changes/33/714833/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.db refs/changes/33/714833/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.db refs/changes/33/714833/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.db refs/changes/33/714833/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.db refs/changes/33/714833/6"}}},"commit":{"parents":[{"commit":"d0b6a78cd55814b0a7d9cfc01aac319badc9df85","subject":"Update TOX/UPPER_CONSTRAINTS_FILE for stable/train","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.db/commit/d0b6a78cd55814b0a7d9cfc01aac319badc9df85"}]}],"author":{"name":"melanie witt","email":"melwittt@gmail.com","date":"2020-03-25 01:40:20.000000000","tz":0},"committer":{"name":"melanie witt","email":"melwittt@gmail.com","date":"2020-04-04 23:57:56.000000000","tz":0},"subject":"DNM: investigating bug 1844929","message":"DNM: investigating bug 1844929\n\nChange-Id: Ie5d7b2b5f60efe9612cef396ed58f9d0378b9cf7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.db/commit/f1c0e3f788f0fdc08fddf36795a7d52b65db9c8e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.db/commit/f1c0e3f788f0fdc08fddf36795a7d52b65db9c8e"}]},"branch":"refs/heads/stable/train"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
