)]}'
{"id":"openstack%2Foslo.db~715504","triplet_id":"openstack%2Foslo.db~master~I56d62f67914f07e7aef74f857c4b42e3bfd3416e","project":"openstack/oslo.db","branch":"master","topic":"update-hacking","hashtags":[],"change_id":"I56d62f67914f07e7aef74f857c4b42e3bfd3416e","subject":"Update hacking for Python3","status":"MERGED","created":"2020-03-27 18:12:59.000000000","updated":"2020-03-30 15:34:09.000000000","submitted":"2020-03-30 15:30:26.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"715504-1585582226038-16c8230b","meta_rev_id":"88549270e9fd80c748026978d78fab5ae51c407f","_number":715504,"virtual_id_number":715504,"owner":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-03-30 15:30:25.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":2,"date":"2020-03-30 12:04:39.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":2,"date":"2020-03-30 12:31:43.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":1,"date":"2020-03-30 12:31:43.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-03-27 21:23:14.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2020-03-30 12:04:39.000000000","updated_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"reviewer":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"state":"REVIEWER"},{"updated":"2020-03-30 12:31:43.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2020-03-30 15:30:25.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"b4f5ecef907fe60b0e1363ac8510a1c8f50ab85f","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2020-03-27 18:12:59.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"7c1de4f93114078e62c07341f593d0d669f69d27","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-27 19:30:03.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/117ab8bc9c5c4454acad0ac7e392771e : SUCCESS in 4m 06s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/28f17a96f248439db10014d43d7750cf : SUCCESS in 1h 15m 16s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/553584c972a747d5afdddcbad8895aa9 : SUCCESS in 5m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/78b1178b14994810a3f5bed132e39f6a : FAILURE in 4m 51s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6f759116cd27401d8fa0ec19899f7265 : SUCCESS in 6m 29s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/59a9cb0a413148c38bab309ab9df7866 : SUCCESS in 7m 38s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c6a1691896534e639d91ff8ead9f5658 : SUCCESS in 7m 54s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/70c4720021b941a6961d2b116ce6221a : SUCCESS in 5m 49s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8735ba5d75964c6dade62c806364290e : SUCCESS in 5m 40s","accounts_in_message":[],"_revision_number":1},{"id":"29efcd17de6666b012c7ab43abaff93d6f4dcdbd","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2020-03-27 19:35:29.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"1f93a72959d7faf7636b02b5c1e1371d428d3431","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-27 21:22:51.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/393b03529cc44afcb0f41e8b51e0ac7a : SUCCESS in 3m 11s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/69a9ddac24bc43b697408304b6dfa77e : SUCCESS in 1h 41m 13s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/bad7506a8f86437c845a24bee38cc98b : SUCCESS in 5m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/93f35412825d499b99f941c97285b809 : SUCCESS in 3m 44s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/530711964c0644afbeb0aae3d8406a68 : SUCCESS in 5m 11s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/4c6d065735ae475bbd99f52a375c8fae : SUCCESS in 6m 30s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/eb43d48026654ea2b409521729c3254d : SUCCESS in 6m 18s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/88abe1dd8651429bb177c2838b9eab03 : SUCCESS in 4m 46s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0be2edfd9b034f82a8c8883e22ea7234 : SUCCESS in 5m 45s","accounts_in_message":[],"_revision_number":2},{"id":"f0812838de958f87b32d503f7e07f9b596a60267","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-03-27 21:23:14.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"2ffe47fa7ea3cf1579fd64c87575639107f320ae","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2020-03-30 10:29:01.000000000","message":"Patch Set 3: Published edit on patch set 2.","accounts_in_message":[],"_revision_number":3},{"id":"a18a189a3c6e7a7399ae35c4cd910c8b4b910aa5","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-03-30 11:46:23.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"883df1bf2789f31b75904264228069965f6608bd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-30 11:57:16.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/ff4063503b5548b89bc6d7d9add07084 : SUCCESS in 3m 25s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/468c988e44754a34900bbc7efdecef79 : SUCCESS in 1h 20m 57s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/b790497e8dd44591ac07d235e1736253 : SUCCESS in 7m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4b995f425f8348b09e5731509291e235 : SUCCESS in 4m 44s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8b92ccb07bdc4450b027af13023e7cc0 : SUCCESS in 5m 12s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/3693ab9447f84486a0a9ddd8ca5e8e08 : SUCCESS in 5m 25s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6d5796d9c28d47619ef92621c1069b66 : SUCCESS in 5m 51s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f5f2f8b5c41b469e9f54d936152b1699 : SUCCESS in 4m 56s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/de8e26ad106c4e2aa81ca1b1b4491541 : SUCCESS in 6m 33s","accounts_in_message":[],"_revision_number":3},{"id":"bbf208caf52f0d00ae7cb98ff20ba16eb338d2d8","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2020-03-30 11:57:45.000000000","message":"Patch Set 4: Published edit on patch set 3.","accounts_in_message":[],"_revision_number":4},{"id":"8b0be15cc842bdc950ad3435e9f162707a9dbb51","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2020-03-30 12:04:39.000000000","message":"Patch Set 4: Code-Review+2","accounts_in_message":[],"_revision_number":4},{"id":"49dc51637525efa747c82c1940797416b10c65fe","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-03-30 12:31:43.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"4ad777905a6cb948557661765546c8dfb915fbee","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-30 13:32:06.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/f16976cc97194312b27e009d5d84cfad : SUCCESS in 4m 28s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/1c69b5c5b4d14447a4898264bd925751 : SUCCESS in 1h 23m 05s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/230f25e6f60d4bfaa59f7d91d26d5a42 : SUCCESS in 5m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/69ec210c05e547a79d77f282eb2611b2 : SUCCESS in 4m 21s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/15e0cb56f26c468db51d4068b03dd762 : SUCCESS in 4m 30s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/f5b77143757c4068b39cab74cc45797d : SUCCESS in 6m 22s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/06436aab733d42d489ec1f598ee58ff8 : SUCCESS in 5m 15s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8eb23c9627d249e8b5eea02cfc033217 : SUCCESS in 4m 30s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9eab45c649634873b6e4ff9f79882a4a : SUCCESS in 7m 10s","accounts_in_message":[],"_revision_number":4},{"id":"6eb3939cb04cb8b5762eec957a417f86029f8e35","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-30 13:32:18.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"3f5ce20d94f8336ba9e44b54797bcefd26e7d5b8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-30 15:30:25.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/ac82c5ba5bb6421385f083921d1cc9ce : SUCCESS in 3m 18s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/200f5a9986a3467087d262c81089319d : SUCCESS in 1h 54m 25s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/b3bb03baec6c4061a127d610d78cd3bb : SUCCESS in 6m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/93276c869e15463986dc8ac3a6f5f472 : SUCCESS in 3m 51s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2c99e2173ac043e38b1ec138043e137c : SUCCESS in 4m 57s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/c4e8e3d94ff2435bac9b9c4ffdfca915 : SUCCESS in 5m 57s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5286d4003c894a2c9a1e6252a0fefb4f : SUCCESS in 7m 16s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/caa37c37d0e84e7a98a04b6dd4181a9e : SUCCESS in 4m 46s","accounts_in_message":[],"_revision_number":4},{"id":"fda9ce47c9b33e28d056eddcc7568b6650107fdc","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-30 15:30:26.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":4},{"id":"e0cfd82664c72685fe7af00859cc9da7c3956e0e","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-30 15:34:09.000000000","message":"Patch Set 4:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/774df9df6c00489bab95eb28bfa88d36 : SUCCESS in 2m 59s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/80b099b1337740e7ad9ed548854905a5 : SUCCESS in 3m 06s","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"4e6fe7cf7d9cce02343ecf0d418ca388ecb2d84d","revisions":{"6b45b57326f8686fe43a6e19f8d593d094896d27":{"kind":"REWORK","_number":1,"created":"2020-03-27 18:12:59.000000000","uploader":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"ref":"refs/changes/04/715504/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.db","ref":"refs/changes/04/715504/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.db refs/changes/04/715504/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.db refs/changes/04/715504/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.db refs/changes/04/715504/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.db refs/changes/04/715504/1"}}},"commit":{"parents":[{"commit":"bf5b45ebba24b4c8737c3197f2cd65f84fd2448c","subject":"Remove \u0027handle_connect_error\u0027","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.db/commit/bf5b45ebba24b4c8737c3197f2cd65f84fd2448c"}]}],"author":{"name":"Andreas Jaeger","email":"aj@suse.com","date":"2020-03-27 18:12:05.000000000","tz":60},"committer":{"name":"Andreas Jaeger","email":"aj@suse.com","date":"2020-03-27 18:12:05.000000000","tz":60},"subject":"Update hacking for Python3","message":"Update hacking for Python3\n\nThe repo is Python 3 now, so update hacking to version 2.0 which\nsupports Python 3.\n\nChange-Id: I56d62f67914f07e7aef74f857c4b42e3bfd3416e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.db/commit/6b45b57326f8686fe43a6e19f8d593d094896d27"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.db/commit/6b45b57326f8686fe43a6e19f8d593d094896d27"}]},"branch":"refs/heads/master"},"6dc96fb9f8a39694ead7ece26dba22b05554e4ae":{"kind":"REWORK","_number":2,"created":"2020-03-27 19:35:29.000000000","uploader":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"ref":"refs/changes/04/715504/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.db","ref":"refs/changes/04/715504/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.db refs/changes/04/715504/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.db refs/changes/04/715504/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.db refs/changes/04/715504/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.db refs/changes/04/715504/2"}}},"commit":{"parents":[{"commit":"bf5b45ebba24b4c8737c3197f2cd65f84fd2448c","subject":"Remove \u0027handle_connect_error\u0027","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.db/commit/bf5b45ebba24b4c8737c3197f2cd65f84fd2448c"}]}],"author":{"name":"Andreas Jaeger","email":"aj@suse.com","date":"2020-03-27 18:12:05.000000000","tz":60},"committer":{"name":"Andreas Jaeger","email":"aj@suse.com","date":"2020-03-27 19:35:24.000000000","tz":60},"subject":"Update hacking for Python3","message":"Update hacking for Python3\n\nThe repo is Python 3 now, so update hacking to version 2.0 which\nsupports Python 3.\n\nblacklist:\n W504 line break after binary operator\n\nFix:\nE305 expected 2 blank lines after class or function definition, found 1\n\nChange-Id: I56d62f67914f07e7aef74f857c4b42e3bfd3416e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.db/commit/6dc96fb9f8a39694ead7ece26dba22b05554e4ae"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.db/commit/6dc96fb9f8a39694ead7ece26dba22b05554e4ae"}]},"branch":"refs/heads/master"},"fa351c3ccb8b8b4780e5aefee1dac3ca0aa7856d":{"kind":"REWORK","_number":3,"created":"2020-03-30 10:29:01.000000000","uploader":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"ref":"refs/changes/04/715504/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.db","ref":"refs/changes/04/715504/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.db refs/changes/04/715504/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.db refs/changes/04/715504/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.db refs/changes/04/715504/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.db refs/changes/04/715504/3"}}},"commit":{"parents":[{"commit":"bf5b45ebba24b4c8737c3197f2cd65f84fd2448c","subject":"Remove \u0027handle_connect_error\u0027","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.db/commit/bf5b45ebba24b4c8737c3197f2cd65f84fd2448c"}]}],"author":{"name":"Andreas Jaeger","email":"aj@suse.com","date":"2020-03-27 18:12:05.000000000","tz":60},"committer":{"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","date":"2020-03-30 10:28:57.000000000","tz":0},"subject":"Update hacking for Python3","message":"Update hacking for Python3\n\nThe repo is Python 3 now, so update hacking to version 3.0 which\nsupports Python 3.\n\nblacklist:\n W504 line break after binary operator\n\nFix:\nE305 expected 2 blank lines after class or function definition, found 1\n\nChange-Id: I56d62f67914f07e7aef74f857c4b42e3bfd3416e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.db/commit/fa351c3ccb8b8b4780e5aefee1dac3ca0aa7856d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.db/commit/fa351c3ccb8b8b4780e5aefee1dac3ca0aa7856d"}]},"branch":"refs/heads/master"},"4e6fe7cf7d9cce02343ecf0d418ca388ecb2d84d":{"kind":"REWORK","_number":4,"created":"2020-03-30 11:57:45.000000000","uploader":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"ref":"refs/changes/04/715504/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.db","ref":"refs/changes/04/715504/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.db refs/changes/04/715504/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.db refs/changes/04/715504/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.db refs/changes/04/715504/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.db refs/changes/04/715504/4"}}},"commit":{"parents":[{"commit":"bf5b45ebba24b4c8737c3197f2cd65f84fd2448c","subject":"Remove \u0027handle_connect_error\u0027","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.db/commit/bf5b45ebba24b4c8737c3197f2cd65f84fd2448c"}]}],"author":{"name":"Andreas Jaeger","email":"aj@suse.com","date":"2020-03-27 18:12:05.000000000","tz":60},"committer":{"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","date":"2020-03-30 11:57:43.000000000","tz":0},"subject":"Update hacking for Python3","message":"Update hacking for Python3\n\nThe repo is Python 3 now, so update hacking to version 3.0 which\nsupports Python 3.\n\nblacklist:\n W504 line break after binary operator\n\nFix:\nE305 expected 2 blank lines after class or function definition, found 1\n\nChange-Id: I56d62f67914f07e7aef74f857c4b42e3bfd3416e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.db/commit/4e6fe7cf7d9cce02343ecf0d418ca388ecb2d84d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.db/commit/4e6fe7cf7d9cce02343ecf0d418ca388ecb2d84d"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
