)]}'
{"id":"openstack%2Foslo.db~717771","triplet_id":"openstack%2Foslo.db~master~I2e4b9082194df6d916000c4f8f179f05d606e1a3","project":"openstack/oslo.db","branch":"master","topic":"modernize_count_test","hashtags":[],"change_id":"I2e4b9082194df6d916000c4f8f179f05d606e1a3","subject":"Modernize use of table.count() with func.count()","status":"MERGED","created":"2020-04-06 14:42:05.000000000","updated":"2020-04-21 13:16:02.000000000","submitted":"2020-04-21 11:40:38.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"717771-1587469238609-795c03e7","meta_rev_id":"ba4cd1c4f0c95e89dc5b5cc246333a03305fde65","_number":717771,"virtual_id_number":717771,"owner":{"_account_id":11816,"name":"mike bayer","display_name":"Mike Bayer","email":"mike_mp@zzzcomputing.com","username":"zzzeek","status":"Red Hat"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-04-21 11:40:38.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":2,"date":"2020-04-21 10:11:40.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":2,"date":"2020-04-16 11:05:11.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":1,"date":"2020-04-21 10:11:40.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-04-16 11:05:11.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2020-04-21 10:11:40.000000000","updated_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"reviewer":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"state":"REVIEWER"},{"updated":"2020-04-21 11:40:38.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"25b392a69ee713c2779220e6c25191522cdad611","author":{"_account_id":11816,"name":"mike bayer","display_name":"Mike Bayer","email":"mike_mp@zzzcomputing.com","username":"zzzeek","status":"Red Hat"},"date":"2020-04-06 14:42:05.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"1e92384ecf94ae45d337dc224e4389862e7ca1ca","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-06 16:24:13.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/9855a6629a384e3ea0f78ab0938e4598 : SUCCESS in 1h 34m 38s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/103f1da4bd0341b5b9bfc7acfda852f1 : SUCCESS in 7m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/78df53d015524fbdb2fa8a6a2095eb59 : SUCCESS in 5m 25s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/06e4c20141d743fc965711b8cf6990d3 : SUCCESS in 11m 02s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/4c872365c7e045ffac93badf58bef71f : SUCCESS in 7m 53s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2c30043e54c1465d8e42d2ec70c5676c : SUCCESS in 6m 42s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6b33510a12934ab0a3eb3d02dcfbed8f : SUCCESS in 7m 42s\n\nWarning:\n  Comments left for invalid file oslo_db/sqlalchemy/enginefacade.py","accounts_in_message":[],"_revision_number":1},{"id":"f9b9dd34bbce42ba4e65e6a919f8444cd91ecef7","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-04-16 11:05:11.000000000","message":"Patch Set 1: Code-Review+2\n\nIntroduced changes [1][2][3] LGTM.\n\n[1] https://docs.sqlalchemy.org/en/13/core/functions.html#sqlalchemy.sql.functions.FunctionElement.select \n[2] https://docs.sqlalchemy.org/en/13/core/selectable.html?highlight\u003dfromclause%20count#sqlalchemy.sql.expression.FromClause.count\n[3] https://docs.sqlalchemy.org/en/13/core/selectable.html?highlight\u003dfromclause%20count#sqlalchemy.sql.expression.Select.select_from","accounts_in_message":[],"_revision_number":1},{"id":"464f68178e0ac812277e16f4a86d24fd734779d0","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2020-04-21 10:11:40.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\nTa","accounts_in_message":[],"_revision_number":1},{"id":"6de19d40b2eab0b29c83b5b891db7b75e24ef2a3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-21 10:11:52.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"f79e507c9cd4b9278e4502c8101aff66f5fec4c8","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-21 11:40:38.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"1dff75469054bc022fe23c64b0b55fb254e89023","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-21 11:40:38.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/e6e0163f3c404b1c8c06a8a481a23523 : SUCCESS in 1h 27m 31s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/fffaded1ae8a4b768362d6229450e8e4 : SUCCESS in 6m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a8ba6350171e4608b2a4f4d339920235 : SUCCESS in 4m 04s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8e3f842b90bb4580b8408c09abf671a7 : SUCCESS in 4m 36s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2f7e129976cf4742be14b465e2b8ae24 : SUCCESS in 7m 29s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0720ecd7e22a4643acda63587277e009 : SUCCESS in 4m 05s","accounts_in_message":[],"_revision_number":1},{"id":"4f9e0c6e1d9c2b153e7068ef137286197dad906d","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-21 11:42:45.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ddb1bdc57e6b4857b3f538ce19aeacb1 : SUCCESS in 1m 47s","accounts_in_message":[],"_revision_number":1},{"id":"4999bf24957e5e6b4e4e1a00f790b80647fed3fc","author":{"_account_id":11816,"name":"mike bayer","display_name":"Mike Bayer","email":"mike_mp@zzzcomputing.com","username":"zzzeek","status":"Red Hat"},"date":"2020-04-21 13:16:02.000000000","message":"Patch Set 1:\n\nthanks all!","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"85573e07823556882117c21f42a098de7d470f7b","revisions":{"85573e07823556882117c21f42a098de7d470f7b":{"kind":"REWORK","_number":1,"created":"2020-04-06 14:42:05.000000000","uploader":{"_account_id":11816,"name":"mike bayer","display_name":"Mike Bayer","email":"mike_mp@zzzcomputing.com","username":"zzzeek","status":"Red Hat"},"ref":"refs/changes/71/717771/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.db","ref":"refs/changes/71/717771/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.db refs/changes/71/717771/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.db refs/changes/71/717771/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.db refs/changes/71/717771/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.db refs/changes/71/717771/1"}}},"commit":{"parents":[{"commit":"4e6fe7cf7d9cce02343ecf0d418ca388ecb2d84d","subject":"Update hacking for Python3","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.db/commit/4e6fe7cf7d9cce02343ecf0d418ca388ecb2d84d"}]}],"author":{"name":"Mike Bayer","email":"mike_mp@zzzcomputing.com","date":"2020-04-06 14:38:29.000000000","tz":-240},"committer":{"name":"Mike Bayer","email":"mike_mp@zzzcomputing.com","date":"2020-04-06 14:38:29.000000000","tz":-240},"subject":"Modernize use of table.count() with func.count()","message":"Modernize use of table.count() with func.count()\n\nThe table.count() function in SQLAlchemy has been deprecated\nsince version 1.1 [1] and is being removed in version 1.4 [2] [3].\nIn order to maintain forwards compatibility as well as\nsuit SQLAlchemy\u0027s CI that includes oslo.db, replace this\none usage with modern use of func.count() as well as\nexplicit connection execution (engine.execute() is also\ngoing away more long term).\n\n[1] https://docs.sqlalchemy.org/en/13/core/selectable.html?highlight\u003dfromclause%20count#sqlalchemy.sql.expression.FromClause.count\n[2] https://gerrit.sqlalchemy.org/c/sqlalchemy/sqlalchemy/+/1833\n[3] https://github.com/sqlalchemy/sqlalchemy/issues/4643\n\nChange-Id: I2e4b9082194df6d916000c4f8f179f05d606e1a3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.db/commit/85573e07823556882117c21f42a098de7d470f7b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.db/commit/85573e07823556882117c21f42a098de7d470f7b"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
