)]}'
{"id":"openstack%2Foslo.db~740236","triplet_id":"openstack%2Foslo.db~master~I9525eda24465bdedb9ae917350263fa91e9b780a","project":"openstack/oslo.db","branch":"master","topic":"drop-os-testr","hashtags":[],"change_id":"I9525eda24465bdedb9ae917350263fa91e9b780a","subject":"requirements: Drop os-testr","status":"MERGED","created":"2020-07-09 10:38:40.000000000","updated":"2020-07-10 17:10:35.000000000","submitted":"2020-07-10 17:07:01.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"740236-1594400821497-55706493","meta_rev_id":"9ee2688b41c70cd743a4fd93957f361a763acb61","_number":740236,"virtual_id_number":740236,"owner":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":27621,"name":"Vishakha Agarwal","email":"agarwalvishakha18@gmail.com","username":"Vishakha"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-07-10 17:07:01.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2020-07-10 15:38:22.000000000","_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":1,"date":"2020-07-09 16:22:51.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":27621,"name":"Vishakha Agarwal","email":"agarwalvishakha18@gmail.com","username":"Vishakha"},{"value":2,"date":"2020-07-10 08:22:35.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":1,"date":"2020-07-10 15:38:22.000000000","_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":27621,"name":"Vishakha Agarwal","email":"agarwalvishakha18@gmail.com","username":"Vishakha"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":27621,"name":"Vishakha Agarwal","email":"agarwalvishakha18@gmail.com","username":"Vishakha"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":27621,"name":"Vishakha Agarwal","email":"agarwalvishakha18@gmail.com","username":"Vishakha"},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-07-09 16:22:51.000000000","updated_by":{"_account_id":27621,"name":"Vishakha Agarwal","email":"agarwalvishakha18@gmail.com","username":"Vishakha"},"reviewer":{"_account_id":27621,"name":"Vishakha Agarwal","email":"agarwalvishakha18@gmail.com","username":"Vishakha"},"state":"REVIEWER"},{"updated":"2020-07-10 08:22:35.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2020-07-10 15:38:22.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2020-07-10 17:07:01.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"8223aba14049907d4384b3bbb6af160c725d7a8f","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2020-07-09 10:38:40.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"ec07150ea46c1dbc1d097585f29e4476a36218e8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-09 12:02:33.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/29a38a7767ae471eaafa698a17b66306 : SUCCESS in 5m 56s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/decfb3a8e8d34dc58995357b493b7c0f : SUCCESS in 1h 19m 14s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/e244065f6fb74f3fa3592df13cdb234e : SUCCESS in 10m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/26edf5889ca044d9a20f1e6faf919056 : SUCCESS in 7m 12s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0dffc4ba7eee40cebafdb9ffa66bc435 : SUCCESS in 9m 22s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/092dc8223b254aaf8b22ff716dcf5a39 : SUCCESS in 9m 57s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/803b7f6da0d04609a2bb4b0f169b3be1 : SUCCESS in 8m 17s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e98e331f3f4d4bd1b98a601137ded524 : SUCCESS in 6m 46s","accounts_in_message":[],"_revision_number":1},{"id":"20b2f76a343b47b4fcb5799c02633d02730a6b7d","author":{"_account_id":27621,"name":"Vishakha Agarwal","email":"agarwalvishakha18@gmail.com","username":"Vishakha"},"date":"2020-07-09 16:22:51.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"64344f6d48c2c8a40dbc29522d74ff964ab30ba8","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-07-10 08:22:35.000000000","message":"Patch Set 1: Code-Review+2\n\nNo longer used, LGTM","accounts_in_message":[],"_revision_number":1},{"id":"4bd4338d3642aa5a70b923c675e340637f4299d5","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-07-10 15:38:22.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"63e916d10e5efe6d2d6a4d52bb032e12d5441fc7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-10 15:38:33.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"9c089a54bea4c5eba7ca2665c813c11bff81763c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-10 17:07:01.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"af661fbdc007fadc1caa01bcaab8dd19ed92a354","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-10 17:07:01.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/072b4cc69ec34e4e9e35c1e778a9cfb2 : SUCCESS in 3m 37s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/e7de96c530ba4eb3b183a9c8c887ab04 : SUCCESS in 1h 26m 16s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/875afbe60593495f834fe6415f154d20 : SUCCESS in 6m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/542d776ad33d423e85a1399e9675f84b : SUCCESS in 6m 57s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/64a66f5061a340a29e2d29461ba9b7f7 : SUCCESS in 6m 07s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8705bf3086044195b648d1eb35f8b83c : SUCCESS in 6m 00s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/994b70f049f34276a08b8b76d6466b43 : SUCCESS in 5m 41s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/543a57beea254de59c3914f5b71ff92f : SUCCESS in 4m 50s","accounts_in_message":[],"_revision_number":1},{"id":"35d28ea65e98b98070b53f361092af5d333ffc2e","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-10 17:10:35.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3f53e8bd65af4eecb374f85e4949683c : SUCCESS in 1m 28s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9a58c6ba15ec4018ac15d100251fbae6 : SUCCESS in 3m 12s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"a9e699227d843b011d4c780c216fe2fe00a549bc","revisions":{"a9e699227d843b011d4c780c216fe2fe00a549bc":{"kind":"REWORK","_number":1,"created":"2020-07-09 10:38:40.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/36/740236/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.db","ref":"refs/changes/36/740236/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.db refs/changes/36/740236/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.db refs/changes/36/740236/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.db refs/changes/36/740236/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.db refs/changes/36/740236/1"}}},"commit":{"parents":[{"commit":"f62ddb15244d5b5aa0c54f2d6b0ae30799835eb2","subject":"Merge \"Bump default tox env from py37 to py38\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.db/commit/f62ddb15244d5b5aa0c54f2d6b0ae30799835eb2"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2020-07-09 10:37:49.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2020-07-09 10:37:49.000000000","tz":60},"subject":"requirements: Drop os-testr","message":"requirements: Drop os-testr\n\nChange-Id: I9525eda24465bdedb9ae917350263fa91e9b780a\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.db/commit/a9e699227d843b011d4c780c216fe2fe00a549bc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.db/commit/a9e699227d843b011d4c780c216fe2fe00a549bc"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
