)]}'
{"id":"openstack%2Foslo.db~744998","triplet_id":"openstack%2Foslo.db~master~I69cbe4ed654e5fe249cb9396449ad6644f519c34","project":"openstack/oslo.db","branch":"master","topic":"mysql-0.10.0","hashtags":[],"change_id":"I69cbe4ed654e5fe249cb9396449ad6644f519c34","subject":"Handle PyMySQL exception updates","status":"ABANDONED","created":"2020-08-05 22:31:05.000000000","updated":"2020-08-06 16:57:03.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"1e2b4e5cac394ef5b5825e0510a419f995368a98","_number":744998,"virtual_id_number":744998,"owner":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2020-08-05 23:47:25.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-08-05 23:47:25.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"47058d54e55458d1aafe60f69e161529cc4602d2","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-08-05 22:31:05.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"7707c0a63a5d07259576268ff0498b993b18bbd3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-05 23:47:25.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/1006b43b5ca24ce2aa2236a02881d91d : SUCCESS in 1h 15m 18s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/e43c2ffba593491ba4e793900b6f09b1 : FAILURE in 5m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/26d9322ad5f04eddbea9e44d9c4cf00e : SUCCESS in 3m 39s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/683ef46aad1445bab6031ed6c99e1604 : FAILURE in 4m 32s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1fe67c166d274633af4036b0ac7701ac : FAILURE in 4m 52s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c413d064d6404f76b6cf8fd1089fa4ba : SUCCESS in 5m 03s","accounts_in_message":[],"_revision_number":1},{"id":"8496ee2beb2c5ad55d30bff3f59c9c93197da142","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-08-06 16:57:03.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"0423f2d3a676ef55ad2204096963dbb33a863927","revisions":{"0423f2d3a676ef55ad2204096963dbb33a863927":{"kind":"REWORK","_number":1,"created":"2020-08-05 22:31:05.000000000","uploader":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"ref":"refs/changes/98/744998/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.db","ref":"refs/changes/98/744998/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.db refs/changes/98/744998/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.db refs/changes/98/744998/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.db refs/changes/98/744998/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.db refs/changes/98/744998/1"}}},"commit":{"parents":[{"commit":"a0f44c1755c0e3c9143752f9b15fd222919a1ec7","subject":"Merge \"requirements: Drop os-testr\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.db/commit/a0f44c1755c0e3c9143752f9b15fd222919a1ec7"}]}],"author":{"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","date":"2020-08-05 22:28:57.000000000","tz":-300},"committer":{"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","date":"2020-08-05 22:28:57.000000000","tz":-300},"subject":"Handle PyMySQL exception updates","message":"Handle PyMySQL exception updates\n\nPyMySQL 0.10.0 updated the way custom errors are raised, resulting in\nwhat would previously have been wrapped as an oslo.db DBError now\nbubbling up as an OperationalError instead.\n\nhttps://github.com/PyMySQL/PyMySQL/commit/c3e5a63514c57d1f4c9d5e7bf4b7e10b0608b0e1\n\nThis adds filtering to catch the new errors and falling back to wrapping\nthem as DBErrors.\n\nChange-Id: I69cbe4ed654e5fe249cb9396449ad6644f519c34\nSigned-off-by: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.db/commit/0423f2d3a676ef55ad2204096963dbb33a863927"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.db/commit/0423f2d3a676ef55ad2204096963dbb33a863927"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
