)]}'
{"id":"openstack%2Foslo.db~747762","triplet_id":"openstack%2Foslo.db~master~Id2f0663b13ed0f81e91a8d44f73d8541015bf844","project":"openstack/oslo.db","branch":"master","topic":"modernize_update_match","hashtags":[],"change_id":"Id2f0663b13ed0f81e91a8d44f73d8541015bf844","subject":"Accommodate immutable URL api","status":"MERGED","created":"2020-08-24 16:01:29.000000000","updated":"2021-04-28 11:14:54.000000000","submitted":"2021-04-28 11:12:59.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":14,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"747762-modernize_update_match","meta_rev_id":"c8b32f33c3ee90c7013f13db080c36a225cb8064","_number":747762,"virtual_id_number":747762,"owner":{"_account_id":11816,"name":"mike bayer","display_name":"Mike Bayer","email":"mike_mp@zzzcomputing.com","username":"zzzeek","status":"Red Hat"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":11816,"name":"mike bayer","display_name":"Mike Bayer","email":"mike_mp@zzzcomputing.com","username":"zzzeek","status":"Red Hat"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-04-28 11:12:48.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},{"value":0,"_account_id":8846,"name":"Hao Wang","display_name":"Hao Wang","email":"sxmatch1986@gmail.com","username":"sxmatch"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},"all":[{"value":2,"date":"2021-01-14 22:23:05.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":11816,"name":"mike bayer","display_name":"Mike Bayer","email":"mike_mp@zzzcomputing.com","username":"zzzeek","status":"Red Hat"},{"value":2,"date":"2021-04-28 09:51:38.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2021-02-02 11:54:00.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},{"value":1,"date":"2021-03-26 00:57:05.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8846,"name":"Hao Wang","display_name":"Hao Wang","email":"sxmatch1986@gmail.com","username":"sxmatch"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"all":[{"value":0,"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":11816,"name":"mike bayer","display_name":"Mike Bayer","email":"mike_mp@zzzcomputing.com","username":"zzzeek","status":"Red Hat"},{"value":1,"date":"2021-04-28 09:51:38.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},{"value":0,"_account_id":8846,"name":"Hao Wang","display_name":"Hao Wang","email":"sxmatch1986@gmail.com","username":"sxmatch"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":11816,"name":"mike bayer","display_name":"Mike Bayer","email":"mike_mp@zzzcomputing.com","username":"zzzeek","status":"Red Hat"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},{"value":0,"_account_id":8846,"name":"Hao Wang","display_name":"Hao Wang","email":"sxmatch1986@gmail.com","username":"sxmatch"}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},{"_account_id":8846,"name":"Hao Wang","display_name":"Hao Wang","email":"sxmatch1986@gmail.com","username":"sxmatch"},{"_account_id":11816,"name":"mike bayer","display_name":"Mike Bayer","email":"mike_mp@zzzcomputing.com","username":"zzzeek","status":"Red Hat"},{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-09-04 15:14:17.000000000","updated_by":{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},"reviewer":{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},"state":"REVIEWER"},{"updated":"2020-09-04 18:57:24.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2020-10-14 09:49:53.000000000","updated_by":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},"reviewer":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},"state":"REVIEWER"},{"updated":"2020-10-14 16:20:06.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-10-15 14:54:24.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2020-12-11 14:26:14.000000000","updated_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"reviewer":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"state":"REVIEWER"},{"updated":"2021-03-26 00:57:05.000000000","updated_by":{"_account_id":8846,"name":"Hao Wang","display_name":"Hao Wang","email":"sxmatch1986@gmail.com","username":"sxmatch"},"reviewer":{"_account_id":8846,"name":"Hao Wang","display_name":"Hao Wang","email":"sxmatch1986@gmail.com","username":"sxmatch"},"state":"REVIEWER"}],"messages":[{"id":"f094785e6085edfce22fcdf0864056412e053048","author":{"_account_id":11816,"name":"mike bayer","display_name":"Mike Bayer","email":"mike_mp@zzzcomputing.com","username":"zzzeek","status":"Red Hat"},"date":"2020-08-24 16:01:29.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"5e7e7249589ed9b2deb260a8bad438b7cbc73f9b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-24 17:36:47.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/198e424796494500b386c3bd1ad69f79 : SUCCESS in 1h 29m 35s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/fdfa6fdfafcb4f50ab8ba70a8735d0f4 : SUCCESS in 7m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6fce4b94f58b46dabf0cf5581f20bb42 : SUCCESS in 5m 21s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/38593a266831459c9e9774d8b6c355db : SUCCESS in 14m 02s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5cbb3536afa0428986f3620efc523bff : SUCCESS in 8m 45s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9380b25487924346a9e8229a27ee8513 : SUCCESS in 5m 02s","accounts_in_message":[],"_revision_number":1},{"id":"114e879eb74a90d4c01ba9300596ca9c78b81a1e","author":{"_account_id":11816,"name":"mike bayer","display_name":"Mike Bayer","email":"mike_mp@zzzcomputing.com","username":"zzzeek","status":"Red Hat"},"date":"2020-08-24 21:04:19.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"31e929e8ea343d9d1bf85654fc5acb9e6232b83a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-24 22:30:24.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/7419412361774d549e47f68ae2b153bb : SUCCESS in 1h 19m 31s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d8622f0e87604e9ea746fb8a0c278b12 : SUCCESS in 9m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/711288517fa54a47814070ad6109a884 : SUCCESS in 4m 40s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/37a062e47418411699812807d86e9809 : SUCCESS in 6m 42s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/cfd9df5aa33a44b490da4d21221b3a6d : SUCCESS in 5m 26s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/41b84c5d067f48139e6c72ba6309124c : SUCCESS in 6m 26s","accounts_in_message":[],"_revision_number":2},{"id":"3c7f58c7005c078322f24a4ebe838e482febd94d","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-09-04 11:25:09.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"4fade5c73bd53828c8c62e9a2decbbe2699836aa","author":{"_account_id":11816,"name":"mike bayer","display_name":"Mike Bayer","email":"mike_mp@zzzcomputing.com","username":"zzzeek","status":"Red Hat"},"date":"2020-09-04 13:17:51.000000000","message":"Patch Set 2: Workflow+1\n\nthis API has merged.   however, I\u0027m not sure if I\u0027m going to get more pushback once I release the initial betas of SQLAlchemy.  probably not, nobody ever tests the betas anyway :).    this is safe to merge and I guess if the API changed the name again or something we\u0027d just update here again.","accounts_in_message":[],"_revision_number":2},{"id":"41d1aa2c23a6ce44a8007f185d3a7d548790d21e","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-09-04 15:13:54.000000000","message":"Patch Set 2: Backport-Candidate+1\n\n\u003e this API has merged.   however, I\u0027m not sure if I\u0027m going to get\n \u003e more pushback once I release the initial betas of SQLAlchemy. \n \u003e probably not, nobody ever tests the betas anyway :).    this is\n \u003e safe to merge and I guess if the API changed the name again or\n \u003e something we\u0027d just update here again.\n\nThanks Mike for your feedback.\n\nWe are currently in feature freeze but I think we can go ahead with these changes as they are not a new feature.\n\nI would appreciate some feedback from Ben concerning the FF before merge this patch. \n\nI only +1 for now to avoid to trigger the merge as you +W\u0027ed.\n\nLets add Ben to the review.","accounts_in_message":[],"_revision_number":2},{"id":"06749d417532b44bda76c3ec08291ca3673cd108","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-09-04 15:14:03.000000000","message":"Patch Set 2: -Backport-Candidate Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"0a7ec10dc59618bf0abd5644db3314a5c333b9a2","author":{"_account_id":11816,"name":"mike bayer","display_name":"Mike Bayer","email":"mike_mp@zzzcomputing.com","username":"zzzeek","status":"Red Hat"},"date":"2020-09-04 15:21:49.000000000","message":"Patch Set 2:\n\nthere\u0027s no rush for this yet, SQLAlchemy 1.4 doesn\u0027t have beta releases yet.   openstack won\u0027t be dealing with trying to get SQLAlchemy 1.4 working for a few months at least.","accounts_in_message":[],"_revision_number":2},{"id":"5ed9e428770dc367091302829ae281f4cbbab95e","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-09-04 15:30:28.000000000","message":"Patch Set 2:\n\n\u003e there\u0027s no rush for this yet, SQLAlchemy 1.4 doesn\u0027t have beta\n \u003e releases yet.   openstack won\u0027t be dealing with trying to get\n \u003e SQLAlchemy 1.4 working for a few months at least.\n\nAck thanks","accounts_in_message":[],"_revision_number":2},{"id":"faaaa8f19297ab09fe6a0273e22bff978608978d","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-09-04 16:47:35.000000000","message":"Patch Set 2: Backport-Candidate-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"82de0004a8bd1fbfce5920293124b24b502516de","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-09-04 16:47:42.000000000","message":"Patch Set 2: -Backport-Candidate Code-Review-1","accounts_in_message":[],"_revision_number":2},{"id":"5d2e6c5fe7b5ceee3bb96234ee70e7fb73576ff4","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-09-04 16:48:20.000000000","message":"Patch Set 2: Workflow-1\n\n-W since it looks like you accidentally did +W by mistake.","accounts_in_message":[],"_revision_number":2},{"id":"7ff6f95e51754e4c620362ee674b53b1194c8855","author":{"_account_id":11816,"name":"mike bayer","display_name":"Mike Bayer","email":"mike_mp@zzzcomputing.com","username":"zzzeek","status":"Red Hat"},"date":"2020-09-04 17:16:17.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"f52c546c2860a97beb8479afec513d8cb8244a65","author":{"_account_id":11816,"name":"mike bayer","display_name":"Mike Bayer","email":"mike_mp@zzzcomputing.com","username":"zzzeek","status":"Red Hat"},"date":"2020-09-04 17:16:25.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"236ed8ed95e2c15bc6d9f70b42c098e45f595d61","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-04 18:39:55.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/d695cf6842ea4bd3bd642a13b3a2891f : SUCCESS in 1h 21m 22s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/831b5c90b08349fc8fd8d71d9efeeec9 : SUCCESS in 5m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/609b45c2292f458c810c9afaa2e72619 : SUCCESS in 4m 53s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/240f753a44684b31906bc2db2ca4cfdb : SUCCESS in 5m 02s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/bbab4544ad744ace9258c46c1eb2f1c4 : SUCCESS in 5m 08s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c48cdb923520464495b394bed95f967a : SUCCESS in 5m 50s","accounts_in_message":[],"_revision_number":3},{"id":"03fa634b57617a4f12192db4eebb680f9857009a","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-09-04 18:57:24.000000000","message":"Patch Set 3: Code-Review+1\n\n(1 comment)\n\nLooks good to me. Wondering if there is still more testing to be done, or if we are ready to proceed with this once Wallaby changes open up.","accounts_in_message":[],"_revision_number":3},{"id":"1864801c20208aee03285e6f1ae0615ab88ab252","author":{"_account_id":11816,"name":"mike bayer","display_name":"Mike Bayer","email":"mike_mp@zzzcomputing.com","username":"zzzeek","status":"Red Hat"},"date":"2020-09-04 19:11:07.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"5334cc15783bf829d400cb32f5b3a445ec4aa8f8","author":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},"date":"2020-10-14 09:49:53.000000000","message":"Patch Set 3: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"66339b62819e6e2f24e1b2f6f8a3589f732c6228","author":{"_account_id":11816,"name":"mike bayer","display_name":"Mike Bayer","email":"mike_mp@zzzcomputing.com","username":"zzzeek","status":"Red Hat"},"date":"2020-10-14 13:11:53.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"891479541142938626c3ab67da7fd1f25cc72732","author":{"_account_id":11816,"name":"mike bayer","display_name":"Mike Bayer","email":"mike_mp@zzzcomputing.com","username":"zzzeek","status":"Red Hat"},"date":"2020-10-14 13:13:08.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"da99dbbfbb00a64dd497762ddc038bb2d0da1e87","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-14 14:34:58.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/f243f85a23c646f589aa849e6b6db67d : SUCCESS in 1h 16m 06s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/85972d8a1b71472cbed31580e91345f2 : SUCCESS in 5m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/129cb2688e524619b315357c30eacf0a : SUCCESS in 4m 48s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c9340abbdab940e98b0a78e879610650 : SUCCESS in 4m 31s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/cf32ce068bd7479e91daeaf33dfe294c : SUCCESS in 4m 54s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bfce5bf9b38e418a923d4a7c314297aa : SUCCESS in 4m 47s","accounts_in_message":[],"_revision_number":4},{"id":"88f577490ea8d8ff82160f1b1e7ffdbbf9f692ac","author":{"_account_id":11816,"name":"mike bayer","display_name":"Mike Bayer","email":"mike_mp@zzzcomputing.com","username":"zzzeek","status":"Red Hat"},"date":"2020-10-14 14:41:17.000000000","message":"Uploaded patch set 5: Patch Set 4 was rebased.","accounts_in_message":[],"_revision_number":5},{"id":"0a2e804f7482452cbfa982b8e2bdd6d690582976","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-14 16:20:06.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/5f569bb503594c6592cb9525d730ef13 : SUCCESS in 1h 30m 17s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/162b079486b84a8eb64e80e87e5a4420 : SUCCESS in 7m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/182d8041b1144190879bd1db64046dbd : SUCCESS in 3m 45s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8681fd7d67b04549ba6398f00e4b24a6 : SUCCESS in 4m 49s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/4f2e48f74ac54468926bc6234e94c83c : SUCCESS in 4m 46s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e69ff1e1a610490f849fd7cb295ff925 : SUCCESS in 4m 32s","accounts_in_message":[],"_revision_number":5},{"id":"655f5e46675271d801cb8a834ead049f7610f00a","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-10-15 14:54:24.000000000","message":"Patch Set 5: Code-Review+2","accounts_in_message":[],"_revision_number":5},{"id":"eb15b87102810ed0a9a4461141f52449898bf401","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2020-12-11 14:26:14.000000000","message":"Patch Set 5: Code-Review-1\n\n(3 comments)\n\nI assume this should be -W ? In any case, I\u0027d like to see some prompts to remove the eventually useless stuff while we have the context.","accounts_in_message":[],"_revision_number":5},{"id":"db77ed18506398e332e621aee85df0b36102e62a","author":{"_account_id":11816,"name":"mike bayer","display_name":"Mike Bayer","email":"mike_mp@zzzcomputing.com","username":"zzzeek","status":"Red Hat"},"date":"2020-12-11 16:03:54.000000000","message":"Patch Set 5:\n\n(1 comment)\n\n\u003e Patch Set 5: Code-Review-1\n\u003e \n\u003e (3 comments)\n\u003e \n\u003e I assume this should be -W ? In any case, I\u0027d like to see some prompts to remove the eventually useless stuff while we have the context.\n\n-W meaning... Workflow -1 ?    I wanted there to be reviews of this even though I indicated WIP.  next rev doesn\u0027t need to be WIP.","accounts_in_message":[],"_revision_number":5},{"id":"cfa07764a0425c05882743289e716f65367ed6b5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11816,"name":"mike bayer","display_name":"Mike Bayer","email":"mike_mp@zzzcomputing.com","username":"zzzeek","status":"Red Hat"},"date":"2020-12-11 16:11:29.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"96864102e92f509fad24b4740765dac32f68fec6","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2020-12-11 16:51:33.000000000","message":"Patch Set 6: Code-Review+2\n\n(3 comments)\n\n\u003e \u003e I assume this should be -W ? In any case, I\u0027d like to see some prompts to remove the eventually useless stuff while we have the context.\n\u003e \n\u003e -W meaning... Workflow -1 ?    I wanted there to be reviews of this even though I indicated WIP.  next rev doesn\u0027t need to be WIP.\n\nYeah, sorry. That\u0027s fair.\n\nPS6 lgtm now","accounts_in_message":[],"_revision_number":6},{"id":"4f190cb074d0a6751bc3a39c1cff8dc4fd5ee242","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-11 17:32:33.000000000","message":"Patch Set 6: Verified-1\n\n(6 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/992925cfb8534628a120057a0a8576dc : SUCCESS in 1h 18m 08s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/3aa42dd427664a8eb427ce7057fa6cfe : FAILURE in 3m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/74deb5c632564f52ba9270c33e89725d : FAILURE in 4m 14s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/523ac3888c7a4a8ead7b54189c2a2a58 : SUCCESS in 5m 30s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/fa030eadeb4845d2a72148d0b9b0c3e3 : SUCCESS in 5m 53s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/278f56bf27164046b1e918a7bc58b7dc : SUCCESS in 6m 55s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/915d35adda004578b05a91a4a21f6770 : SUCCESS in 5m 45s","accounts_in_message":[],"_revision_number":6},{"id":"fe709643b20ec9c0a764b418dc4a975e50d4b450","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11816,"name":"mike bayer","display_name":"Mike Bayer","email":"mike_mp@zzzcomputing.com","username":"zzzeek","status":"Red Hat"},"date":"2020-12-11 18:18:47.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"ddba51ec08f6e3dc07b286cda571e2b476208d2c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-11 19:50:05.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/2b498ab12130419491f044eb5c187c90 : SUCCESS in 1h 25m 23s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/4024b6888cbd4dacbcb1de6076e9df33 : FAILURE in 3m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/03926e02e79442b08120f80f43a4b8c3 : SUCCESS in 4m 11s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0113e516e3654af897f6fbf3288338b9 : SUCCESS in 5m 21s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2362d15b78f24711adf02c8a7fc8e36b : SUCCESS in 6m 45s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/453f0934073c4bb688554df48e5b73d6 : SUCCESS in 5m 30s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/248e34c70ab742ee86727204672b06ae : SUCCESS in 5m 25s","accounts_in_message":[],"_revision_number":7},{"id":"78fe7d760433275d1d92b0e3d7293d3d84ceac00","author":{"_account_id":11816,"name":"mike bayer","display_name":"Mike Bayer","email":"mike_mp@zzzcomputing.com","username":"zzzeek","status":"Red Hat"},"date":"2021-01-14 19:15:44.000000000","message":"Patch Set 7:\n\nrecheck","accounts_in_message":[],"_revision_number":7},{"id":"ed163355db1176f25a83a3e7bcac94579a16382a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-14 20:51:54.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/a5862984584f413db33368315640b110 : SUCCESS in 1h 32m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c7f271bd76d043eebc9bd28eaccf51b9 : SUCCESS in 5m 03s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/def0f1ff7e6142ea8f49244d3b8d2a6f : SUCCESS in 4m 39s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/4edbd810925e4c92928f1f94644cb00f : SUCCESS in 5m 09s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/4d10ed72c8bd46e7aca673d5f1303fe9 : SUCCESS in 7m 57s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0517bb6173a240a2832eb198ce508704 : SUCCESS in 5m 09s","accounts_in_message":[],"_revision_number":7},{"id":"044cc29ce10cc0cee23ac26ca4f0e0a2bb1cfbc2","author":{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},"date":"2021-01-14 22:23:05.000000000","message":"Patch Set 7: Code-Review+2","accounts_in_message":[],"_revision_number":7},{"id":"3a9142df9018a4fda49e93df9ceaa5eea00ed302","author":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},"date":"2021-02-02 11:54:00.000000000","message":"Patch Set 7: Code-Review+1","accounts_in_message":[],"_revision_number":7},{"id":"8ab3b161f4ce7f2b4886494f85c234b35ca325c6","author":{"_account_id":8846,"name":"Hao Wang","display_name":"Hao Wang","email":"sxmatch1986@gmail.com","username":"sxmatch"},"date":"2021-03-26 00:57:05.000000000","message":"Patch Set 7: Code-Review+1","accounts_in_message":[],"_revision_number":7},{"id":"a6a48315915b7df71fde7a88fa3b5dd77dc82149","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-04-28 09:51:38.000000000","message":"Patch Set 7: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":7},{"id":"f39d036e08553037ea318baeed71506e40a93fbe","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-28 09:52:07.000000000","message":"Patch Set 7: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":7},{"id":"a687853d3e3f1a9530cdb7d1818dfe0db7b26e2c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-28 11:12:48.000000000","message":"Patch Set 7: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/5af6e28092a74b06a30d83d64e0e8b00 : SUCCESS in 1h 20m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/15a7d5dd1c1f4ed48706574b779f28ff : SUCCESS in 13m 12s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5b387acd63fc48228b09d23ae6d21800 : SUCCESS in 11m 37s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/05bbf4dddc6340a98909a9d4baa8278f : SUCCESS in 7m 44s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/238c075e50a4490ab8ed601f50358abc : SUCCESS in 5m 43s","accounts_in_message":[],"_revision_number":7},{"id":"19cdd967ae5ca8f3201d9099795a8d403a6cb896","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-28 11:12:59.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":7},{"id":"c8b32f33c3ee90c7013f13db080c36a225cb8064","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-28 11:14:54.000000000","message":"Patch Set 7:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a747c069fbee4804b3f7c609f00756ef : SUCCESS in 53s","accounts_in_message":[],"_revision_number":7}],"current_revision_number":7,"current_revision":"f37269ae603bc0e09210379e03a476ff2193c74c","revisions":{"a03b68431fc4d1e81eb28a5c747480ed1a3e90eb":{"kind":"REWORK","_number":1,"created":"2020-08-24 16:01:29.000000000","uploader":{"_account_id":11816,"name":"mike bayer","display_name":"Mike Bayer","email":"mike_mp@zzzcomputing.com","username":"zzzeek","status":"Red Hat"},"ref":"refs/changes/62/747762/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.db","ref":"refs/changes/62/747762/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.db refs/changes/62/747762/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.db refs/changes/62/747762/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.db refs/changes/62/747762/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.db refs/changes/62/747762/1"}}},"commit":{"parents":[{"commit":"9201de3ca9e4ea82ba8fc7cf735b49d06a3e9f11","subject":"Bump bandit version","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.db/commit/9201de3ca9e4ea82ba8fc7cf735b49d06a3e9f11"}]}],"author":{"name":"Mike Bayer","email":"mike_mp@zzzcomputing.com","date":"2020-08-24 16:00:04.000000000","tz":-240},"committer":{"name":"Mike Bayer","email":"mike_mp@zzzcomputing.com","date":"2020-08-24 16:00:04.000000000","tz":-240},"subject":"Accommodate immutable URL api","message":"Accommodate immutable URL api\n\nSQLAlchemy 1.4 is likely going to switch the URL\nobject to a new immutable URL.  This patch makes the\nadjustments needed, using duck-typing to check if the\nURL object is of the new style.\n\nWIP as I am still cross-testing the patch on the\nSQLAlchemy side.\n\nChange-Id: Id2f0663b13ed0f81e91a8d44f73d8541015bf844\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.db/commit/a03b68431fc4d1e81eb28a5c747480ed1a3e90eb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.db/commit/a03b68431fc4d1e81eb28a5c747480ed1a3e90eb"}]},"branch":"refs/heads/master"},"1ec90f4997bc3125499cb781492413068f7cb194":{"kind":"REWORK","_number":2,"created":"2020-08-24 21:04:19.000000000","uploader":{"_account_id":11816,"name":"mike bayer","display_name":"Mike Bayer","email":"mike_mp@zzzcomputing.com","username":"zzzeek","status":"Red Hat"},"ref":"refs/changes/62/747762/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.db","ref":"refs/changes/62/747762/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.db refs/changes/62/747762/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.db refs/changes/62/747762/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.db refs/changes/62/747762/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.db refs/changes/62/747762/2"}}},"commit":{"parents":[{"commit":"9201de3ca9e4ea82ba8fc7cf735b49d06a3e9f11","subject":"Bump bandit version","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.db/commit/9201de3ca9e4ea82ba8fc7cf735b49d06a3e9f11"}]}],"author":{"name":"Mike Bayer","email":"mike_mp@zzzcomputing.com","date":"2020-08-24 16:00:04.000000000","tz":-240},"committer":{"name":"Mike Bayer","email":"mike_mp@zzzcomputing.com","date":"2020-08-24 21:04:13.000000000","tz":-240},"subject":"Accommodate immutable URL api","message":"Accommodate immutable URL api\n\nSQLAlchemy 1.4 is likely going to switch the URL\nobject to a new immutable URL.  This patch makes the\nadjustments needed, using duck-typing to check if the\nURL object is of the new style.\n\nWIP as I am still cross-testing the patch on the\nSQLAlchemy side.\n\nChange-Id: Id2f0663b13ed0f81e91a8d44f73d8541015bf844\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.db/commit/1ec90f4997bc3125499cb781492413068f7cb194"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.db/commit/1ec90f4997bc3125499cb781492413068f7cb194"}]},"branch":"refs/heads/master"},"3724bdc16f735ad057a479fe06c8fff08054fbf8":{"kind":"REWORK","_number":3,"created":"2020-09-04 17:16:17.000000000","uploader":{"_account_id":11816,"name":"mike bayer","display_name":"Mike Bayer","email":"mike_mp@zzzcomputing.com","username":"zzzeek","status":"Red Hat"},"ref":"refs/changes/62/747762/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.db","ref":"refs/changes/62/747762/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.db refs/changes/62/747762/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.db refs/changes/62/747762/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.db refs/changes/62/747762/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.db refs/changes/62/747762/3"}}},"commit":{"parents":[{"commit":"9201de3ca9e4ea82ba8fc7cf735b49d06a3e9f11","subject":"Bump bandit version","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.db/commit/9201de3ca9e4ea82ba8fc7cf735b49d06a3e9f11"}]}],"author":{"name":"Mike Bayer","email":"mike_mp@zzzcomputing.com","date":"2020-08-24 16:00:04.000000000","tz":-240},"committer":{"name":"Mike Bayer","email":"mike_mp@zzzcomputing.com","date":"2020-09-04 17:15:58.000000000","tz":-240},"subject":"Accommodate immutable URL api","message":"Accommodate immutable URL api\n\nSQLAlchemy 1.4 is likely going to switch the URL\nobject to a new immutable URL.  This patch makes the\nadjustments needed, using duck-typing to check if the\nURL object is of the new style.\n\nWIP as I am still cross-testing the patch on the\nSQLAlchemy side.\n\nChange-Id: Id2f0663b13ed0f81e91a8d44f73d8541015bf844\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.db/commit/3724bdc16f735ad057a479fe06c8fff08054fbf8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.db/commit/3724bdc16f735ad057a479fe06c8fff08054fbf8"}]},"branch":"refs/heads/master"},"1181af2069a834f9a9883cbd4a796089766d00e7":{"kind":"REWORK","_number":4,"created":"2020-10-14 13:11:53.000000000","uploader":{"_account_id":11816,"name":"mike bayer","display_name":"Mike Bayer","email":"mike_mp@zzzcomputing.com","username":"zzzeek","status":"Red Hat"},"ref":"refs/changes/62/747762/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.db","ref":"refs/changes/62/747762/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.db refs/changes/62/747762/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.db refs/changes/62/747762/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.db refs/changes/62/747762/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.db refs/changes/62/747762/4"}}},"commit":{"parents":[{"commit":"9201de3ca9e4ea82ba8fc7cf735b49d06a3e9f11","subject":"Bump bandit version","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.db/commit/9201de3ca9e4ea82ba8fc7cf735b49d06a3e9f11"}]}],"author":{"name":"Mike Bayer","email":"mike_mp@zzzcomputing.com","date":"2020-08-24 16:00:04.000000000","tz":-240},"committer":{"name":"Mike Bayer","email":"mike_mp@zzzcomputing.com","date":"2020-10-14 13:10:06.000000000","tz":-240},"subject":"Accommodate immutable URL api","message":"Accommodate immutable URL api\n\nSQLAlchemy 1.4 is likely going to switch the URL\nobject to a new immutable URL.  This patch makes the\nadjustments needed, using duck-typing to check if the\nURL object is of the new style.\n\nWIP as I am still cross-testing the patch on the\nSQLAlchemy side.\n\nChange-Id: Id2f0663b13ed0f81e91a8d44f73d8541015bf844\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.db/commit/1181af2069a834f9a9883cbd4a796089766d00e7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.db/commit/1181af2069a834f9a9883cbd4a796089766d00e7"}]},"branch":"refs/heads/master"},"8f887f657cd869ec431464046f218d9e62129ccc":{"kind":"TRIVIAL_REBASE","_number":5,"created":"2020-10-14 14:41:17.000000000","uploader":{"_account_id":11816,"name":"mike bayer","display_name":"Mike Bayer","email":"mike_mp@zzzcomputing.com","username":"zzzeek","status":"Red Hat"},"ref":"refs/changes/62/747762/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.db","ref":"refs/changes/62/747762/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.db refs/changes/62/747762/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.db refs/changes/62/747762/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.db refs/changes/62/747762/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.db refs/changes/62/747762/5"}}},"commit":{"parents":[{"commit":"aad193a6dc2d3c93f5cc2209ec6cd3d9046b091c","subject":"Imported Translations from Zanata","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.db/commit/aad193a6dc2d3c93f5cc2209ec6cd3d9046b091c"}]}],"author":{"name":"Mike Bayer","email":"mike_mp@zzzcomputing.com","date":"2020-08-24 16:00:04.000000000","tz":-240},"committer":{"name":"Mike Bayer","email":"mike_mp@zzzcomputing.com","date":"2020-10-14 13:20:19.000000000","tz":-240},"subject":"Accommodate immutable URL api","message":"Accommodate immutable URL api\n\nSQLAlchemy 1.4 is likely going to switch the URL\nobject to a new immutable URL.  This patch makes the\nadjustments needed, using duck-typing to check if the\nURL object is of the new style.\n\nWIP as I am still cross-testing the patch on the\nSQLAlchemy side.\n\nChange-Id: Id2f0663b13ed0f81e91a8d44f73d8541015bf844\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.db/commit/8f887f657cd869ec431464046f218d9e62129ccc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.db/commit/8f887f657cd869ec431464046f218d9e62129ccc"}]},"branch":"refs/heads/master"},"1e1f1b56fa8b248aa31aeb1f90b58e1f5e6390dd":{"kind":"REWORK","_number":6,"created":"2020-12-11 16:11:29.000000000","uploader":{"_account_id":11816,"name":"mike bayer","display_name":"Mike Bayer","email":"mike_mp@zzzcomputing.com","username":"zzzeek","status":"Red Hat"},"ref":"refs/changes/62/747762/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.db","ref":"refs/changes/62/747762/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.db refs/changes/62/747762/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.db refs/changes/62/747762/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.db refs/changes/62/747762/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.db refs/changes/62/747762/6"}}},"commit":{"parents":[{"commit":"aad193a6dc2d3c93f5cc2209ec6cd3d9046b091c","subject":"Imported Translations from Zanata","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.db/commit/aad193a6dc2d3c93f5cc2209ec6cd3d9046b091c"}]}],"author":{"name":"Mike Bayer","email":"mike_mp@zzzcomputing.com","date":"2020-08-24 16:00:04.000000000","tz":-240},"committer":{"name":"Mike Bayer","email":"mike_mp@zzzcomputing.com","date":"2020-12-11 16:10:13.000000000","tz":-300},"subject":"Accommodate immutable URL api","message":"Accommodate immutable URL api\n\nSQLAlchemy 1.4 has modified the URL object to be immutable.\nThis patch makes the adjustments needed, using duck-typing\nto check if the URL object is of the new style.  When\nrequirements reach SQLAlchemy 1.4 as the minimum required\nversion, these conditionals should be removed in favor of\nthe noted option in each.\n\nChange-Id: Id2f0663b13ed0f81e91a8d44f73d8541015bf844\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.db/commit/1e1f1b56fa8b248aa31aeb1f90b58e1f5e6390dd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.db/commit/1e1f1b56fa8b248aa31aeb1f90b58e1f5e6390dd"}]},"branch":"refs/heads/master"},"f37269ae603bc0e09210379e03a476ff2193c74c":{"kind":"REWORK","_number":7,"created":"2020-12-11 18:18:47.000000000","uploader":{"_account_id":11816,"name":"mike bayer","display_name":"Mike Bayer","email":"mike_mp@zzzcomputing.com","username":"zzzeek","status":"Red Hat"},"ref":"refs/changes/62/747762/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.db","ref":"refs/changes/62/747762/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.db refs/changes/62/747762/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.db refs/changes/62/747762/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.db refs/changes/62/747762/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.db refs/changes/62/747762/7"}}},"commit":{"parents":[{"commit":"aad193a6dc2d3c93f5cc2209ec6cd3d9046b091c","subject":"Imported Translations from Zanata","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.db/commit/aad193a6dc2d3c93f5cc2209ec6cd3d9046b091c"}]}],"author":{"name":"Mike Bayer","email":"mike_mp@zzzcomputing.com","date":"2020-08-24 16:00:04.000000000","tz":-240},"committer":{"name":"Mike Bayer","email":"mike_mp@zzzcomputing.com","date":"2020-12-11 18:18:42.000000000","tz":-300},"subject":"Accommodate immutable URL api","message":"Accommodate immutable URL api\n\nSQLAlchemy 1.4 has modified the URL object to be immutable.\nThis patch makes the adjustments needed, using duck-typing\nto check if the URL object is of the new style.  When\nrequirements reach SQLAlchemy 1.4 as the minimum required\nversion, these conditionals should be removed in favor of\nthe noted option in each.\n\nChange-Id: Id2f0663b13ed0f81e91a8d44f73d8541015bf844\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.db/commit/f37269ae603bc0e09210379e03a476ff2193c74c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.db/commit/f37269ae603bc0e09210379e03a476ff2193c74c"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
