)]}'
{"id":"openstack%2Foslo.db~756277","triplet_id":"openstack%2Foslo.db~master~I5ac9a4dcd76768d84aa2cee4343f2a2530899f2d","project":"openstack/oslo.db","branch":"master","topic":"ed25519","hashtags":[],"change_id":"I5ac9a4dcd76768d84aa2cee4343f2a2530899f2d","subject":"Add PyNaCl requirement to authenticate with auth_ed25519","status":"ABANDONED","created":"2020-10-06 12:13:05.000000000","updated":"2021-06-22 10:05:37.000000000","total_comment_count":6,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"56fafd12b9a7353c47223057ca2e2accfdaea2bd","_number":756277,"virtual_id_number":756277,"owner":{"_account_id":20778,"name":"Damien Ciabrini","email":"dciabrin@redhat.com","username":"dciabrin"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":11816,"name":"mike bayer","display_name":"Mike Bayer","email":"mike_mp@zzzcomputing.com","username":"zzzeek","status":"Red Hat"},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2020-10-06 13:30:36.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"disliked":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},"all":[{"value":-1,"date":"2020-10-06 14:35:19.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":11816,"name":"mike bayer","display_name":"Mike Bayer","email":"mike_mp@zzzcomputing.com","username":"zzzeek","status":"Red Hat"},{"value":-1,"date":"2020-10-06 12:28:17.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":-1,"date":"2021-06-22 10:03:22.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":-1,"default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":11816,"name":"mike bayer","display_name":"Mike Bayer","email":"mike_mp@zzzcomputing.com","username":"zzzeek","status":"Red Hat"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":11816,"name":"mike bayer","display_name":"Mike Bayer","email":"mike_mp@zzzcomputing.com","username":"zzzeek","status":"Red Hat"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"date":"2021-06-22 10:03:22.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11816,"name":"mike bayer","display_name":"Mike Bayer","email":"mike_mp@zzzcomputing.com","username":"zzzeek","status":"Red Hat"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-10-06 12:28:17.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2020-10-06 13:30:36.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-10-06 14:35:19.000000000","updated_by":{"_account_id":11816,"name":"mike bayer","display_name":"Mike Bayer","email":"mike_mp@zzzcomputing.com","username":"zzzeek","status":"Red Hat"},"reviewer":{"_account_id":11816,"name":"mike bayer","display_name":"Mike Bayer","email":"mike_mp@zzzcomputing.com","username":"zzzeek","status":"Red Hat"},"state":"REVIEWER"},{"updated":"2020-10-07 10:55:21.000000000","updated_by":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},"reviewer":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},"state":"REVIEWER"}],"messages":[{"id":"86e19a2f779c80f1e919f6b38e9e5070d187bff2","author":{"_account_id":20778,"name":"Damien Ciabrini","email":"dciabrin@redhat.com","username":"dciabrin"},"date":"2020-10-06 12:13:05.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"b2930da3d2229b3eb5f91b378119eae7ed200823","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-10-06 12:23:02.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"bbe2ead0263b2ac2ff1d0b7c345afd54919844bb","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-10-06 12:28:17.000000000","message":"Patch Set 1: Code-Review-1\n\nPyNaCl should be added to our lower-constraints.txt [1] too, without this tests will fail.\n\nhttp://codesearch.openstack.org/?q\u003dPyNaCl\u0026i\u003dnope\u0026files\u003d\u0026repos\u003dopenstack/oslo.db","accounts_in_message":[],"_revision_number":1},{"id":"f001de64e8d0936ce4db6efa2ebe1c03e96ce54b","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-10-06 12:30:07.000000000","message":"Patch Set 1:\n\n\u003e PyNaCl should be added to our lower-constraints.txt [1] too,\n \u003e without this tests will fail.\n \u003e \n \u003e http://codesearch.openstack.org/?q\u003dPyNaCl\u0026i\u003dnope\u0026files\u003d\u0026repos\u003dopenstack/oslo.db\n\nHere is the line to add:\n\n```\nPyNaCl\u003d\u003d1.4.0\n````","accounts_in_message":[],"_revision_number":1},{"id":"19dd16187d7abea83661b37c400da8ecbdea0195","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-06 13:30:36.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/e20fe390cfe54afdb7627fd537dace30 : FAILURE in 3m 58s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/891ca9c511094f40b9e5539c92e77cfc : SUCCESS in 1h 13m 09s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/531fdf1e378245c393471ae671fe28c1 : SUCCESS in 5m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/eefd04b86a0b4ded9df78110f8127245 : SUCCESS in 6m 08s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/68e16958ccd445268a15be03f7bf3f28 : SUCCESS in 6m 58s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/be63e00cef90407f82fe36c15c6fd8c3 : SUCCESS in 6m 29s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c040a28d820e4a828d94daeeff5b2c25 : SUCCESS in 5m 03s","accounts_in_message":[],"_revision_number":1},{"id":"feea7bbaed7b7f0e0972f858b5855df8604292fe","author":{"_account_id":11816,"name":"mike bayer","display_name":"Mike Bayer","email":"mike_mp@zzzcomputing.com","username":"zzzeek","status":"Red Hat"},"date":"2020-10-06 14:35:19.000000000","message":"Patch Set 1: Code-Review-1\n\n(2 comments)\n\nso, I can add a unit test to force the dependency to be here but pymysql itself is not in the dependencies.    pynacl is an optional dependency of pymysql.    It would be best if it were packaged downstream with pymysql itself, we should be able to make that happen.","accounts_in_message":[],"_revision_number":1},{"id":"8db9b17bcfcce85dca377d5a9dbaada872974cc5","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-10-07 08:22:52.000000000","message":"Patch Set 1:\n\n(2 comments)\n\nIf you decide to continue with these changes don\u0027t forget to add it to lower-constraints to fix the gate :)","accounts_in_message":[],"_revision_number":1},{"id":"76dc515f082d28f5b95b34f6777173e3b48d2a78","author":{"_account_id":11816,"name":"mike bayer","display_name":"Mike Bayer","email":"mike_mp@zzzcomputing.com","username":"zzzeek","status":"Red Hat"},"date":"2020-10-07 12:49:57.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"08ed9795bcf8ef7975123af286aee183bc75f0b9","author":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},"date":"2021-06-22 10:03:22.000000000","message":"Patch Set 1: Code-Review-1\n\nI think we can close this patch.","accounts_in_message":[],"_revision_number":1},{"id":"56fafd12b9a7353c47223057ca2e2accfdaea2bd","tag":"autogenerated:gerrit:abandon","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-06-22 10:05:37.000000000","message":"Abandoned\n\nI close this review as we seems to follow another approach. Feel free to reopen it if needed.","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"baa74598501584f2b62c898fe4fd4133b08a90e0","revisions":{"baa74598501584f2b62c898fe4fd4133b08a90e0":{"kind":"REWORK","_number":1,"created":"2020-10-06 12:13:05.000000000","uploader":{"_account_id":20778,"name":"Damien Ciabrini","email":"dciabrin@redhat.com","username":"dciabrin"},"ref":"refs/changes/77/756277/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.db","ref":"refs/changes/77/756277/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.db refs/changes/77/756277/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.db refs/changes/77/756277/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.db refs/changes/77/756277/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.db refs/changes/77/756277/1"}}},"commit":{"parents":[{"commit":"80190b395c12aca192f013b076831f3687ccd5a3","subject":"Adding pre-commit","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.db/commit/80190b395c12aca192f013b076831f3687ccd5a3"}]}],"author":{"name":"Damien Ciabrini","email":"dciabrin@redhat.com","date":"2020-10-06 10:03:24.000000000","tz":120},"committer":{"name":"Damien Ciabrini","email":"dciabrin@redhat.com","date":"2020-10-06 12:05:31.000000000","tz":120},"subject":"Add PyNaCl requirement to authenticate with auth_ed25519","message":"Add PyNaCl requirement to authenticate with auth_ed25519\n\nSince PyMySQL 0.10.0, mariadb clients can connect to the\ndatabase with users that have been configured to authenticate\nwith auth_ed25519 [1], a new secure authentication protocol.\n\nWhen auth_ed25519 is configured, PyMySQL requires PyNaCl\nto answer to the crypto challenge that it receives from the\nserver. The authentication is transparent for oslo.db.\n\nSince oslo.db depends on a PyMySQL version that supports\nauth_ed25519, add PyNaCl to the list of requirements so that\noslo.db clients can always use auth_ed25519 if they want to.\nBy default, it is not is use and PyMySQL keeps connecting\nas usual (PyNaCl is not used).\n\n[1] https://mariadb.com/kb/en/authentication-plugin-ed25519\n\nRelated-Bug: #1866093\nChange-Id: I5ac9a4dcd76768d84aa2cee4343f2a2530899f2d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.db/commit/baa74598501584f2b62c898fe4fd4133b08a90e0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.db/commit/baa74598501584f2b62c898fe4fd4133b08a90e0"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
