)]}'
{"id":"openstack%2Foslo.db~760538","triplet_id":"openstack%2Foslo.db~master~I59335b4f318bae2e29ab139cdea089a4d6e14305","project":"openstack/oslo.db","branch":"master","topic":"remove-sqlalchemy-migrate","hashtags":[],"change_id":"I59335b4f318bae2e29ab139cdea089a4d6e14305","subject":"Deprecate \u0027oslo_db.sqlalchemy.migration\u0027 module","status":"MERGED","created":"2020-10-30 10:40:37.000000000","updated":"2020-10-30 16:39:07.000000000","submitted":"2020-10-30 16:37:01.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"760538-1604075821574-294064b8","meta_rev_id":"3e1c68012294117998b0452fde5ccad54e9ca91c","_number":760538,"virtual_id_number":760538,"owner":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"date":"2020-10-30 13:19:46.000000000","_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":11816,"name":"mike bayer","display_name":"Mike Bayer","email":"mike_mp@zzzcomputing.com","username":"zzzeek","status":"Red Hat"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-10-30 16:37:01.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2020-10-30 15:12:19.000000000","_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":2,"date":"2020-10-30 12:36:50.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11816,"name":"mike bayer","display_name":"Mike Bayer","email":"mike_mp@zzzcomputing.com","username":"zzzeek","status":"Red Hat"},{"value":2,"date":"2020-10-30 10:42:13.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":1,"date":"2020-10-30 15:12:19.000000000","_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":11816,"name":"mike bayer","display_name":"Mike Bayer","email":"mike_mp@zzzcomputing.com","username":"zzzeek","status":"Red Hat"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":11816,"name":"mike bayer","display_name":"Mike Bayer","email":"mike_mp@zzzcomputing.com","username":"zzzeek","status":"Red Hat"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11816,"name":"mike bayer","display_name":"Mike Bayer","email":"mike_mp@zzzcomputing.com","username":"zzzeek","status":"Red Hat"},{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-10-30 10:42:13.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2020-10-30 12:36:50.000000000","updated_by":{"_account_id":11816,"name":"mike bayer","display_name":"Mike Bayer","email":"mike_mp@zzzcomputing.com","username":"zzzeek","status":"Red Hat"},"reviewer":{"_account_id":11816,"name":"mike bayer","display_name":"Mike Bayer","email":"mike_mp@zzzcomputing.com","username":"zzzeek","status":"Red Hat"},"state":"REVIEWER"},{"updated":"2020-10-30 15:12:19.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2020-10-30 16:37:01.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"465f0b89bb0b952deb3e36e9b32f6110dd3d32f6","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2020-10-30 10:40:37.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"1afdaff823b00a532ec46952779768576f3dbdde","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-10-30 10:42:13.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"c3644f0142b1287448b8cfaa9ec13b4c5e03b156","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-30 12:00:19.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/7c6c6a95d6fd4fa292e6bfc4298a4de3 : SUCCESS in 1h 16m 56s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/735f1ebacfeb46bb8f13b464f5636fcd : SUCCESS in 4m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3e8659529d6d4cc5b915ab1e5bdf1242 : SUCCESS in 4m 58s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/81d37a59a9ff400a8be2835eaef6c4db : SUCCESS in 5m 42s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9fa71c221aac4ab7b997f1ce92b996c4 : SUCCESS in 5m 35s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/99cdde62f9bf461ea4ab8a8d70601bbb : SUCCESS in 5m 43s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a46bb174b82b4fc092395ec2fde1731c : SUCCESS in 4m 36s","accounts_in_message":[],"_revision_number":1},{"id":"94bc1da789456ed2831e2ffd90a3fee58ff06d20","author":{"_account_id":11816,"name":"mike bayer","display_name":"Mike Bayer","email":"mike_mp@zzzcomputing.com","username":"zzzeek","status":"Red Hat"},"date":"2020-10-30 12:35:34.000000000","message":"Patch Set 1:\n\nit is my understanding that most major Openstack components are still fully on sqlalchemy-migrate with no plans (AFAIK?)  (or resources?) to change that any time soon:\n\nNova\nKeystone\nCinder\nGlance\n\nbasically Neutron is the only core component that\u0027s moved to Alembic AFAIK ?","accounts_in_message":[],"_revision_number":1},{"id":"d1480c18d5253af21d34d263afeacc0d7515055d","author":{"_account_id":11816,"name":"mike bayer","display_name":"Mike Bayer","email":"mike_mp@zzzcomputing.com","username":"zzzeek","status":"Red Hat"},"date":"2020-10-30 12:36:50.000000000","message":"Patch Set 1: Code-Review+2\n\nthough yes we are just deprecating, not removing!   coffee first next time.   of course :)","accounts_in_message":[],"_revision_number":1},{"id":"f41d58a18073ba66d842fb908e84810647996d61","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2020-10-30 13:05:13.000000000","message":"Patch Set 1:\n\n\u003e it is my understanding that most major Openstack components are\n \u003e still fully on sqlalchemy-migrate with no plans (AFAIK?)  (or\n \u003e resources?) to change that any time soon:\n \u003e \n \u003e Nova\n \u003e Keystone\n \u003e Cinder\n \u003e Glance\n \u003e \n \u003e basically Neutron is the only core component that\u0027s moved to\n \u003e Alembic AFAIK ?\n\nDesignate is on the list too, but I\u0027m working on resolving this. I want to resolve nova and glance this cycle and in fact have already started on the latter [1]. Keystone and Cinder would be more work but are totally possible.\n\n[1] https://review.opendev.org/#/q/status:open+project:openstack/glance+branch:master+topic:bp/remove-sqlalchemy-migrate","accounts_in_message":[],"_revision_number":1},{"id":"f50fc87bd67355cafcce6c39a9f855a00131eebc","author":{"_account_id":11816,"name":"mike bayer","display_name":"Mike Bayer","email":"mike_mp@zzzcomputing.com","username":"zzzeek","status":"Red Hat"},"date":"2020-10-30 13:13:54.000000000","message":"Patch Set 1:\n\n\u003e \u003e it is my understanding that most major Openstack components are\n \u003e \u003e still fully on sqlalchemy-migrate with no plans (AFAIK?)  (or\n \u003e \u003e resources?) to change that any time soon:\n \u003e \u003e\n \u003e \u003e Nova\n \u003e \u003e Keystone\n \u003e \u003e Cinder\n \u003e \u003e Glance\n \u003e \u003e\n \u003e \u003e basically Neutron is the only core component that\u0027s moved to\n \u003e \u003e Alembic AFAIK ?\n \u003e \n \u003e Designate is on the list too, but I\u0027m working on resolving this. I\n \u003e want to resolve nova and glance this cycle and in fact have already\n \u003e started on the latter [1]. Keystone and Cinder would be more work\n \u003e but are totally possible.\n \u003e \n \u003e [1] https://review.opendev.org/#/q/status:open+project:openstack/glance+branch:master+topic:bp/remove-sqlalchemy-migrate\n\n wow on Nova?   That would be amazing but wow that\u0027s going to be really tough unless people there are open towards it.","accounts_in_message":[],"_revision_number":1},{"id":"1096d18321063defbf97865894bc6539ef71804f","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2020-10-30 13:19:46.000000000","message":"Patch Set 1:\n\n\u003e \u003e \u003e it is my understanding that most major Openstack components are\n \u003e \u003e \u003e still fully on sqlalchemy-migrate with no plans (AFAIK?)  (or\n \u003e \u003e \u003e resources?) to change that any time soon:\n \u003e \u003e \u003e\n \u003e \u003e \u003e Nova\n \u003e \u003e \u003e Keystone\n \u003e \u003e \u003e Cinder\n \u003e \u003e \u003e Glance\n \u003e \u003e \u003e\n \u003e \u003e \u003e basically Neutron is the only core component that\u0027s moved to\n \u003e \u003e \u003e Alembic AFAIK ?\n \u003e \u003e\n \u003e \u003e Designate is on the list too, but I\u0027m working on resolving this.\n \u003e I\n \u003e \u003e want to resolve nova and glance this cycle and in fact have\n \u003e already\n \u003e \u003e started on the latter [1]. Keystone and Cinder would be more work\n \u003e \u003e but are totally possible.\n \u003e \u003e\n \u003e \u003e [1] https://review.opendev.org/#/q/status:open+project:openstack/glance+branch:master+topic:bp/remove-sqlalchemy-migrate\n \u003e \n \u003e wow on Nova?   That would be amazing but wow that\u0027s going to be\n \u003e really tough unless people there are open towards it.\n\nJust discussed it yesterday, in fact. We\u0027re collapsing all pre-Ussuri migrations this cycle so there\u0027s a natural inflection point. I just need to spec/PoC it up. Wish me luck :)\n\nhttps://etherpad.opendev.org/p/nova-wallaby-ptg (line 363 rn)","accounts_in_message":[],"_revision_number":1},{"id":"4c2564dab8f6c68ae2346ea3bf2cfc0fe4cf1d48","author":{"_account_id":11816,"name":"mike bayer","display_name":"Mike Bayer","email":"mike_mp@zzzcomputing.com","username":"zzzeek","status":"Red Hat"},"date":"2020-10-30 13:32:47.000000000","message":"Patch Set 1:\n\n\u003e \u003e \u003e \u003e it is my understanding that most major Openstack components\n \u003e are\n \u003e \u003e \u003e \u003e still fully on sqlalchemy-migrate with no plans (AFAIK?)  (or\n \u003e \u003e \u003e \u003e resources?) to change that any time soon:\n \u003e \u003e \u003e \u003e\n \u003e \u003e \u003e \u003e Nova\n \u003e \u003e \u003e \u003e Keystone\n \u003e \u003e \u003e \u003e Cinder\n \u003e \u003e \u003e \u003e Glance\n \u003e \u003e \u003e \u003e\n \u003e \u003e \u003e \u003e basically Neutron is the only core component that\u0027s moved to\n \u003e \u003e \u003e \u003e Alembic AFAIK ?\n \u003e \u003e \u003e\n \u003e \u003e \u003e Designate is on the list too, but I\u0027m working on resolving\n \u003e this.\n \u003e \u003e I\n \u003e \u003e \u003e want to resolve nova and glance this cycle and in fact have\n \u003e \u003e already\n \u003e \u003e \u003e started on the latter [1]. Keystone and Cinder would be more\n \u003e work\n \u003e \u003e \u003e but are totally possible.\n \u003e \u003e \u003e\n \u003e \u003e \u003e [1] https://review.opendev.org/#/q/status:open+project:openstack/glance+branch:master+topic:bp/remove-sqlalchemy-migrate\n \u003e \u003e\n \u003e \u003e wow on Nova?   That would be amazing but wow that\u0027s going to be\n \u003e \u003e really tough unless people there are open towards it.\n \u003e \n \u003e Just discussed it yesterday, in fact. We\u0027re collapsing all\n \u003e pre-Ussuri migrations this cycle so there\u0027s a natural inflection\n \u003e point. I just need to spec/PoC it up. Wish me luck :)\n \u003e \n \u003e https://etherpad.opendev.org/p/nova-wallaby-ptg (line 363 rn)\n\nthanks so much for taking that on!    im here.","accounts_in_message":[],"_revision_number":1},{"id":"5cf488f07abcdb1af2a6e37a862741c8ea4be210","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-10-30 15:12:19.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"ee66ac407fce219eca70421b3c1f486b6425c9f2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-30 15:12:31.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"25aca41c929517f08f3f2ffce4cc8d7749edb7bb","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-30 16:37:01.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"19b60a7a3b5dce89acb82d1a76d68590e5418f72","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-30 16:37:01.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/3a3a432cb4a94726a2032600e01314d2 : SUCCESS in 1h 18m 22s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f1f4f7b091eb4ab782e7fc04ee3e9c68 : SUCCESS in 4m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e5c01e936b234f479d68b178e30ff4cd : SUCCESS in 4m 16s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5369918712dc4b04b0efeed3e2359f03 : SUCCESS in 5m 00s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/55f689261c084e5794346ff7c0cb5bb8 : SUCCESS in 4m 59s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/090fbcf83d594ce1b551654b5e4be5a7 : SUCCESS in 5m 08s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7156de31deef4732939dbdfdda200087 : SUCCESS in 5m 51s","accounts_in_message":[],"_revision_number":1},{"id":"a7f5f8a952597c68048f34e05b2732c176444708","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-30 16:39:07.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4c732197d29346ffbbed52c8d4b8d18c : SUCCESS in 1m 26s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d66bae3c39ae4006abcb08a4ac026c0f : SUCCESS in 1m 12s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"fc855875236fd3bf760237fc64092f4a9d864fcb","revisions":{"fc855875236fd3bf760237fc64092f4a9d864fcb":{"kind":"REWORK","_number":1,"created":"2020-10-30 10:40:37.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/38/760538/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.db","ref":"refs/changes/38/760538/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.db refs/changes/38/760538/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.db refs/changes/38/760538/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.db refs/changes/38/760538/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.db refs/changes/38/760538/1"}}},"commit":{"parents":[{"commit":"f62ddb15244d5b5aa0c54f2d6b0ae30799835eb2","subject":"Merge \"Bump default tox env from py37 to py38\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.db/commit/f62ddb15244d5b5aa0c54f2d6b0ae30799835eb2"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2020-10-30 10:12:13.000000000","tz":0},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2020-10-30 10:40:24.000000000","tz":0},"subject":"Deprecate \u0027oslo_db.sqlalchemy.migration\u0027 module","message":"Deprecate \u0027oslo_db.sqlalchemy.migration\u0027 module\n\nThis module hasn\u0027t been touched with years and \u0027sqlalchemy-migrate\u0027 is\neffectively abandoned. It\u0027s time to drop support for this an encourage\nthe laggards to switch to alembic and use it directly.\n\nChange-Id: I59335b4f318bae2e29ab139cdea089a4d6e14305\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.db/commit/fc855875236fd3bf760237fc64092f4a9d864fcb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.db/commit/fc855875236fd3bf760237fc64092f4a9d864fcb"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
