)]}'
{"id":"openstack%2Foslo.db~763248","triplet_id":"openstack%2Foslo.db~master~I30f236c9a3d3d4e278c5551603be3091157387af","project":"openstack/oslo.db","branch":"master","topic":"cleanup_disposes","hashtags":[],"change_id":"I30f236c9a3d3d4e278c5551603be3091157387af","subject":"clean up test engine disposal","status":"ABANDONED","created":"2020-11-18 19:26:55.000000000","updated":"2021-06-22 12:40:07.000000000","total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"324a9b6abb9ee194a819cdef40b15c16be8da792","_number":763248,"virtual_id_number":763248,"owner":{"_account_id":11816,"name":"mike bayer","display_name":"Mike Bayer","email":"mike_mp@zzzcomputing.com","username":"zzzeek","status":"Red Hat"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2020-11-18 21:15:44.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-11-18 21:15:44.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-06-22 10:39:05.000000000","updated_by":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},"reviewer":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},"state":"CC"}],"messages":[{"id":"4f5e621834f424f15bd1159647c97d8c4b7d25c9","author":{"_account_id":11816,"name":"mike bayer","display_name":"Mike Bayer","email":"mike_mp@zzzcomputing.com","username":"zzzeek","status":"Red Hat"},"date":"2020-11-18 19:26:55.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"e6080df0bb21d4b51ab456ed729d7b6ed058c5ec","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-18 21:15:44.000000000","message":"Patch Set 1: Verified-1\n\n(3 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/22034f6bd6984799bb42abf0d6b21c6f : SUCCESS in 1h 45m 36s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/abdbfe3c8533432c97cb880d9e948ca6 : SUCCESS in 6m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fc0fd4ca1b7e42a3b48b8a4c80f1e138 : FAILURE in 5m 12s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0f4bb92721244e2c9d1ee99174ccd7a6 : FAILURE in 7m 15s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3b9b1688756b432fb671c77fa2a8ddbc : FAILURE in 6m 49s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/41af16957caa44ee819e0b195f39774e : FAILURE in 8m 09s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a34c1b0b8626414298cf05ada989c8dd : SUCCESS in 4m 49s","accounts_in_message":[],"_revision_number":1},{"id":"624bb036e0fcd970115ed5171f554e2d93de2cb1","author":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},"date":"2021-06-22 10:39:05.000000000","message":"Patch Set 1:\n\nHi,\n\nThe pach is work in progress and no update for a long time. I would like to know if it\u0027s still revelant? Do you plan to work on it?","accounts_in_message":[],"_revision_number":1},{"id":"668a6bb2c3bfcda68694e17d1e77f111c809bd9e","author":{"_account_id":11816,"name":"mike bayer","display_name":"Mike Bayer","email":"mike_mp@zzzcomputing.com","username":"zzzeek","status":"Red Hat"},"date":"2021-06-22 12:39:56.000000000","message":"Patch Set 1:\n\nI dont even have memory of what this is.    :)    will abandon","accounts_in_message":[],"_revision_number":1},{"id":"324a9b6abb9ee194a819cdef40b15c16be8da792","tag":"autogenerated:gerrit:abandon","author":{"_account_id":11816,"name":"mike bayer","display_name":"Mike Bayer","email":"mike_mp@zzzcomputing.com","username":"zzzeek","status":"Red Hat"},"date":"2021-06-22 12:40:07.000000000","message":"Abandoned\n\nnot clear what I was working on here","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"11bdb891dd4dbeb5c324f23c4e2df4e97804c877","revisions":{"11bdb891dd4dbeb5c324f23c4e2df4e97804c877":{"kind":"REWORK","_number":1,"created":"2020-11-18 19:26:55.000000000","uploader":{"_account_id":11816,"name":"mike bayer","display_name":"Mike Bayer","email":"mike_mp@zzzcomputing.com","username":"zzzeek","status":"Red Hat"},"ref":"refs/changes/48/763248/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.db","ref":"refs/changes/48/763248/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.db refs/changes/48/763248/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.db refs/changes/48/763248/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.db refs/changes/48/763248/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.db refs/changes/48/763248/1"}}},"commit":{"parents":[{"commit":"503db6071acebdabffa994ef86829eb438249db4","subject":"Merge \"Deprecate the \u0027oslo_db.sqlalchemy.migration_cli\u0027 module\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.db/commit/503db6071acebdabffa994ef86829eb438249db4"}]}],"author":{"name":"Mike Bayer","email":"mike_mp@zzzcomputing.com","date":"2020-11-18 19:25:53.000000000","tz":-300},"committer":{"name":"Mike Bayer","email":"mike_mp@zzzcomputing.com","date":"2020-11-18 19:26:48.000000000","tz":-300},"subject":"clean up test engine disposal","message":"clean up test engine disposal\n\nwip\n\nChange-Id: I30f236c9a3d3d4e278c5551603be3091157387af\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.db/commit/11bdb891dd4dbeb5c324f23c4e2df4e97804c877"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.db/commit/11bdb891dd4dbeb5c324f23c4e2df4e97804c877"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
