)]}'
{"id":"openstack%2Foslo.db~778454","triplet_id":"openstack%2Foslo.db~master~Ice1a90edc32603c85449b80fe78fd437aa7e8081","project":"openstack/oslo.db","branch":"master","topic":"remove_sql_max_pool_size","hashtags":[],"change_id":"Ice1a90edc32603c85449b80fe78fd437aa7e8081","subject":"Remove the sql_max_pool_size option.","status":"MERGED","created":"2021-03-03 12:57:43.000000000","updated":"2021-05-03 12:35:57.000000000","submitted":"2021-05-03 12:34:27.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"778454-remove_sql_max_pool_size","meta_rev_id":"054d268f019581e7fd7e177a63d4b53ca41ad818","_number":778454,"virtual_id_number":778454,"owner":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":27954,"name":"Moisés Guimarães de Medeiros","email":"guimaraes@pm.me","username":"moguimar"},{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-05-03 12:34:26.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":27954,"name":"Moisés Guimarães de Medeiros","email":"guimaraes@pm.me","username":"moguimar"},"all":[{"value":2,"date":"2021-05-03 10:54:32.000000000","_account_id":27954,"name":"Moisés Guimarães de Medeiros","email":"guimaraes@pm.me","username":"moguimar"},{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":2,"date":"2021-04-28 10:20:57.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":27954,"name":"Moisés Guimarães de Medeiros","email":"guimaraes@pm.me","username":"moguimar"},"all":[{"value":1,"date":"2021-05-03 10:54:32.000000000","_account_id":27954,"name":"Moisés Guimarães de Medeiros","email":"guimaraes@pm.me","username":"moguimar"},{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":27954,"name":"Moisés Guimarães de Medeiros","email":"guimaraes@pm.me","username":"moguimar"},{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":27954,"name":"Moisés Guimarães de Medeiros","email":"guimaraes@pm.me","username":"moguimar"},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-03-03 12:58:52.000000000","updated_by":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},"reviewer":{"_account_id":27954,"name":"Moisés Guimarães de Medeiros","email":"guimaraes@pm.me","username":"moguimar"},"state":"REVIEWER"},{"updated":"2021-03-03 12:58:52.000000000","updated_by":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2021-03-03 16:00:31.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-04-15 13:53:14.000000000","updated_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"reviewer":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"state":"REVIEWER"}],"messages":[{"id":"c34c62f470760b963bfbaf8723b0b5b43f6c5912","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},"date":"2021-03-03 12:57:43.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"13fde280ff0899bfef6c816cdabe6130c1252539","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},"date":"2021-03-03 14:02:26.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"4ba49687ea8038207095b1ab1d92bf3d3e42495a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-03 16:00:31.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/62f8a68bd2eb41c498b5b90b61549318 : SUCCESS in 1h 47m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/815c5e0de290460181dacb3aa863f691 : FAILURE in 5m 23s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/dcc649131a814dea8fcb0c0f787c533c : FAILURE in 5m 48s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d14d22a41d3d42f7853f26900ea2ac21 : FAILURE in 6m 36s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/926ec8950222496cba86ac6a9de3a66e : FAILURE in 7m 18s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f1c509340e7344a8812898b29b5467fe : SUCCESS in 5m 59s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a79b851208a94f43a19ba8034cf42d0d : SUCCESS in 5m 24s","accounts_in_message":[],"_revision_number":2},{"id":"9cd4f27985fc0bd3965d21294982aecccaff9359","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},"date":"2021-04-07 14:47:50.000000000","message":"Uploaded patch set 3: Patch Set 2 was rebased.","accounts_in_message":[],"_revision_number":3},{"id":"669c9641b59c436f9e0e4994699b40ff88b0825b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-07 16:23:42.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/bcaf959349b444c38a309f6f204f9268 : SUCCESS in 1h 28m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5ecbece9464344ab916bdfd2525c1030 : SUCCESS in 5m 14s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/cba2bb2d22684f3386ae54af31eddd41 : FAILURE in 5m 43s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/017b586a27c4471a91fbb0db49977cb5 : FAILURE in 6m 26s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/62fcd0f90c124399822b53194fdd12d5 : FAILURE in 6m 14s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f9e9e670a0814611b9ac1da7e002fa79 : SUCCESS in 6m 13s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/14f700078def4d9ab97ba533181e41e3 : SUCCESS in 5m 16s","accounts_in_message":[],"_revision_number":3},{"id":"1ee6fc8d210159a6f592f6502a157ef7c907f816","author":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},"date":"2021-04-08 08:09:59.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"eb0f76fb000c46c6468377aa3f01e4053530e98f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-08 10:04:54.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/71278d5d936f46578aed57f4e5b359b1 : SUCCESS in 1h 45m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4bc644a77a524844b7a3c6cecb0bd23c : SUCCESS in 6m 43s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0727d2cabe8b4ec48ee594b7c002bd7b : FAILURE in 7m 05s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0d488ed0dd9241819610adc88901347b : FAILURE in 6m 54s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/89d36ceba1334199a49c3528b4825d32 : FAILURE in 6m 59s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/20c8bb18833a4d1e80ec429adc65b01b : SUCCESS in 6m 42s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3b8b664eb87140afbee92268fbbabc8a : SUCCESS in 5m 18s","accounts_in_message":[],"_revision_number":3},{"id":"cbaff74897d37ab2e7f3b4a926910b7c4167e80c","author":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},"date":"2021-04-13 10:05:11.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"9c3db611ef3d40c796b90974019edd5d2b74b22d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-13 11:57:49.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/ef429274dd5a4c648c8648f9f8f68bd5 : SUCCESS in 1h 50m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/18ef4f8e00ce4e68a93c11c437f8dd7a : SUCCESS in 4m 59s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/284a00ff057540d19fb4eebd23eff53d : FAILURE in 6m 31s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e9b355cdd07942728ff3611666ea698a : FAILURE in 5m 04s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f3541e0d25384aa0ad5cd7f6edbd93d6 : FAILURE in 6m 25s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3b5521fb4aa5429c835606cd7c5bfc7a : SUCCESS in 5m 47s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a9cad0970b0d490492873cc51884a3c7 : SUCCESS in 4m 51s","accounts_in_message":[],"_revision_number":3},{"id":"332b41850e8a07e16a3dfbd6f4f38385ec29eb2b","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2021-04-15 13:53:14.000000000","message":"Patch Set 3: Code-Review-1\n\nTest failures look real. Seems you missed updating some tests.","accounts_in_message":[],"_revision_number":3},{"id":"67fff29cfaab226623d9e396342997c52808be4f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},"date":"2021-04-23 10:34:06.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"b7e5db511fa1bd42893d2601ddf095fb8dc6abb1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-23 12:02:44.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/00ff1e97dbd94c6cab0be093bf8670a9 : SUCCESS in 1h 16m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/21ddba8fb6f34886b91304db0f29a7b7 : SUCCESS in 4m 50s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d4a3408987234f76b368f09171834fbf : FAILURE in 5m 54s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/80afde67aec94414aea760134a0502a0 : FAILURE in 5m 23s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/744f14d3e89d45549405947a99cd4bcc : FAILURE in 4m 59s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8f0b98f428e34ce7a96bc6d6ba4b7935 : SUCCESS in 5m 16s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/af909c4ef6404545965016773de13ac4 : SUCCESS in 6m 33s","accounts_in_message":[],"_revision_number":4},{"id":"468c793d44c8d2858bcf30ec1dac677b08696fe3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},"date":"2021-04-26 10:35:05.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"38f0ef278b9714f79c94840ea8961745447cca39","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-26 11:56:29.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/b551cad13bf34f19a2064df9788c5c10 : SUCCESS in 1h 19m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ec50d1414be647be9328189981bf542d : SUCCESS in 5m 22s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/47382ffdcb264fedb2a78453370ba03e : SUCCESS in 5m 47s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f0b14200b9d44d62b574fa45c748c005 : SUCCESS in 5m 35s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/823be639305a4ffcba93129fa72c1cc2 : SUCCESS in 5m 56s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/94129c14c9e04a0cb2d5e71cc912ecb8 : SUCCESS in 8m 49s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9d4bf3f8de3e47ff977bb59a45e154a8 : SUCCESS in 5m 27s","accounts_in_message":[],"_revision_number":5},{"id":"e48c8cdc051ae10c23538e71eb0d6c7ad514c59d","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-04-28 10:20:57.000000000","message":"Patch Set 5: Code-Review+2\n\nFailure have been addressed by PS5.\nChanges LGTM\nFrom an user point of view I didn\u0027t find real occurence of usage other than commented and deprecated config [1].\n\nI\u0027m good to go with that.\nThanks.\n\n[1] https://codesearch.opendev.org/?q\u003dsql_max_pool\u0026i\u003dnope\u0026files\u003d\u0026excludeFiles\u003d\u0026repos\u003d","accounts_in_message":[],"_revision_number":5},{"id":"9957c592d96ae43af93c66431cdb0cb6b89d5492","author":{"_account_id":27954,"name":"Moisés Guimarães de Medeiros","email":"guimaraes@pm.me","username":"moguimar"},"date":"2021-05-03 10:54:32.000000000","message":"Patch Set 5: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":5},{"id":"f0a0986a33fc5ca44bd77e0ef39f8139b1731690","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-03 10:54:55.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"4af2e413a71234fb1c6549a832b3640f6d8b7a55","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-03 12:34:26.000000000","message":"Patch Set 5: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/193e45e129264d969506c0717d7810ad : SUCCESS in 1h 34m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9f52eee956744fdaa1be9aab04f72797 : SUCCESS in 4m 05s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b26dd45d74764c71a12c27bc056d61c2 : SUCCESS in 7m 20s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/810e5a4e511d49e8bf1bb48945c1ce15 : SUCCESS in 9m 27s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/34b5436252514953a41f11e7449137c5 : SUCCESS in 6m 01s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/629acd6049c140239c27902466c2bdf0 : SUCCESS in 3m 58s","accounts_in_message":[],"_revision_number":5},{"id":"b4d20c7a6caf84468069b47d928b697e33c2400b","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-03 12:34:27.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":5},{"id":"054d268f019581e7fd7e177a63d4b53ca41ad818","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-03 12:35:57.000000000","message":"Patch Set 5:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/46bd281f7b264e1c8292c8a504cf336a : SUCCESS in 57s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/fff79be4f16a47559ab7583f5c167739 : SUCCESS in 1m 02s","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"94141edc5b381e294777378aa1500810794a69b9","revisions":{"c1d64660a2cea5102581a5ef6b4788d972e31f4f":{"kind":"REWORK","_number":1,"created":"2021-03-03 12:57:43.000000000","uploader":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},"ref":"refs/changes/54/778454/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.db","ref":"refs/changes/54/778454/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.db refs/changes/54/778454/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.db refs/changes/54/778454/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.db refs/changes/54/778454/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.db refs/changes/54/778454/1"}}},"commit":{"parents":[{"commit":"9aa6114d746af5f74741d589f9c21280c446a83a","subject":"Merge \"Use py3 as the default runtime for tox\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.db/commit/9aa6114d746af5f74741d589f9c21280c446a83a"}]}],"author":{"name":"Daniel Bengtsson","email":"dbengt@redhat.com","date":"2021-03-03 12:59:51.000000000","tz":60},"committer":{"name":"Daniel Bengtsson","email":"dbengt@redhat.com","date":"2021-03-03 12:59:51.000000000","tz":60},"subject":"Remove the sql_max_pool_size option.","message":"Remove the sql_max_pool_size option.\n\nThe option was replaced 6 years ago[1] by max_pool_size.\n\n[1] https://opendev.org/openstack/oslo.db/commit/1b7c295d522e26aa8bf09242dcc1355c3fb1ab3e\n\nChange-Id: Ice1a90edc32603c85449b80fe78fd437aa7e8081\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.db/commit/c1d64660a2cea5102581a5ef6b4788d972e31f4f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.db/commit/c1d64660a2cea5102581a5ef6b4788d972e31f4f"}]},"branch":"refs/heads/master"},"cf004971536ff07fc4d7466c5cbf3ff3eb3360d7":{"kind":"REWORK","_number":2,"created":"2021-03-03 14:02:26.000000000","uploader":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},"ref":"refs/changes/54/778454/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.db","ref":"refs/changes/54/778454/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.db refs/changes/54/778454/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.db refs/changes/54/778454/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.db refs/changes/54/778454/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.db refs/changes/54/778454/2"}}},"commit":{"parents":[{"commit":"9aa6114d746af5f74741d589f9c21280c446a83a","subject":"Merge \"Use py3 as the default runtime for tox\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.db/commit/9aa6114d746af5f74741d589f9c21280c446a83a"}]}],"author":{"name":"Daniel Bengtsson","email":"dbengt@redhat.com","date":"2021-03-03 12:59:51.000000000","tz":60},"committer":{"name":"Daniel Bengtsson","email":"dbengt@redhat.com","date":"2021-03-03 14:09:10.000000000","tz":60},"subject":"Remove the sql_max_pool_size option.","message":"Remove the sql_max_pool_size option.\n\nThe option was replaced 6 years ago[1] by max_pool_size.\n\n[1] https://opendev.org/openstack/oslo.db/commit/1b7c295d522e26aa8bf09242dcc1355c3fb1ab3e\n\nChange-Id: Ice1a90edc32603c85449b80fe78fd437aa7e8081\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.db/commit/cf004971536ff07fc4d7466c5cbf3ff3eb3360d7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.db/commit/cf004971536ff07fc4d7466c5cbf3ff3eb3360d7"}]},"branch":"refs/heads/master"},"8b5c6e77e5321503ddb15846f038e64a1d0398e9":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2021-04-07 14:47:50.000000000","uploader":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},"ref":"refs/changes/54/778454/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.db","ref":"refs/changes/54/778454/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.db refs/changes/54/778454/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.db refs/changes/54/778454/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.db refs/changes/54/778454/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.db refs/changes/54/778454/3"}}},"commit":{"parents":[{"commit":"9119edc86ffa207bdf244e0d21e163c96a06ee8e","subject":"Merge \"Remove lower-constraints remnants\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.db/commit/9119edc86ffa207bdf244e0d21e163c96a06ee8e"}]}],"author":{"name":"Daniel Bengtsson","email":"dbengt@redhat.com","date":"2021-03-03 12:59:51.000000000","tz":60},"committer":{"name":"Daniel Bengtsson","email":"dbengt@redhat.com","date":"2021-04-07 14:54:32.000000000","tz":120},"subject":"Remove the sql_max_pool_size option.","message":"Remove the sql_max_pool_size option.\n\nThe option was replaced 6 years ago[1] by max_pool_size.\n\n[1] https://opendev.org/openstack/oslo.db/commit/1b7c295d522e26aa8bf09242dcc1355c3fb1ab3e\n\nChange-Id: Ice1a90edc32603c85449b80fe78fd437aa7e8081\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.db/commit/8b5c6e77e5321503ddb15846f038e64a1d0398e9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.db/commit/8b5c6e77e5321503ddb15846f038e64a1d0398e9"}]},"branch":"refs/heads/master"},"71cd73b495714818243e9605c7fa8329ffa000a4":{"kind":"REWORK","_number":4,"created":"2021-04-23 10:34:06.000000000","uploader":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},"ref":"refs/changes/54/778454/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.db","ref":"refs/changes/54/778454/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.db refs/changes/54/778454/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.db refs/changes/54/778454/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.db refs/changes/54/778454/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.db refs/changes/54/778454/4"}}},"commit":{"parents":[{"commit":"9119edc86ffa207bdf244e0d21e163c96a06ee8e","subject":"Merge \"Remove lower-constraints remnants\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.db/commit/9119edc86ffa207bdf244e0d21e163c96a06ee8e"}]}],"author":{"name":"Daniel Bengtsson","email":"dbengt@redhat.com","date":"2021-03-03 12:59:51.000000000","tz":60},"committer":{"name":"Daniel Bengtsson","email":"dbengt@redhat.com","date":"2021-04-23 10:40:51.000000000","tz":120},"subject":"Remove the sql_max_pool_size option.","message":"Remove the sql_max_pool_size option.\n\nThe option was replaced 6 years ago[1] by max_pool_size.\n\n[1] https://opendev.org/openstack/oslo.db/commit/1b7c295d522e26aa8bf09242dcc1355c3fb1ab3e\n\nChange-Id: Ice1a90edc32603c85449b80fe78fd437aa7e8081\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.db/commit/71cd73b495714818243e9605c7fa8329ffa000a4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.db/commit/71cd73b495714818243e9605c7fa8329ffa000a4"}]},"branch":"refs/heads/master"},"94141edc5b381e294777378aa1500810794a69b9":{"kind":"REWORK","_number":5,"created":"2021-04-26 10:35:05.000000000","uploader":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},"ref":"refs/changes/54/778454/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.db","ref":"refs/changes/54/778454/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.db refs/changes/54/778454/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.db refs/changes/54/778454/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.db refs/changes/54/778454/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.db refs/changes/54/778454/5"}}},"commit":{"parents":[{"commit":"9119edc86ffa207bdf244e0d21e163c96a06ee8e","subject":"Merge \"Remove lower-constraints remnants\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.db/commit/9119edc86ffa207bdf244e0d21e163c96a06ee8e"}]}],"author":{"name":"Daniel Bengtsson","email":"dbengt@redhat.com","date":"2021-03-03 12:59:51.000000000","tz":60},"committer":{"name":"Daniel Bengtsson","email":"dbengt@redhat.com","date":"2021-04-26 10:41:58.000000000","tz":120},"subject":"Remove the sql_max_pool_size option.","message":"Remove the sql_max_pool_size option.\n\nThe option was replaced 6 years ago[1] by max_pool_size.\n\n[1] https://opendev.org/openstack/oslo.db/commit/1b7c295d522e26aa8bf09242dcc1355c3fb1ab3e\n\nChange-Id: Ice1a90edc32603c85449b80fe78fd437aa7e8081\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.db/commit/94141edc5b381e294777378aa1500810794a69b9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.db/commit/94141edc5b381e294777378aa1500810794a69b9"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":27954,"name":"Moisés Guimarães de Medeiros","email":"guimaraes@pm.me","username":"moguimar"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":27954,"name":"Moisés Guimarães de Medeiros","email":"guimaraes@pm.me","username":"moguimar"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
