)]}'
{"id":"openstack%2Foslo.db~781361","triplet_id":"openstack%2Foslo.db~stable%2Fwallaby~I9583729a8603d76c33b248f67ccc0bf2b0bd6f61","project":"openstack/oslo.db","branch":"stable/wallaby","topic":"create-wallaby","hashtags":[],"change_id":"I9583729a8603d76c33b248f67ccc0bf2b0bd6f61","subject":"Update .gitreview for stable/wallaby","status":"MERGED","created":"2021-03-18 11:16:50.000000000","updated":"2021-04-16 14:00:08.000000000","submitted":"2021-04-16 13:48:36.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"781361-create-wallaby","meta_rev_id":"0f1ebe1b838b2eea204e6e24b62d5d63d5e663e4","_number":781361,"virtual_id_number":781361,"owner":{"_account_id":22816,"name":"OpenStack Release Bot","email":"infra-root@openstack.org","username":"release","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-04-16 13:48:24.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":2,"date":"2021-04-16 10:21:24.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":1,"date":"2021-04-16 10:21:24.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}],"reviewers":{"REVIEWER":[{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-03-18 12:17:12.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-03-18 12:54:29.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2021-04-16 10:21:24.000000000","updated_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"reviewer":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"state":"REVIEWER"}],"messages":[{"id":"7163ca732f0ab3361bd518ef35939c7688f9eaea","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22816,"name":"OpenStack Release Bot","email":"infra-root@openstack.org","username":"release","tags":["SERVICE_USER"]},"date":"2021-03-18 11:16:50.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"c346f104cd4f82a4e72513d0010df678304174dc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-18 12:17:12.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c228a42e447641fbaab5cceeda16b3ac : FAILURE in 5m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/dadcda2e5fd04bc6a763cf7fd6d8a188 : FAILURE in 5m 32s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c7123f15813247fe86447bc7ab30d7ae : SUCCESS in 5m 45s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3b9bf54d15754e1c9599b714a8088e11 : SUCCESS in 6m 58s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/953237f8338a4aa0a94241527a60ae86 : SUCCESS in 6m 46s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9c018e7fc8fe4625a7bc3541e3b718a0 : SUCCESS in 7m 37s","accounts_in_message":[],"_revision_number":1},{"id":"c0b1f4105645a7a502c5fe8b3d1c185e62fcc524","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-03-18 12:54:29.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.","accounts_in_message":[],"_revision_number":2},{"id":"617a94ab23621962c5a2c98122fe85edcae1e2b3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-18 13:31:04.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/bb5d9a58ca914419bcddcb10e1727214 : FAILURE in 7m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7a96fc19ea574630b6e3efc8185eeda9 : SUCCESS in 5m 04s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1b3035cee74a4f40aac2af6ab901cce9 : SUCCESS in 6m 30s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/da85da261e8d496085a3a16a887b7f52 : SUCCESS in 5m 42s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/0c8f51834f8a4d98882dc3e21f6a09b8 : SUCCESS in 8m 09s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cee6222b76e940bdbfedd6b4efca3bea : SUCCESS in 6m 09s","accounts_in_message":[],"_revision_number":2},{"id":"61bfb49507a7563d75b2e6b8d044fe206191eb7b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-03-23 10:21:16.000000000","message":"Uploaded patch set 3: Patch Set 2 was rebased.","accounts_in_message":[],"_revision_number":3},{"id":"3e369da89595fcb8f05e981e65c4dc1f68244649","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-23 10:34:32.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/b337aa8fc62148469b38e437d15c6fd6 : FAILURE in 4m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fcc9e52dd2e149969992082d6704ec43 : SUCCESS in 5m 13s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1d19d0b4abeb451c8f7257a56a8ed7f5 : SUCCESS in 6m 49s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f709e054e46e465e95e11c1bae2924dd : SUCCESS in 6m 28s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d4a77ba72eaa4c0bb527835eec3448c2 : SUCCESS in 7m 47s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d71ed17d74344be9b558ea2c26f1d6d7 : SUCCESS in 5m 15s","accounts_in_message":[],"_revision_number":3},{"id":"d1d4a6640abb55e487f242573bf71f5ba008c40d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-03-23 12:29:11.000000000","message":"Uploaded patch set 4: Patch Set 3 was rebased.","accounts_in_message":[],"_revision_number":4},{"id":"35e34a5ad8f1f518bae7e957c3a19c1968595008","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-23 12:41:13.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/be6d6e7f6d9848a3a335ff9f288ed61f : FAILURE in 6m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/576cd5484699497bb43fe2dc72f6f15c : SUCCESS in 7m 53s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/25f0441b13ad4da8b50637942d1e4bd5 : SUCCESS in 6m 29s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/79e557420b1949b192e005757e494272 : SUCCESS in 6m 47s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b65e533015b049a5887cf5eeae302eaf : SUCCESS in 5m 47s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/31cdf809c9c547f5be9cf129050da51c : SUCCESS in 5m 41s","accounts_in_message":[],"_revision_number":4},{"id":"2b0af61a09eb0c4d96df16c64504661763bb80af","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2021-04-16 10:20:35.000000000","message":"Topic create-wallaby-patch1 removed","accounts_in_message":[],"_revision_number":4},{"id":"84eb65001739b9dc5dc63bf8aed1a16292d1517f","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2021-04-16 10:20:38.000000000","message":"Topic set to create-wallaby","accounts_in_message":[],"_revision_number":4},{"id":"83ef65d071dd0edba8f8413e464c76b8580957d8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2021-04-16 10:20:55.000000000","message":"Patch Set 5: Patch Set 4 was rebased","accounts_in_message":[],"_revision_number":5},{"id":"46f4d65fc5e261b750b6a127576d1dd8adb88080","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2021-04-16 10:21:24.000000000","message":"Patch Set 5: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":5},{"id":"40d5bcb1f21687ef7f5c4b82314e57ec200af419","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-16 12:03:08.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/2de5535c025941bebfd18a4b2e6556a5 : SUCCESS in 1h 27m 38s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/5b9f6f8a8afd43059062b98d9f24226b : SUCCESS in 5m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/20f850a3c37046189da36c66ebb9f4cc : SUCCESS in 5m 06s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e9e002345f92486aae8a7f5a7ef2ca4f : SUCCESS in 5m 15s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8e6933b696264fce95b3f2fa265ec9bb : SUCCESS in 5m 59s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/4a6871c41a074cdf92e255150b5548e3 : SUCCESS in 5m 30s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a178ff4419c54ce1aa9d6c8407cc0397 : SUCCESS in 4m 58s","accounts_in_message":[],"_revision_number":5},{"id":"7856849796e2925af9525094308991b5e36495be","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-16 12:04:28.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"b7af15083d038a949d8cb6547db4846aa4b5f2a1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-16 13:48:24.000000000","message":"Patch Set 5: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/d87334ea90a944d0a956daee00cba825 : SUCCESS in 1h 38m 12s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/dcff3665eea2415d946d2604c325ecb2 : SUCCESS in 6m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/25959ac19b66402a9d6ae8a7df8095e6 : SUCCESS in 5m 33s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b208bbf5341b4540a02217e68b54ea63 : SUCCESS in 5m 10s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/059943d4e4a444a1b0a60152fc417915 : SUCCESS in 5m 50s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d20b6ca57e214eabbf875bf8186bb12e : SUCCESS in 6m 26s","accounts_in_message":[],"_revision_number":5},{"id":"40a4cbe751373651269fedbf61a27f78add97591","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-16 13:48:36.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":5},{"id":"0f1ebe1b838b2eea204e6e24b62d5d63d5e663e4","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-16 14:00:08.000000000","message":"Patch Set 5:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3b7598ed6e6344f0aa762f4326b98ed2 : SUCCESS in 59s","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"f5b3c68c16c65ba583d9e33b0316115d53636ece","revisions":{"3a6f0b8f568a5a876c19fcf9e7876b9579473748":{"kind":"REWORK","_number":1,"created":"2021-03-18 11:16:50.000000000","uploader":{"_account_id":22816,"name":"OpenStack Release Bot","email":"infra-root@openstack.org","username":"release","tags":["SERVICE_USER"]},"ref":"refs/changes/61/781361/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.db","ref":"refs/changes/61/781361/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.db refs/changes/61/781361/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.db refs/changes/61/781361/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.db refs/changes/61/781361/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.db refs/changes/61/781361/1"}}},"commit":{"parents":[{"commit":"503db6071acebdabffa994ef86829eb438249db4","subject":"Merge \"Deprecate the \u0027oslo_db.sqlalchemy.migration_cli\u0027 module\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.db/commit/503db6071acebdabffa994ef86829eb438249db4"}]}],"author":{"name":"OpenStack Release Bot","email":"infra-root@openstack.org","date":"2021-03-18 11:16:49.000000000","tz":0},"committer":{"name":"OpenStack Release Bot","email":"infra-root@openstack.org","date":"2021-03-18 11:16:49.000000000","tz":0},"subject":"Update .gitreview for stable/wallaby","message":"Update .gitreview for stable/wallaby\n\nChange-Id: I9583729a8603d76c33b248f67ccc0bf2b0bd6f61\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.db/commit/3a6f0b8f568a5a876c19fcf9e7876b9579473748"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.db/commit/3a6f0b8f568a5a876c19fcf9e7876b9579473748"}]},"branch":"refs/heads/stable/wallaby"},"f4c318de2a20121fbf5d668702bf220dfa16e2f1":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2021-03-18 12:54:29.000000000","uploader":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"ref":"refs/changes/61/781361/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.db","ref":"refs/changes/61/781361/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.db refs/changes/61/781361/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.db refs/changes/61/781361/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.db refs/changes/61/781361/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.db refs/changes/61/781361/2"}}},"commit":{"parents":[{"commit":"bc49425097afb66e9379771b47f4a8e2fc364658","subject":"Fix the conflict status with hacking.","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.db/commit/bc49425097afb66e9379771b47f4a8e2fc364658"}]}],"author":{"name":"OpenStack Release Bot","email":"infra-root@openstack.org","date":"2021-03-18 11:16:49.000000000","tz":0},"committer":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2021-03-18 12:54:12.000000000","tz":60},"subject":"Update .gitreview for stable/wallaby","message":"Update .gitreview for stable/wallaby\n\nChange-Id: I9583729a8603d76c33b248f67ccc0bf2b0bd6f61\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.db/commit/f4c318de2a20121fbf5d668702bf220dfa16e2f1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.db/commit/f4c318de2a20121fbf5d668702bf220dfa16e2f1"}]},"branch":"refs/heads/stable/wallaby"},"f52e2770a2da5b0aa52923423ec158124fa76f66":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2021-03-23 10:21:16.000000000","uploader":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"ref":"refs/changes/61/781361/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.db","ref":"refs/changes/61/781361/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.db refs/changes/61/781361/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.db refs/changes/61/781361/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.db refs/changes/61/781361/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.db refs/changes/61/781361/3"}}},"commit":{"parents":[{"commit":"cf7e2a26245c8576851128d3d70f0fd3aef4e1bf","subject":"move flake8 as a pre-commit local target","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.db/commit/cf7e2a26245c8576851128d3d70f0fd3aef4e1bf"}]}],"author":{"name":"OpenStack Release Bot","email":"infra-root@openstack.org","date":"2021-03-18 11:16:49.000000000","tz":0},"committer":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2021-03-23 10:20:55.000000000","tz":60},"subject":"Update .gitreview for stable/wallaby","message":"Update .gitreview for stable/wallaby\n\nChange-Id: I9583729a8603d76c33b248f67ccc0bf2b0bd6f61\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.db/commit/f52e2770a2da5b0aa52923423ec158124fa76f66"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.db/commit/f52e2770a2da5b0aa52923423ec158124fa76f66"}]},"branch":"refs/heads/stable/wallaby"},"384e5f3aefdc70b2d98a786c4be98189b85d8cfe":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2021-03-23 12:29:11.000000000","uploader":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"ref":"refs/changes/61/781361/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.db","ref":"refs/changes/61/781361/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.db refs/changes/61/781361/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.db refs/changes/61/781361/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.db refs/changes/61/781361/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.db refs/changes/61/781361/4"}}},"commit":{"parents":[{"commit":"5d89318a57d4e6056bf1f4a431832d9db8bd372c","subject":"move flake8 as a pre-commit local target","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.db/commit/5d89318a57d4e6056bf1f4a431832d9db8bd372c"}]}],"author":{"name":"OpenStack Release Bot","email":"infra-root@openstack.org","date":"2021-03-18 11:16:49.000000000","tz":0},"committer":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2021-03-23 12:28:57.000000000","tz":60},"subject":"Update .gitreview for stable/wallaby","message":"Update .gitreview for stable/wallaby\n\nChange-Id: I9583729a8603d76c33b248f67ccc0bf2b0bd6f61\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.db/commit/384e5f3aefdc70b2d98a786c4be98189b85d8cfe"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.db/commit/384e5f3aefdc70b2d98a786c4be98189b85d8cfe"}]},"branch":"refs/heads/stable/wallaby"},"f5b3c68c16c65ba583d9e33b0316115d53636ece":{"kind":"TRIVIAL_REBASE","_number":5,"created":"2021-04-16 10:20:55.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/61/781361/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.db","ref":"refs/changes/61/781361/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.db refs/changes/61/781361/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.db refs/changes/61/781361/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.db refs/changes/61/781361/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.db refs/changes/61/781361/5"}}},"commit":{"parents":[{"commit":"8d1e11374b479070bb252330ef7df108492ac1f1","subject":"Unblock wallaby gate","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.db/commit/8d1e11374b479070bb252330ef7df108492ac1f1"}]}],"author":{"name":"OpenStack Release Bot","email":"infra-root@openstack.org","date":"2021-03-18 11:16:49.000000000","tz":0},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2021-04-16 10:20:55.000000000","tz":0},"subject":"Update .gitreview for stable/wallaby","message":"Update .gitreview for stable/wallaby\n\nChange-Id: I9583729a8603d76c33b248f67ccc0bf2b0bd6f61\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.db/commit/f5b3c68c16c65ba583d9e33b0316115d53636ece"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.db/commit/f5b3c68c16c65ba583d9e33b0316115d53636ece"}]},"branch":"refs/heads/stable/wallaby","description":"Rebase"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
