)]}'
{"id":"openstack%2Foslo.db~781470","triplet_id":"openstack%2Foslo.db~master~Iae20f13ba6293b94b5bd131941263e097eba0518","project":"openstack/oslo.db","branch":"master","topic":"improve-flake8","hashtags":[],"change_id":"Iae20f13ba6293b94b5bd131941263e097eba0518","subject":"move flake8 as a pre-commit local target","status":"MERGED","created":"2021-03-18 13:37:56.000000000","updated":"2021-03-19 16:03:10.000000000","submitted":"2021-03-19 15:59:03.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":5,"unresolved_comment_count":2,"has_review_started":true,"submission_id":"781470-improve-flake8","meta_rev_id":"b41f7b1ee6cedb43c8e0450d5f59e414aed3e81c","_number":781470,"virtual_id_number":781470,"owner":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},{"value":0,"_account_id":27954,"name":"Moisés Guimarães de Medeiros","email":"guimaraes@pm.me","username":"moguimar"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-03-19 15:58:44.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},"all":[{"value":2,"date":"2021-03-18 16:37:21.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},{"value":2,"date":"2021-03-19 14:07:36.000000000","_account_id":27954,"name":"Moisés Guimarães de Medeiros","email":"guimaraes@pm.me","username":"moguimar"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":27954,"name":"Moisés Guimarães de Medeiros","email":"guimaraes@pm.me","username":"moguimar"},"all":[{"value":0,"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},{"value":1,"date":"2021-03-19 14:07:36.000000000","_account_id":27954,"name":"Moisés Guimarães de Medeiros","email":"guimaraes@pm.me","username":"moguimar"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},{"value":0,"_account_id":27954,"name":"Moisés Guimarães de Medeiros","email":"guimaraes@pm.me","username":"moguimar"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":27954,"name":"Moisés Guimarães de Medeiros","email":"guimaraes@pm.me","username":"moguimar"},{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-03-18 14:44:46.000000000","updated_by":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},"reviewer":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},"state":"REVIEWER"},{"updated":"2021-03-18 14:49:20.000000000","updated_by":{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},"reviewer":{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},"state":"REVIEWER"},{"updated":"2021-03-18 18:08:09.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-03-19 14:07:36.000000000","updated_by":{"_account_id":27954,"name":"Moisés Guimarães de Medeiros","email":"guimaraes@pm.me","username":"moguimar"},"reviewer":{"_account_id":27954,"name":"Moisés Guimarães de Medeiros","email":"guimaraes@pm.me","username":"moguimar"},"state":"REVIEWER"}],"messages":[{"id":"c2571610ce2a573ead74fbe78127366481b99229","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-03-18 13:37:56.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"1af569738c6dfdfeb410c3a0e6af1751aa0857ad","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-03-18 14:36:23.000000000","message":"Uploaded patch set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"dd3b02464d93ad9cd5a112d3d373f9ece7d80573","author":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},"date":"2021-03-18 14:44:46.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"4fa5b3450d31914e71b907c5f86ab5e6716e50f7","author":{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},"date":"2021-03-18 14:49:20.000000000","message":"Patch Set 2: Code-Review+2\n\n(1 comment)\n\nThis seems better than pinning flake8, so lgtm. We could probably keep the new hacking in this repo, but I wouldn\u0027t block it over that.","accounts_in_message":[],"_revision_number":2},{"id":"38ee6b7e9d496f10eeadfd85deb532ec9dee7cf7","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-03-18 14:58:17.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"bb828df025c4c42eaeb16b3a49ac9145c0ff70cf","author":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},"date":"2021-03-18 15:06:41.000000000","message":"Patch Set 2: Backport-Candidate+1","accounts_in_message":[],"_revision_number":2},{"id":"42f5c323a891f37e5fbf04277f24eb579909710e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-03-18 15:44:57.000000000","message":"Patch Set 3: Commit message was updated.","accounts_in_message":[],"_revision_number":3},{"id":"6de883d70ff65e7d1f108a69b1a5d2615fa05c30","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-03-18 15:45:54.000000000","message":"Patch Set 4: Commit message was updated.","accounts_in_message":[],"_revision_number":4},{"id":"8145939b0ef7ae2f552086fddc374979afc3f567","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-03-18 15:46:31.000000000","message":"Patch Set 5: Commit message was updated.","accounts_in_message":[],"_revision_number":5},{"id":"fbfef93c4a812ddac1f2bb45d5b9642416d65d2c","author":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},"date":"2021-03-18 15:51:12.000000000","message":"Patch Set 5: Code-Review+1","accounts_in_message":[],"_revision_number":5},{"id":"7861a0f9ddb85a14194b6268de6c50c39a64ab62","author":{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},"date":"2021-03-18 15:58:19.000000000","message":"Patch Set 5: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"e009cfe93b253ac797e56ff7aae25aa18a0e7e72","author":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},"date":"2021-03-18 15:59:51.000000000","message":"Patch Set 5:\n\nThanks a lot for the launchpad. I waiting the merge to start to bakcport the patch on other projects.","accounts_in_message":[],"_revision_number":5},{"id":"1754ef12227958d53b037bbd85e9e15bae69cd7c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-03-18 16:24:28.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"a8694ac9b8c173b9af0f379c716a20a584ced3ac","author":{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},"date":"2021-03-18 16:28:48.000000000","message":"Patch Set 6: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"fef2ffee7f61dae2b2de6450b5c2b38db7f690d4","author":{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},"date":"2021-03-18 16:34:05.000000000","message":"Patch Set 6:\n\nJust to record the outcome of our IRC conversation here, I was mistaken in thinking that hacking 4.0.0 was on the wallaby branch. The change merged before branching, but it wasn\u0027t released so it\u0027s not included in stable/wallaby. Moving this back to hacking 3 is the right thing to do so it can be cleanly backported.","accounts_in_message":[],"_revision_number":6},{"id":"576f5ab53f7584bae407dfb484cd72b45127d156","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-03-18 16:34:18.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"d5c1530f87662f812f94055a0537f0e1ee9d9312","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-03-18 16:36:06.000000000","message":"Uploaded patch set 8.","accounts_in_message":[],"_revision_number":8},{"id":"80042e3cec37ba4dd0ff6ab3062b96b466e6fe4d","author":{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},"date":"2021-03-18 16:37:21.000000000","message":"Patch Set 8: Code-Review+2\n\nThanks!","accounts_in_message":[],"_revision_number":8},{"id":"e5a05b107213ebbcef6995c1509b2124aa701acc","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-03-18 16:38:13.000000000","message":"Patch Set 8:\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"f456307cee8dcbd13b95f4c0b8a193cf1d25465d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-18 18:08:09.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/cb766eb6d5a2431c9bda4e3f701e1435 : SUCCESS in 1h 25m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5d6e0ab163534f75bb75766067993848 : SUCCESS in 5m 44s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f16a1909b39145529753beb5c2ccee93 : SUCCESS in 6m 27s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0ba9501489294e81b13140e3e6dbcf26 : SUCCESS in 6m 22s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/5b4a2fcfa58640079119332cd0fcdadc : SUCCESS in 6m 55s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bf7d729d6d5a4181a8c033061c412661 : SUCCESS in 6m 01s","accounts_in_message":[],"_revision_number":8},{"id":"71d84ad106ae72c991ffe0c751d98409dfdc6957","author":{"_account_id":27954,"name":"Moisés Guimarães de Medeiros","email":"guimaraes@pm.me","username":"moguimar"},"date":"2021-03-19 14:07:36.000000000","message":"Patch Set 8: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":8},{"id":"96b5e8509f90d0a31fe84e98c24aaa20850654a1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-19 14:07:55.000000000","message":"Patch Set 8: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":8},{"id":"a4ec98f26a6403dd89413c59b4337270cf02e979","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-19 15:58:44.000000000","message":"Patch Set 8: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/7fa7693cccb24d6993ac0ad9130cc8c5 : SUCCESS in 1h 43m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/db5b63282353436ca68a1216fc2ede6c : SUCCESS in 5m 00s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3c49b1245a8a441fadd02f6356eb434e : SUCCESS in 5m 49s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9521c6e4e2d04aed9361dd0d185ec544 : SUCCESS in 5m 50s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b172ac3b11484cdf81e5149434a30e1a : SUCCESS in 6m 00s","accounts_in_message":[],"_revision_number":8},{"id":"7b16e358e173ac140bebf56291ab95bd4013f4f7","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-19 15:59:03.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":8},{"id":"b41f7b1ee6cedb43c8e0450d5f59e414aed3e81c","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-19 16:03:10.000000000","message":"Patch Set 8:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d5c14c5477344dfeab9634e23c7c2293 : SUCCESS in 1m 03s","accounts_in_message":[],"_revision_number":8}],"current_revision_number":8,"current_revision":"7443bcd079e1c52e8527b5f4b9ce1431aa7cf317","revisions":{"74a0a9f29864b5abc14b1d048c36546d125c94af":{"kind":"REWORK","_number":1,"created":"2021-03-18 13:37:56.000000000","uploader":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"ref":"refs/changes/70/781470/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.db","ref":"refs/changes/70/781470/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.db refs/changes/70/781470/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.db refs/changes/70/781470/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.db refs/changes/70/781470/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.db refs/changes/70/781470/1"}}},"commit":{"parents":[{"commit":"e6d6b90bf440ac0631f659d75b1c54a206d71af3","subject":"Fix the conflict status with hacking.","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.db/commit/e6d6b90bf440ac0631f659d75b1c54a206d71af3"}]}],"author":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2021-03-18 13:36:18.000000000","tz":60},"committer":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2021-03-18 13:36:22.000000000","tz":60},"subject":"[WIP] move flake8 as local target","message":"[WIP] move flake8 as local target\n\nThe goal here is to better handle versions of flake8 and hacking to avoid\nconficts each two weeks.\n\nChange-Id: Iae20f13ba6293b94b5bd131941263e097eba0518\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.db/commit/74a0a9f29864b5abc14b1d048c36546d125c94af"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.db/commit/74a0a9f29864b5abc14b1d048c36546d125c94af"}]},"branch":"refs/heads/master"},"e98d57f07c8733db5d3d2043718c05dc916bd128":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2021-03-18 14:36:23.000000000","uploader":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"ref":"refs/changes/70/781470/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.db","ref":"refs/changes/70/781470/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.db refs/changes/70/781470/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.db refs/changes/70/781470/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.db refs/changes/70/781470/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.db refs/changes/70/781470/2"}}},"commit":{"parents":[{"commit":"e6d6b90bf440ac0631f659d75b1c54a206d71af3","subject":"Fix the conflict status with hacking.","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.db/commit/e6d6b90bf440ac0631f659d75b1c54a206d71af3"}]}],"author":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2021-03-18 13:36:18.000000000","tz":60},"committer":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2021-03-18 14:35:15.000000000","tz":60},"subject":"move flake8 as a pre-commit local target","message":"move flake8 as a pre-commit local target\n\nThe goal here is to avoid conflicts between flake8 and hacking version each\n2 days.\n\nInspired from nova\u0027s approach [1].\n\nThe flake8 version to install will be determined by hacking and\nrequirements will stay aligned instead of relying on different versions.\n\n[1] https://opendev.org/openstack/nova/src/branch/master/.pre-commit-config.yaml#L26-L35\n[2] https://opendev.org/openstack/hacking/src/branch/master/requirements.txt#L1\n\nChange-Id: Iae20f13ba6293b94b5bd131941263e097eba0518\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.db/commit/e98d57f07c8733db5d3d2043718c05dc916bd128"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.db/commit/e98d57f07c8733db5d3d2043718c05dc916bd128"}]},"branch":"refs/heads/master"},"9dbe17afcd103a79e7af86e1af80f8cde0b1ac20":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2021-03-18 15:44:57.000000000","uploader":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"ref":"refs/changes/70/781470/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.db","ref":"refs/changes/70/781470/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.db refs/changes/70/781470/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.db refs/changes/70/781470/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.db refs/changes/70/781470/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.db refs/changes/70/781470/3"}}},"commit":{"parents":[{"commit":"e6d6b90bf440ac0631f659d75b1c54a206d71af3","subject":"Fix the conflict status with hacking.","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.db/commit/e6d6b90bf440ac0631f659d75b1c54a206d71af3"}]}],"author":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2021-03-18 13:36:18.000000000","tz":60},"committer":{"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","date":"2021-03-18 15:44:57.000000000","tz":0},"subject":"move flake8 as a pre-commit local target","message":"move flake8 as a pre-commit local target\n\nThe goal here is to avoid conflicts between flake8 and hacking version each\n2 days.\n\nInspired from nova\u0027s approach [1].\n\nThe flake8 version to install will be determined by hacking and\nrequirements will stay aligned instead of relying on different versions.\n\n[1] https://opendev.org/openstack/nova/src/branch/master/.pre-commit-config.yaml#L26-L35\n[2] https://opendev.org/openstack/hacking/src/branch/master/requirements.txt#L1\n\nChange-Id: Iae20f13ba6293b94b5bd131941263e097eba0518\nCloses-Bug: 1920038\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.db/commit/9dbe17afcd103a79e7af86e1af80f8cde0b1ac20"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.db/commit/9dbe17afcd103a79e7af86e1af80f8cde0b1ac20"}]},"branch":"refs/heads/master","description":"Edit commit message"},"7ebfec48dfe25dd01f1488f0666d73ad45d3ee0d":{"kind":"NO_CODE_CHANGE","_number":4,"created":"2021-03-18 15:45:54.000000000","uploader":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"ref":"refs/changes/70/781470/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.db","ref":"refs/changes/70/781470/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.db refs/changes/70/781470/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.db refs/changes/70/781470/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.db refs/changes/70/781470/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.db refs/changes/70/781470/4"}}},"commit":{"parents":[{"commit":"e6d6b90bf440ac0631f659d75b1c54a206d71af3","subject":"Fix the conflict status with hacking.","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.db/commit/e6d6b90bf440ac0631f659d75b1c54a206d71af3"}]}],"author":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2021-03-18 13:36:18.000000000","tz":60},"committer":{"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","date":"2021-03-18 15:45:54.000000000","tz":0},"subject":"move flake8 as a pre-commit local target","message":"move flake8 as a pre-commit local target\n\nThe goal here is to avoid conflicts between flake8 and hacking version each\n2 days.\n\nInspired from nova\u0027s approach [1].\n\nThe flake8 version to install will be determined by hacking and\nrequirements [2] will stay aligned instead of relying on different versions.\n\n[1] https://opendev.org/openstack/nova/src/branch/master/.pre-commit-config.yaml#L26-L35\n[2] https://opendev.org/openstack/hacking/src/branch/master/requirements.txt#L1\n\nChange-Id: Iae20f13ba6293b94b5bd131941263e097eba0518\nCloses-Bug: 1920038\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.db/commit/7ebfec48dfe25dd01f1488f0666d73ad45d3ee0d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.db/commit/7ebfec48dfe25dd01f1488f0666d73ad45d3ee0d"}]},"branch":"refs/heads/master","description":"Edit commit message"},"d443a2d8a6730f2939616b499bb965633f105242":{"kind":"NO_CODE_CHANGE","_number":5,"created":"2021-03-18 15:46:31.000000000","uploader":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"ref":"refs/changes/70/781470/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.db","ref":"refs/changes/70/781470/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.db refs/changes/70/781470/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.db refs/changes/70/781470/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.db refs/changes/70/781470/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.db refs/changes/70/781470/5"}}},"commit":{"parents":[{"commit":"e6d6b90bf440ac0631f659d75b1c54a206d71af3","subject":"Fix the conflict status with hacking.","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.db/commit/e6d6b90bf440ac0631f659d75b1c54a206d71af3"}]}],"author":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2021-03-18 13:36:18.000000000","tz":60},"committer":{"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","date":"2021-03-18 15:46:31.000000000","tz":0},"subject":"move flake8 as a pre-commit local target","message":"move flake8 as a pre-commit local target\n\nThe goal here is to avoid conflicts between flake8 and hacking version each\n2 days.\n\nInspired from nova\u0027s approach [1].\n\nThe flake8 version to install will be determined by hacking [2] and\nrequirements will stay aligned instead of relying on different versions.\n\n[1] https://opendev.org/openstack/nova/src/branch/master/.pre-commit-config.yaml#L26-L35\n[2] https://opendev.org/openstack/hacking/src/branch/master/requirements.txt#L1\n\nChange-Id: Iae20f13ba6293b94b5bd131941263e097eba0518\nCloses-Bug: 1920038\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.db/commit/d443a2d8a6730f2939616b499bb965633f105242"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.db/commit/d443a2d8a6730f2939616b499bb965633f105242"}]},"branch":"refs/heads/master","description":"Edit commit message"},"33dcd3e13fa9e5d3aa3f21a642aa7194fa156906":{"kind":"REWORK","_number":6,"created":"2021-03-18 16:24:28.000000000","uploader":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"ref":"refs/changes/70/781470/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.db","ref":"refs/changes/70/781470/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.db refs/changes/70/781470/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.db refs/changes/70/781470/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.db refs/changes/70/781470/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.db refs/changes/70/781470/6"}}},"commit":{"parents":[{"commit":"e6d6b90bf440ac0631f659d75b1c54a206d71af3","subject":"Fix the conflict status with hacking.","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.db/commit/e6d6b90bf440ac0631f659d75b1c54a206d71af3"}]}],"author":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2021-03-18 13:36:18.000000000","tz":60},"committer":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2021-03-18 16:24:14.000000000","tz":60},"subject":"move flake8 as a pre-commit local target","message":"move flake8 as a pre-commit local target\n\nThe goal here is to avoid conflicts between flake8 and hacking version each\n2 days.\n\nInspired from nova\u0027s approach [1].\n\nThe flake8 version to install will be determined by hacking and\nrequirements will stay aligned instead of relying on different versions.\n\n[1] https://opendev.org/openstack/nova/src/branch/master/.pre-commit-config.yaml#L26-L35\n[2] https://opendev.org/openstack/hacking/src/branch/master/requirements.txt#L1\n\nChange-Id: Iae20f13ba6293b94b5bd131941263e097eba0518\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.db/commit/33dcd3e13fa9e5d3aa3f21a642aa7194fa156906"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.db/commit/33dcd3e13fa9e5d3aa3f21a642aa7194fa156906"}]},"branch":"refs/heads/master"},"e00d26fae4b351356d361a4f38c8ac023af9c73e":{"kind":"REWORK","_number":7,"created":"2021-03-18 16:34:18.000000000","uploader":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"ref":"refs/changes/70/781470/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.db","ref":"refs/changes/70/781470/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.db refs/changes/70/781470/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.db refs/changes/70/781470/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.db refs/changes/70/781470/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.db refs/changes/70/781470/7"}}},"commit":{"parents":[{"commit":"e6d6b90bf440ac0631f659d75b1c54a206d71af3","subject":"Fix the conflict status with hacking.","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.db/commit/e6d6b90bf440ac0631f659d75b1c54a206d71af3"}]}],"author":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2021-03-18 13:36:18.000000000","tz":60},"committer":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2021-03-18 16:34:08.000000000","tz":60},"subject":"move flake8 as a pre-commit local target","message":"move flake8 as a pre-commit local target\n\nThe goal here is to avoid conflicts between flake8 and hacking version each\n2 days.\n\nInspired from nova\u0027s approach [1].\n\nThe flake8 version to install will be determined by hacking and\nrequirements will stay aligned instead of relying on different versions.\n\n[1] https://opendev.org/openstack/nova/src/branch/master/.pre-commit-config.yaml#L26-L35\n[2] https://opendev.org/openstack/hacking/src/branch/master/requirements.txt#L1\n\nChange-Id: Iae20f13ba6293b94b5bd131941263e097eba0518\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.db/commit/e00d26fae4b351356d361a4f38c8ac023af9c73e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.db/commit/e00d26fae4b351356d361a4f38c8ac023af9c73e"}]},"branch":"refs/heads/master"},"7443bcd079e1c52e8527b5f4b9ce1431aa7cf317":{"kind":"REWORK","_number":8,"created":"2021-03-18 16:36:06.000000000","uploader":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"ref":"refs/changes/70/781470/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.db","ref":"refs/changes/70/781470/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.db refs/changes/70/781470/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.db refs/changes/70/781470/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.db refs/changes/70/781470/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.db refs/changes/70/781470/8"}}},"commit":{"parents":[{"commit":"e6d6b90bf440ac0631f659d75b1c54a206d71af3","subject":"Fix the conflict status with hacking.","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.db/commit/e6d6b90bf440ac0631f659d75b1c54a206d71af3"}]}],"author":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2021-03-18 13:36:18.000000000","tz":60},"committer":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2021-03-18 16:35:56.000000000","tz":60},"subject":"move flake8 as a pre-commit local target","message":"move flake8 as a pre-commit local target\n\nThe goal here is to avoid conflicts between flake8 and hacking version each\n2 days.\n\nInspired from nova\u0027s approach [1].\n\nThe flake8 version to install will be determined by hacking and\nrequirements will stay aligned instead of relying on different versions.\n\n[1] https://opendev.org/openstack/nova/src/branch/master/.pre-commit-config.yaml#L26-L35\n[2] https://opendev.org/openstack/hacking/src/branch/master/requirements.txt#L1\n\nChange-Id: Iae20f13ba6293b94b5bd131941263e097eba0518\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.db/commit/7443bcd079e1c52e8527b5f4b9ce1431aa7cf317"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.db/commit/7443bcd079e1c52e8527b5f4b9ce1431aa7cf317"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":27954,"name":"Moisés Guimarães de Medeiros","email":"guimaraes@pm.me","username":"moguimar"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":27954,"name":"Moisés Guimarães de Medeiros","email":"guimaraes@pm.me","username":"moguimar"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
