)]}'
{"id":"openstack%2Foslo.db~782476","triplet_id":"openstack%2Foslo.db~stable%2Fwallaby~I170ba9d49dacb468a889fbf1c3bd9a37cc15a1b9","project":"openstack/oslo.db","branch":"stable/wallaby","topic":"create-wallaby-patch1","hashtags":[],"change_id":"I170ba9d49dacb468a889fbf1c3bd9a37cc15a1b9","subject":"Remove contradictions in lower-constraints","status":"ABANDONED","created":"2021-03-23 12:29:11.000000000","updated":"2021-03-26 13:21:39.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"d3c69d69921f893be1a44337513b14b317083742","_number":782476,"virtual_id_number":782476,"owner":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2021-03-23 14:39:13.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-03-23 12:40:19.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-03-23 14:26:14.000000000","updated_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"reviewer":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"state":"REVIEWER"}],"messages":[{"id":"3775b2ed24d5b624d4da5d9fbd9930fec24901e5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-03-23 12:29:11.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"fbe088e1aabbd4b69fde8e28c322fa4cd25108f9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-23 12:40:19.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/9e60211cd91e412cb402a0c9e0ec6561 : FAILURE in 3m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/992c951081b143078eaf90ff6dbb67b0 : FAILURE in 5m 32s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c9eda1674c3d452294fe5a44de7f7935 : SUCCESS in 5m 10s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/bc437105b75c41f09a38da1104097aac : SUCCESS in 6m 09s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/55681d53c8d2491093d2674c417489e2 : SUCCESS in 5m 56s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d84cb328be824afa85662acd2b2345b7 : SUCCESS in 5m 35s","accounts_in_message":[],"_revision_number":1},{"id":"7414578d39698828a26ebf261b535e645ebebe6e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2021-03-23 14:26:14.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"8594fc55538fff9fa821355e2dbdd57c4789862b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-23 14:39:13.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/617cdcba7d5c4b51a4fe9a3e1183138e : SUCCESS in 5m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d52945652617448687fb831aab73038b : FAILURE in 4m 48s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/28e0ed98a2d34fd29a97a9cb8742f749 : SUCCESS in 6m 18s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/75299bb4f72d4f37baaac8f4db0800d3 : SUCCESS in 6m 24s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e562577f7fc24fcdb4d39bab1262681d : SUCCESS in 6m 24s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cd692a4a47c54b528f14c511eba0b568 : SUCCESS in 6m 39s","accounts_in_message":[],"_revision_number":2},{"id":"54ea5629493230aeabc1b8a71419fbfa790820c6","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2021-03-23 15:03:46.000000000","message":"Patch Set 2:\n\nNow lower-constraints is passing but pre-commit\u0027s flake8 version (3.8.3) conflicts with what \u0027hacking\u0027 requires (flake8\u003c3.8.0). To resolve this, the pre-commit fix [1] has to be squashed into this patch.\n\n[1] https://review.opendev.org/c/openstack/oslo.db/+/781470","accounts_in_message":[],"_revision_number":2},{"id":"3071d6ce8f1ed63ee8d9619c6250e01dcf341fd0","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-03-26 13:21:34.000000000","message":"Patch Set 2:\n\nDuplicated with https://review.opendev.org/c/openstack/oslo.db/+/782402","accounts_in_message":[],"_revision_number":2},{"id":"d3c69d69921f893be1a44337513b14b317083742","tag":"autogenerated:gerrit:abandon","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-03-26 13:21:39.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"589e07e9392922faef1d52a7fb79574eca6349db","revisions":{"6492dbe705bf251f5e71074114b57e12464e2d5e":{"kind":"REWORK","_number":1,"created":"2021-03-23 12:29:11.000000000","uploader":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"ref":"refs/changes/76/782476/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.db","ref":"refs/changes/76/782476/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.db refs/changes/76/782476/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.db refs/changes/76/782476/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.db refs/changes/76/782476/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.db refs/changes/76/782476/1"}}},"commit":{"parents":[{"commit":"503db6071acebdabffa994ef86829eb438249db4","subject":"Merge \"Deprecate the \u0027oslo_db.sqlalchemy.migration_cli\u0027 module\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.db/commit/503db6071acebdabffa994ef86829eb438249db4"}]}],"author":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2021-03-23 12:26:12.000000000","tz":60},"committer":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2021-03-23 12:28:57.000000000","tz":60},"subject":"Upgrade bandit in lower-constraints","message":"Upgrade bandit in lower-constraints\n\nThis will avoid the following issue seen recently on wallaby and on\nprojects where lower-constraints tests haven\u0027t been yet dropped:\n\n```\nERROR: Could not find a version that satisfies the requirement\nbandit\u003c1.7.0,\u003e\u003d1.6.0\nERROR: No matching distribution found for bandit\u003c1.7.0,\u003e\u003d1.6.0\n```\n\nChange-Id: I170ba9d49dacb468a889fbf1c3bd9a37cc15a1b9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.db/commit/6492dbe705bf251f5e71074114b57e12464e2d5e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.db/commit/6492dbe705bf251f5e71074114b57e12464e2d5e"}]},"branch":"refs/heads/stable/wallaby"},"589e07e9392922faef1d52a7fb79574eca6349db":{"kind":"REWORK","_number":2,"created":"2021-03-23 14:26:14.000000000","uploader":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"ref":"refs/changes/76/782476/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.db","ref":"refs/changes/76/782476/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.db refs/changes/76/782476/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.db refs/changes/76/782476/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.db refs/changes/76/782476/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.db refs/changes/76/782476/2"}}},"commit":{"parents":[{"commit":"503db6071acebdabffa994ef86829eb438249db4","subject":"Merge \"Deprecate the \u0027oslo_db.sqlalchemy.migration_cli\u0027 module\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.db/commit/503db6071acebdabffa994ef86829eb438249db4"}]}],"author":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2021-03-23 12:26:12.000000000","tz":60},"committer":{"name":"Elod Illes","email":"elod.illes@est.tech","date":"2021-03-23 14:22:17.000000000","tz":60},"subject":"Remove contradictions in lower-constraints","message":"Remove contradictions in lower-constraints\n\nThis will avoid the following issue seen recently on wallaby and on\nprojects where lower-constraints tests haven\u0027t been yet dropped:\n\n```\nERROR: Could not find a version that satisfies the requirement\nbandit\u003c1.7.0,\u003e\u003d1.6.0\nERROR: No matching distribution found for bandit\u003c1.7.0,\u003e\u003d1.6.0\n```\n\nAlso some other packages lower constraints has to bump to avoid\nfurther contradictions and slow run of the pip resolver.\n\nChange-Id: I170ba9d49dacb468a889fbf1c3bd9a37cc15a1b9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.db/commit/589e07e9392922faef1d52a7fb79574eca6349db"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.db/commit/589e07e9392922faef1d52a7fb79574eca6349db"}]},"branch":"refs/heads/stable/wallaby"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
