)]}'
{"id":"openstack%2Foslo.db~797426","triplet_id":"openstack%2Foslo.db~master~Idf6fd858fad9521c7c5ba82c31b6d3077756abd9","project":"openstack/oslo.db","branch":"master","topic":"sqlalchemy-1-4-14","hashtags":[],"change_id":"Idf6fd858fad9521c7c5ba82c31b6d3077756abd9","subject":"types: Set \u0027cache_ok\u0027","status":"MERGED","created":"2021-06-22 10:57:23.000000000","updated":"2021-06-25 11:41:01.000000000","submitted":"2021-06-25 11:39:41.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"797426-sqlalchemy-1-4-14","meta_rev_id":"db11230d349239dd5b34fd80368a6fd4de769305","_number":797426,"virtual_id_number":797426,"owner":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},{"value":0,"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"value":0,"_account_id":11816,"name":"mike bayer","display_name":"Mike Bayer","email":"mike_mp@zzzcomputing.com","username":"zzzeek","status":"Red Hat"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-06-25 11:39:41.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},"all":[{"value":2,"date":"2021-06-23 17:06:08.000000000","_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},{"value":0,"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"value":0,"_account_id":11816,"name":"mike bayer","display_name":"Mike Bayer","email":"mike_mp@zzzcomputing.com","username":"zzzeek","status":"Red Hat"},{"value":2,"date":"2021-06-25 10:16:06.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2021-06-23 10:45:12.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"all":[{"value":0,"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},{"value":0,"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"value":0,"_account_id":11816,"name":"mike bayer","display_name":"Mike Bayer","email":"mike_mp@zzzcomputing.com","username":"zzzeek","status":"Red Hat"},{"value":1,"date":"2021-06-25 10:16:06.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},{"value":0,"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"value":0,"_account_id":11816,"name":"mike bayer","display_name":"Mike Bayer","email":"mike_mp@zzzcomputing.com","username":"zzzeek","status":"Red Hat"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"_account_id":11816,"name":"mike bayer","display_name":"Mike Bayer","email":"mike_mp@zzzcomputing.com","username":"zzzeek","status":"Red Hat"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-06-22 10:59:50.000000000","updated_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"reviewer":{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},"state":"REVIEWER"},{"updated":"2021-06-22 10:59:50.000000000","updated_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"reviewer":{"_account_id":11816,"name":"mike bayer","display_name":"Mike Bayer","email":"mike_mp@zzzcomputing.com","username":"zzzeek","status":"Red Hat"},"state":"REVIEWER"},{"updated":"2021-06-22 11:12:30.000000000","updated_by":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},"reviewer":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},"state":"REVIEWER"},{"updated":"2021-06-22 11:33:40.000000000","updated_by":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"reviewer":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"state":"REVIEWER"},{"updated":"2021-06-22 11:38:55.000000000","updated_by":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"reviewer":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"state":"REVIEWER"},{"updated":"2021-06-22 11:51:14.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2021-06-22 12:16:16.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"84e446e77223bef358526947941e2b2cb7716167","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2021-06-22 10:57:23.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"8fad293849aee030ce093b8417c29be2da06f3d0","author":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},"date":"2021-06-22 11:12:30.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"3c97f0a4618bf3082a3daac0a8d870eb23f13769","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2021-06-22 11:33:40.000000000","message":"Patch Set 1: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"10ae71384e8d520ef690ef98485e100a30d3023b","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2021-06-22 11:38:55.000000000","message":"Patch Set 1: Code-Review+1\n\nyes please","accounts_in_message":[],"_revision_number":1},{"id":"1465c05a6090d6ab0d4fd0c96acb159b223b8043","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-06-22 11:51:14.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"26017d125c81f4faff1222806aa752790b8dd7b9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-22 12:16:16.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/9a3bb1212a9143eca24e5190d8a7b103 : SUCCESS in 1h 17m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/046bb5ea7d684ad798a68d1d3817ede9 : SUCCESS in 5m 32s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/58b9c3fade214ad69ae0690aa95565b4 : SUCCESS in 6m 30s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/cbdbdfadeb304ab98f3ca7414f521772 : SUCCESS in 6m 11s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/37aad5626a7545e2b15d9d1feee3df4e : SUCCESS in 6m 47s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1463d253edef4e1c8c9c1c325996456a : FAILURE in 5m 47s","accounts_in_message":[],"_revision_number":1},{"id":"c2538bcb66b5d630d87f4a18ff6fab8e57d837f3","author":{"_account_id":11816,"name":"mike bayer","display_name":"Mike Bayer","email":"mike_mp@zzzcomputing.com","username":"zzzeek","status":"Red Hat"},"date":"2021-06-22 12:43:50.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"3b0b154e0cbc6ed98a092bb573e2ffeead522fb4","author":{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},"date":"2021-06-22 16:17:20.000000000","message":"Patch Set 1:\n\nLooks like this has upset the docs job:\n\ndocstring of oslo_db.sqlalchemy.types.JsonEncodedType.cache_ok:37: WARNING: undefined label: sql_caching","accounts_in_message":[],"_revision_number":1},{"id":"410f39f47f05bba23f8186f94ce2c90356f7efb1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2021-06-22 17:10:56.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"1cb32bef7574034c23d7bb34a81035d306b70503","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2021-06-22 17:11:16.000000000","message":"Patch Set 1:\n\n\u003e Patch Set 1:\n\u003e \n\u003e Looks like this has upset the docs job:\n\u003e \n\u003e docstring of oslo_db.sqlalchemy.types.JsonEncodedType.cache_ok:37: WARNING: undefined label: sql_caching\n\nWeird, looks like we were inheriting from SQLAlchemy\u0027s docs. Fixed now.","accounts_in_message":[],"_revision_number":1},{"id":"169dea1011e88ae99f1624b91e86f937549fb7d9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-22 18:30:52.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/eb6e75901e70452a9bd0016d40505974 : SUCCESS in 1h 12m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/06435f9b6eb244f2bdb751404a953474 : SUCCESS in 4m 43s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6aad312a3e3d49449f7671a99fee32e3 : SUCCESS in 10m 43s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f2809fcadcdf4bc4b3019960d5b35f9b : SUCCESS in 7m 10s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/1cef194350964ab09afcb456914f59df : SUCCESS in 8m 03s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ac5d554a558b4d3db8cdcce48642c482 : SUCCESS in 7m 08s","accounts_in_message":[],"_revision_number":2},{"id":"ef202fac151b1ad78bb6c9ed87f128d947790d9b","author":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},"date":"2021-06-23 10:45:12.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"229f37935124c2812ea4ed4c85387b8d48fef491","author":{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},"date":"2021-06-23 17:06:08.000000000","message":"Patch Set 2: Code-Review+2\n\nDocs look good now.","accounts_in_message":[],"_revision_number":2},{"id":"a45cf8ba3e0f6ab810acbc3787bed23ebbaf7a9f","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-06-25 10:16:06.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1\n\nLGTM Thanks","accounts_in_message":[],"_revision_number":2},{"id":"cd5dc5849924c5542ea5d11cc25aecd2785d925f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-25 10:16:26.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"a4cabdca54fe3552f6798a85ac35e8bbe3706290","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-25 11:39:41.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/96f8bd620b3c415b84f1fb3d598fb86a : SUCCESS in 1h 19m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/971b7c82c3984bf7b297858d9307b9ee : SUCCESS in 7m 43s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a833e2b762dd45e7b2fa642328d8d46f : SUCCESS in 6m 56s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/4ca5039508054f898104a25f146a2f90 : SUCCESS in 9m 25s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/621a1aafdb174f72b33279eb1510b697 : SUCCESS in 8m 35s","accounts_in_message":[],"_revision_number":2},{"id":"09cd33465296d2e6bf494ff8bd8c6e338130ba57","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-25 11:39:41.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"db11230d349239dd5b34fd80368a6fd4de769305","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-25 11:41:01.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b9be78af546747fa88bedb306f0b60df : SUCCESS in 1m 04s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"1dc20f646b558354e4ba434f4132a1b7979d563e","revisions":{"6af408fe42aceae1a01f411d662684d187b97892":{"kind":"REWORK","_number":1,"created":"2021-06-22 10:57:23.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/26/797426/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.db","ref":"refs/changes/26/797426/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.db refs/changes/26/797426/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.db refs/changes/26/797426/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.db refs/changes/26/797426/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.db refs/changes/26/797426/1"}}},"commit":{"parents":[{"commit":"be2cc6a2ce40f6666d1e2b0f13054e7be7bc792e","subject":"Followup of \"Added handler for mysql 8.0.19 duplicate key error update\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.db/commit/be2cc6a2ce40f6666d1e2b0f13054e7be7bc792e"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2021-06-22 10:54:01.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2021-06-22 10:54:01.000000000","tz":60},"subject":"types: Set \u0027cache_ok\u0027","message":"types: Set \u0027cache_ok\u0027\n\nResolves the following warning that has been popping up on tests:\n\n  SAWarning: TypeDecorator SoftDeleteInteger() will not produce a cache\n  key because the ``cache_ok`` flag is not set to True.  Set this flag\n  to True if this type object\u0027s state is safe to use in a cache key, or\n  False to disable this warning.\n\nBoth SoftDeleteInteger and the various subclasses of \u0027JsonEncodedType\u0027\nare hashable and should return a consistent hash and key. For more\ninformation, refer to the SQLAlchemy docs [1].\n\n[1] https://docs.sqlalchemy.org/en/14/core/custom_types.html#sqlalchemy.types.TypeDecorator.cache_ok\n\nChange-Id: Idf6fd858fad9521c7c5ba82c31b6d3077756abd9\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.db/commit/6af408fe42aceae1a01f411d662684d187b97892"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.db/commit/6af408fe42aceae1a01f411d662684d187b97892"}]},"branch":"refs/heads/master"},"1dc20f646b558354e4ba434f4132a1b7979d563e":{"kind":"REWORK","_number":2,"created":"2021-06-22 17:10:56.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/26/797426/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.db","ref":"refs/changes/26/797426/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.db refs/changes/26/797426/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.db refs/changes/26/797426/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.db refs/changes/26/797426/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.db refs/changes/26/797426/2"}}},"commit":{"parents":[{"commit":"be2cc6a2ce40f6666d1e2b0f13054e7be7bc792e","subject":"Followup of \"Added handler for mysql 8.0.19 duplicate key error update\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.db/commit/be2cc6a2ce40f6666d1e2b0f13054e7be7bc792e"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2021-06-22 10:54:01.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2021-06-22 17:10:46.000000000","tz":60},"subject":"types: Set \u0027cache_ok\u0027","message":"types: Set \u0027cache_ok\u0027\n\nResolves the following warning that has been popping up on tests:\n\n  SAWarning: TypeDecorator SoftDeleteInteger() will not produce a cache\n  key because the ``cache_ok`` flag is not set to True.  Set this flag\n  to True if this type object\u0027s state is safe to use in a cache key, or\n  False to disable this warning.\n\nBoth SoftDeleteInteger and the various subclasses of \u0027JsonEncodedType\u0027\nare hashable and should return a consistent hash and key. For more\ninformation, refer to the SQLAlchemy docs [1].\n\n[1] https://docs.sqlalchemy.org/en/14/core/custom_types.html#sqlalchemy.types.TypeDecorator.cache_ok\n\nChange-Id: Idf6fd858fad9521c7c5ba82c31b6d3077756abd9\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.db/commit/1dc20f646b558354e4ba434f4132a1b7979d563e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.db/commit/1dc20f646b558354e4ba434f4132a1b7979d563e"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
