)]}'
{"id":"openstack%2Foslo.db~799224","triplet_id":"openstack%2Foslo.db~master~If537934713986997fa531946551c8a408d83c925","project":"openstack/oslo.db","branch":"master","topic":"session_thread_master","hashtags":[],"change_id":"If537934713986997fa531946551c8a408d83c925","subject":"[WIP] Make SQLAlchemy session thread-safe","status":"ABANDONED","created":"2021-07-02 15:48:37.000000000","updated":"2023-04-17 09:41:30.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"26dc35da37900b395c60dd75e37421feb2148980","_number":799224,"virtual_id_number":799224,"owner":{"_account_id":27838,"name":"Vadym Markov","email":"vmarkov@mirantis.com","username":"vmarkov"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2021-07-02 16:23:21.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-07-02 16:23:21.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"e43947c8db2e5fd7669eeb33bf65e41951ac8150","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27838,"name":"Vadym Markov","email":"vmarkov@mirantis.com","username":"vmarkov"},"date":"2021-07-02 15:48:37.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"2b47262c1a4462b2089cb0720d0cd365e034c2e9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-02 16:23:21.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/9235c0d5e36441c8854d14e55cce915d : FAILURE in 28m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b917edbd6def472198bf8fb38a40ef22 : SUCCESS in 5m 31s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f2de93dfdd464924ab98c1504feadff3 : FAILURE in 6m 24s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ad67258f41c246528cb15bf48072897b : FAILURE in 7m 06s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/9bd430fe101244fbb6b6cf8ebc788628 : FAILURE in 8m 01s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/92a2b41b4cc245a4bc44a65585998a3c : SUCCESS in 7m 35s","accounts_in_message":[],"_revision_number":1},{"id":"26dc35da37900b395c60dd75e37421feb2148980","tag":"autogenerated:gerrit:abandon","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2023-04-17 09:41:30.000000000","message":"Abandoned\n\nThis has been in a WIP state for a long time. Closing to keep our review queue clear. Please feel free to reopen if you pick this up again.","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"7fccc3a8bfac19fc2b6f433b5a4ef9feb0aeb92e","revisions":{"7fccc3a8bfac19fc2b6f433b5a4ef9feb0aeb92e":{"kind":"REWORK","_number":1,"created":"2021-07-02 15:48:37.000000000","uploader":{"_account_id":27838,"name":"Vadym Markov","email":"vmarkov@mirantis.com","username":"vmarkov"},"ref":"refs/changes/24/799224/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.db","ref":"refs/changes/24/799224/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.db refs/changes/24/799224/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.db refs/changes/24/799224/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.db refs/changes/24/799224/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.db refs/changes/24/799224/1"}}},"commit":{"parents":[{"commit":"f3bdd4d4220a884e4752ce6b61c2605bfd72f2fb","subject":"Merge \"types: Set \u0027cache_ok\u0027\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.db/commit/f3bdd4d4220a884e4752ce6b61c2605bfd72f2fb"}]}],"author":{"name":"Vadym Markov","email":"vmarkov@mirantis.com","date":"2021-06-18 15:26:38.000000000","tz":180},"committer":{"name":"Vadym Markov","email":"vmarkov@mirantis.com","date":"2021-07-02 15:41:33.000000000","tz":180},"subject":"[WIP] Make SQLAlchemy session thread-safe","message":"[WIP] Make SQLAlchemy session thread-safe\n\nUsed scoped session instead of the regular one. Tests are totally broken\nfor now\n\nChange-Id: If537934713986997fa531946551c8a408d83c925\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.db/commit/7fccc3a8bfac19fc2b6f433b5a4ef9feb0aeb92e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.db/commit/7fccc3a8bfac19fc2b6f433b5a4ef9feb0aeb92e"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
