)]}'
{"id":"openstack%2Foslo.db~802562","triplet_id":"openstack%2Foslo.db~master~Idb3926d73ff3c11abe9c273062aca1b299b9f72e","project":"openstack/oslo.db","branch":"master","topic":"sqlalchemy-20","hashtags":[],"change_id":"Idb3926d73ff3c11abe9c273062aca1b299b9f72e","subject":"tox: Simplify test running","status":"MERGED","created":"2021-07-27 17:30:19.000000000","updated":"2021-07-27 20:53:34.000000000","submitted":"2021-07-27 20:52:04.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"802562-sqlalchemy-20","meta_rev_id":"231ca9eb616c2ff98058b482a4362ad5ecc8d54a","_number":802562,"virtual_id_number":802562,"owner":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11816,"name":"mike bayer","display_name":"Mike Bayer","email":"mike_mp@zzzcomputing.com","username":"zzzeek","status":"Red Hat"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-07-27 20:52:04.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11816,"name":"mike bayer","display_name":"Mike Bayer","email":"mike_mp@zzzcomputing.com","username":"zzzeek","status":"Red Hat"},"all":[{"value":2,"date":"2021-07-27 19:08:36.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11816,"name":"mike bayer","display_name":"Mike Bayer","email":"mike_mp@zzzcomputing.com","username":"zzzeek","status":"Red Hat"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11816,"name":"mike bayer","display_name":"Mike Bayer","email":"mike_mp@zzzcomputing.com","username":"zzzeek","status":"Red Hat"},"all":[{"value":1,"date":"2021-07-27 19:08:36.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11816,"name":"mike bayer","display_name":"Mike Bayer","email":"mike_mp@zzzcomputing.com","username":"zzzeek","status":"Red Hat"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":11816,"name":"mike bayer","display_name":"Mike Bayer","email":"mike_mp@zzzcomputing.com","username":"zzzeek","status":"Red Hat"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11816,"name":"mike bayer","display_name":"Mike Bayer","email":"mike_mp@zzzcomputing.com","username":"zzzeek","status":"Red Hat"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-07-27 18:55:39.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-07-27 19:08:36.000000000","updated_by":{"_account_id":11816,"name":"mike bayer","display_name":"Mike Bayer","email":"mike_mp@zzzcomputing.com","username":"zzzeek","status":"Red Hat"},"reviewer":{"_account_id":11816,"name":"mike bayer","display_name":"Mike Bayer","email":"mike_mp@zzzcomputing.com","username":"zzzeek","status":"Red Hat"},"state":"REVIEWER"}],"messages":[{"id":"42af9c0befb881f658255dbe7ad6d6a3f3785338","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2021-07-27 17:30:19.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"00d89588e6de4f179096afbeb7f40f183bac77ec","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-27 18:55:39.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/ee6639086a574611ab79dad690df6934 : SUCCESS in 1h 23m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0dd1879b668c4ff0b465ce85573801dd : SUCCESS in 5m 31s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1f2b6f86c97c4e2aa16c06dcf0f21f43 : SUCCESS in 5m 44s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6934052af0914335925cedbd03957f91 : SUCCESS in 5m 21s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/30b0fda78c0145bea35f6970fa10d292 : SUCCESS in 6m 34s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1b0611bc687444479eb573416857b6e6 : SUCCESS in 5m 57s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ab72a962d91b44f8a9381dda6ece16ee : SUCCESS in 5m 29s","accounts_in_message":[],"_revision_number":1},{"id":"283330da5c0d3041be9476710d0f162ceb06f8b3","author":{"_account_id":11816,"name":"mike bayer","display_name":"Mike Bayer","email":"mike_mp@zzzcomputing.com","username":"zzzeek","status":"Red Hat"},"date":"2021-07-27 19:08:36.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\ni read the code and i said, wait were we running 2x as longer? then the commit message backed it up!","accounts_in_message":[],"_revision_number":1},{"id":"ec2444a1ca61f87d7607a9939207fb8cb995462e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-27 19:09:04.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"0f2aae7a48ea52833fdc65834cd97d9013c6f51f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-27 20:52:04.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/5f3b297f2038469d84cc6037230cdf57 : SUCCESS in 1h 37m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/10a438486d75428aa78c09bc81041abe : SUCCESS in 4m 44s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2230b43341f145a9a42ca837ccddd571 : SUCCESS in 5m 13s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d1f353dfda714834936cee8082c1fe89 : SUCCESS in 5m 08s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7bee2a1c109f4733b01ad1a06615c577 : SUCCESS in 5m 09s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/47af0e5b0c3f4df998e7e56388ce4879 : SUCCESS in 4m 41s","accounts_in_message":[],"_revision_number":1},{"id":"eedcce4aedc55a843894f94d28c221ea495fbb09","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-27 20:52:04.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"231ca9eb616c2ff98058b482a4362ad5ecc8d54a","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-27 20:53:34.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b4033004787c40079be61d5b77fbb60a : SUCCESS in 1m 10s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b993e28a06734a8c957fbbdf59a6a216 : SUCCESS in 54s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"f048068216c655b6babc5740fc4b5e7efcf02ab1","revisions":{"f048068216c655b6babc5740fc4b5e7efcf02ab1":{"kind":"REWORK","_number":1,"created":"2021-07-27 17:30:19.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/62/802562/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.db","ref":"refs/changes/62/802562/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.db refs/changes/62/802562/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.db refs/changes/62/802562/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.db refs/changes/62/802562/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.db refs/changes/62/802562/1"}}},"commit":{"parents":[{"commit":"f3bdd4d4220a884e4752ce6b61c2605bfd72f2fb","subject":"Merge \"types: Set \u0027cache_ok\u0027\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.db/commit/f3bdd4d4220a884e4752ce6b61c2605bfd72f2fb"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2021-07-16 11:14:34.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2021-07-16 12:01:45.000000000","tz":60},"subject":"tox: Simplify test running","message":"tox: Simplify test running\n\nWe were actually running tests four times. Twice with eventlet disabled\nand twice with it enabled. Correct this.\n\nChange-Id: Idb3926d73ff3c11abe9c273062aca1b299b9f72e\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.db/commit/f048068216c655b6babc5740fc4b5e7efcf02ab1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.db/commit/f048068216c655b6babc5740fc4b5e7efcf02ab1"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":11816,"name":"mike bayer","display_name":"Mike Bayer","email":"mike_mp@zzzcomputing.com","username":"zzzeek","status":"Red Hat"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":11816,"name":"mike bayer","display_name":"Mike Bayer","email":"mike_mp@zzzcomputing.com","username":"zzzeek","status":"Red Hat"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
