)]}'
{"id":"openstack%2Foslo.db~802563","triplet_id":"openstack%2Foslo.db~master~I110392586fb032507d122ae9a412a6bba7c5ff2d","project":"openstack/oslo.db","branch":"master","topic":"sqlalchemy-20","hashtags":[],"change_id":"I110392586fb032507d122ae9a412a6bba7c5ff2d","subject":"tests: Clean up base test","status":"MERGED","created":"2021-07-27 17:30:19.000000000","updated":"2021-08-10 14:29:37.000000000","submitted":"2021-08-10 14:28:12.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"802563-sqlalchemy-20","meta_rev_id":"52e1da25e79f37b1509cb48c6977004849ba0672","_number":802563,"virtual_id_number":802563,"owner":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11816,"name":"mike bayer","display_name":"Mike Bayer","email":"mike_mp@zzzcomputing.com","username":"zzzeek","status":"Red Hat"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-08-10 14:28:12.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11816,"name":"mike bayer","display_name":"Mike Bayer","email":"mike_mp@zzzcomputing.com","username":"zzzeek","status":"Red Hat"},"all":[{"value":2,"date":"2021-08-10 14:29:37.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":11816,"name":"mike bayer","display_name":"Mike Bayer","email":"mike_mp@zzzcomputing.com","username":"zzzeek","status":"Red Hat"},{"value":2,"date":"2021-08-10 13:04:10.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"all":[{"value":0,"_account_id":11816,"name":"mike bayer","display_name":"Mike Bayer","email":"mike_mp@zzzcomputing.com","username":"zzzeek","status":"Red Hat"},{"value":1,"date":"2021-08-10 13:04:10.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":11816,"name":"mike bayer","display_name":"Mike Bayer","email":"mike_mp@zzzcomputing.com","username":"zzzeek","status":"Red Hat"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11816,"name":"mike bayer","display_name":"Mike Bayer","email":"mike_mp@zzzcomputing.com","username":"zzzeek","status":"Red Hat"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-07-27 18:47:35.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-07-27 19:07:17.000000000","updated_by":{"_account_id":11816,"name":"mike bayer","display_name":"Mike Bayer","email":"mike_mp@zzzcomputing.com","username":"zzzeek","status":"Red Hat"},"reviewer":{"_account_id":11816,"name":"mike bayer","display_name":"Mike Bayer","email":"mike_mp@zzzcomputing.com","username":"zzzeek","status":"Red Hat"},"state":"REVIEWER"},{"updated":"2021-08-10 13:04:10.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"}],"messages":[{"id":"a3d8be26e748f6e317753f6b785ee1b8f7d8c298","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2021-07-27 17:30:19.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"2eefb418b8ae35bea67db464b602c338cdbab191","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-27 18:47:35.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/a36ede56d6b9499c85c95d277b6f3030 : SUCCESS in 1h 15m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/29cd5b3548604b2082537922f9deeba0 : SUCCESS in 4m 49s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3d84f4d6148a4cbcaa37a8676062b67f : SUCCESS in 5m 14s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/fcf923b7b70d4c2aa5e1bd7a5e32ab9b : SUCCESS in 7m 21s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/466623bbb84143f8b4b660357b3c39cd : SUCCESS in 8m 44s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2d95d538f75542bfb1be988caf422f7d : SUCCESS in 7m 55s","accounts_in_message":[],"_revision_number":1},{"id":"fedd2d3fa0d130af199f7cd9bbe4de7388288351","author":{"_account_id":11816,"name":"mike bayer","display_name":"Mike Bayer","email":"mike_mp@zzzcomputing.com","username":"zzzeek","status":"Red Hat"},"date":"2021-07-27 19:07:17.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)\n\nLFTM, havent followed this stuff in a long time","accounts_in_message":[],"_revision_number":1},{"id":"2bec3ed60fa189bef9ddbf5f116140eaaf367a73","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2021-07-28 09:05:55.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"20406f7b397ea68136ae419bad50c104ace5309f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2021-07-29 10:44:29.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.","accounts_in_message":[],"_revision_number":2},{"id":"84dda15c36028b174a3519e136ce97b468c21d4a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-29 12:12:09.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/f06bea2152e64fd7b556eb2e520e9dc1 : SUCCESS in 1h 14m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/64c5f8ac5ce44dbdbc53fb3567423478 : SUCCESS in 5m 01s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e09d9b3cd63b489e9b03b4ea4e511c5e : SUCCESS in 6m 17s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/87759fad99034c10983bdd8cb3de5e59 : SUCCESS in 6m 13s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/49485544cbe649b2945aea73a7d53d01 : SUCCESS in 7m 32s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fe676bac44fc42bd8f5c710a3d5701de : SUCCESS in 6m 13s","accounts_in_message":[],"_revision_number":2},{"id":"262d05781c1c951055f5ddde34d862ceb180d418","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2021-07-29 15:37:53.000000000","message":"Uploaded patch set 3: Patch Set 2 was rebased.","accounts_in_message":[],"_revision_number":3},{"id":"9de68be469790f4385e9a3988833a4c1a0e44265","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-29 17:44:54.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/5b08127d4eba4656821ff4974922ef2e : TIMED_OUT in 2h 05m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8f709b0830c444a48ddee74f6a912b28 : SUCCESS in 5m 17s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/02bcf18487644628a2806c69824bfd6a : SUCCESS in 6m 11s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/cca85f7db7b7493e837d8ee970c4928c : SUCCESS in 6m 34s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/606376f57344415f9de6756bd368f0e6 : SUCCESS in 7m 07s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a4760ad537ce45fbae4e545d60b823e2 : SUCCESS in 7m 12s","accounts_in_message":[],"_revision_number":3},{"id":"c293ec7818d80464f1ce8e6d3c69aa4c6e83ca2d","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2021-08-10 10:20:33.000000000","message":"Patch Set 3:\n\nrecheck timeout","accounts_in_message":[],"_revision_number":3},{"id":"2fbac8199626649fb2a1db10cbef42025c9a7e54","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-10 11:54:38.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/3b4ab01f39374660ad04f580a06b8e1c : SUCCESS in 1h 29m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/867f148242e04201bbd3b00253cc316a : SUCCESS in 5m 13s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5f8bddca48144495855de8468607a476 : SUCCESS in 5m 57s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/79e3bb88212b4c31a59ed11d20850f4a : SUCCESS in 6m 09s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/15bb8559e8124eb381ca60abfda1de52 : SUCCESS in 5m 10s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/07ce817e163e4872aa2b852b810e7f83 : SUCCESS in 6m 13s","accounts_in_message":[],"_revision_number":3},{"id":"0665a7b75a6d392ae3e84246f29bfd5af3b7ecd5","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-08-10 13:04:10.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"c95d525a30247673b16a208d9aa762f8dab23177","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-10 13:04:39.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"e3ae5295663edc3d636b705967f635234a531416","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-10 14:28:12.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/57f3918c6c0b4cdda02b5b10fddd1374 : SUCCESS in 1h 20m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6317c6e2aa3c4932924225f5d2d0cadc : SUCCESS in 4m 40s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4f4cfd1bb503450ebe65d5b3bcc9a584 : SUCCESS in 5m 12s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f845b64c71e34246ae960604aba44cdf : SUCCESS in 4m 56s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6decb8b4d05644bca407a860cbd262be : SUCCESS in 5m 09s","accounts_in_message":[],"_revision_number":3},{"id":"369dd1d5bb983a3fb4cbe71a1f9599ccb8902ecc","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-10 14:28:12.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3},{"id":"3923f8cdf9bcdec76c90c4e8b5ea391e1879eac9","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-10 14:29:37.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dfb6f3e5d5ac4aacb4dbf64766a5eef3 : SUCCESS in 1m 01s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"73e376d2fd5c96a9a3ddd1aea43197f82dd767c2","revisions":{"16e321d44517443327ab4dd3cf16542c0ed8ecb3":{"kind":"REWORK","_number":1,"created":"2021-07-27 17:30:19.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/63/802563/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.db","ref":"refs/changes/63/802563/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.db refs/changes/63/802563/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.db refs/changes/63/802563/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.db refs/changes/63/802563/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.db refs/changes/63/802563/1"}}},"commit":{"parents":[{"commit":"f048068216c655b6babc5740fc4b5e7efcf02ab1","subject":"tox: Simplify test running","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.db/commit/f048068216c655b6babc5740fc4b5e7efcf02ab1"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2021-07-16 11:27:07.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2021-07-16 12:01:45.000000000","tz":60},"subject":"tests: Clean up base test","message":"tests: Clean up base test\n\nThis base test isn\u0027t actually used by anything but we\u0027re going to change\nthat shortly. Before we do, address a longstanding TODO and make it\nsubclass \u0027oslotest.BaseTestCase\u0027. Because \u0027oslotest.BaseTestCase\u0027\nalready configures all of the fixtures we\u0027re currently using, our own\nbase test class ends up being a simple wrapper. This will also change in\nthe future.\n\nChange-Id: I110392586fb032507d122ae9a412a6bba7c5ff2d\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.db/commit/16e321d44517443327ab4dd3cf16542c0ed8ecb3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.db/commit/16e321d44517443327ab4dd3cf16542c0ed8ecb3"}]},"branch":"refs/heads/master"},"e0042db52be3591c39ac05973e8fb677527018b8":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2021-07-29 10:44:29.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/63/802563/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.db","ref":"refs/changes/63/802563/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.db refs/changes/63/802563/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.db refs/changes/63/802563/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.db refs/changes/63/802563/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.db refs/changes/63/802563/2"}}},"commit":{"parents":[{"commit":"a3ee5b99426f9da5fac2efc399968ee5f865a049","subject":"Merge \"tests: Remove \u0027ModelsMigrationsSync.check_foreign_keys\u0027\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.db/commit/a3ee5b99426f9da5fac2efc399968ee5f865a049"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2021-07-16 11:27:07.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2021-07-29 09:54:51.000000000","tz":60},"subject":"tests: Clean up base test","message":"tests: Clean up base test\n\nThis base test isn\u0027t actually used by anything but we\u0027re going to change\nthat shortly. Before we do, address a longstanding TODO and make it\nsubclass \u0027oslotest.BaseTestCase\u0027. Because \u0027oslotest.BaseTestCase\u0027\nalready configures all of the fixtures we\u0027re currently using, our own\nbase test class ends up being a simple wrapper. This will also change in\nthe future.\n\nChange-Id: I110392586fb032507d122ae9a412a6bba7c5ff2d\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.db/commit/e0042db52be3591c39ac05973e8fb677527018b8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.db/commit/e0042db52be3591c39ac05973e8fb677527018b8"}]},"branch":"refs/heads/master"},"73e376d2fd5c96a9a3ddd1aea43197f82dd767c2":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2021-07-29 15:37:53.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/63/802563/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.db","ref":"refs/changes/63/802563/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.db refs/changes/63/802563/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.db refs/changes/63/802563/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.db refs/changes/63/802563/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.db refs/changes/63/802563/3"}}},"commit":{"parents":[{"commit":"1b48a34bd11f40fc516995ed693d193f28c80b25","subject":"Merge \"Drop checks for IBM DB2\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.db/commit/1b48a34bd11f40fc516995ed693d193f28c80b25"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2021-07-16 11:27:07.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2021-07-29 15:36:38.000000000","tz":60},"subject":"tests: Clean up base test","message":"tests: Clean up base test\n\nThis base test isn\u0027t actually used by anything but we\u0027re going to change\nthat shortly. Before we do, address a longstanding TODO and make it\nsubclass \u0027oslotest.BaseTestCase\u0027. Because \u0027oslotest.BaseTestCase\u0027\nalready configures all of the fixtures we\u0027re currently using, our own\nbase test class ends up being a simple wrapper. This will also change in\nthe future.\n\nChange-Id: I110392586fb032507d122ae9a412a6bba7c5ff2d\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.db/commit/73e376d2fd5c96a9a3ddd1aea43197f82dd767c2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.db/commit/73e376d2fd5c96a9a3ddd1aea43197f82dd767c2"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":11816,"name":"mike bayer","display_name":"Mike Bayer","email":"mike_mp@zzzcomputing.com","username":"zzzeek","status":"Red Hat"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
