)]}'
{"id":"openstack%2Foslo.db~802564","triplet_id":"openstack%2Foslo.db~master~I6dae26aeb4b2a5cc4f39ed637807a27e44d40025","project":"openstack/oslo.db","branch":"master","topic":"sqlalchemy-20","hashtags":[],"change_id":"I6dae26aeb4b2a5cc4f39ed637807a27e44d40025","subject":"tests: Enfeeble \u0027oslo_db.tests.utils.BaseTestCase\u0027","status":"MERGED","created":"2021-07-27 17:30:19.000000000","updated":"2021-08-10 14:40:20.000000000","submitted":"2021-08-10 14:38:51.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"802564-sqlalchemy-20","meta_rev_id":"c02604336c4d2b963a9197dc7c0e8e1dea253a8f","_number":802564,"virtual_id_number":802564,"owner":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11816,"name":"mike bayer","display_name":"Mike Bayer","email":"mike_mp@zzzcomputing.com","username":"zzzeek","status":"Red Hat"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-08-10 14:38:50.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11816,"name":"mike bayer","display_name":"Mike Bayer","email":"mike_mp@zzzcomputing.com","username":"zzzeek","status":"Red Hat"},"all":[{"value":2,"date":"2021-08-09 12:38:45.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11816,"name":"mike bayer","display_name":"Mike Bayer","email":"mike_mp@zzzcomputing.com","username":"zzzeek","status":"Red Hat"},{"value":2,"date":"2021-08-10 08:31:09.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"all":[{"value":0,"_account_id":11816,"name":"mike bayer","display_name":"Mike Bayer","email":"mike_mp@zzzcomputing.com","username":"zzzeek","status":"Red Hat"},{"value":1,"date":"2021-08-10 08:31:09.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":11816,"name":"mike bayer","display_name":"Mike Bayer","email":"mike_mp@zzzcomputing.com","username":"zzzeek","status":"Red Hat"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11816,"name":"mike bayer","display_name":"Mike Bayer","email":"mike_mp@zzzcomputing.com","username":"zzzeek","status":"Red Hat"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-07-27 18:50:48.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-08-09 12:38:45.000000000","updated_by":{"_account_id":11816,"name":"mike bayer","display_name":"Mike Bayer","email":"mike_mp@zzzcomputing.com","username":"zzzeek","status":"Red Hat"},"reviewer":{"_account_id":11816,"name":"mike bayer","display_name":"Mike Bayer","email":"mike_mp@zzzcomputing.com","username":"zzzeek","status":"Red Hat"},"state":"REVIEWER"},{"updated":"2021-08-10 08:31:09.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"}],"messages":[{"id":"06592687260f1834b247334f561e8e7366cc42a0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2021-07-27 17:30:19.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"0653b045736d0f56b23fe821caef9697bc50ca2c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-27 18:50:48.000000000","message":"Patch Set 1: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/d6c0a43ff0a2411d816b816463c4ffc3 : SUCCESS in 1h 13m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6dad546d5be74e53b189dc8b78ae105e : FAILURE in 4m 58s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2735c6e4388b4d8283094da1ee54a66b : SUCCESS in 6m 45s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/371371113322423bb438b2c02cbb92b4 : SUCCESS in 6m 57s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/909361c90d264a1788e357bc10c7bdc8 : SUCCESS in 5m 26s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/73ef7f5a4e3f4c0eb1e066f9927475fa : SUCCESS in 6m 02s","accounts_in_message":[],"_revision_number":1},{"id":"250ba15797955d5d00cb9c94abcc89bfe79e0c43","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2021-07-28 09:07:17.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"af78d6924add9f35f3a7d2b805e9a223fb6eea74","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-28 10:53:55.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/4e632755c0f9468898dc9e4ce389e09e : SUCCESS in 1h 40m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c6ef744d39e4423e9abb306481350bf9 : SUCCESS in 6m 07s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ecb277a50f9e4dfb88d542e4cee73233 : SUCCESS in 5m 44s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f90b8b1ec3fe4c53983e4d30c02a2c2b : SUCCESS in 5m 40s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/64f5c17092074e20ab0043f5ef53f087 : SUCCESS in 6m 19s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/35bc64d3aed54d1aa7fed338390f9c6b : SUCCESS in 5m 23s","accounts_in_message":[],"_revision_number":2},{"id":"b9f0538de2cac28b605d92f0856a1c4b9b2d2498","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2021-07-28 14:38:18.000000000","message":"Uploaded patch set 3: New patch set was added with same tree, parent, and commit message as Patch Set 2.","accounts_in_message":[],"_revision_number":3},{"id":"42e6033eae086fe0f54acfa96686d2766fc74074","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-28 16:04:19.000000000","message":"Patch Set 3:\n\nBuild succeeded (check pipeline).\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/a095cf9558df44408e57fa7ee9c9342f : SUCCESS in 1h 20m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2edd5b72e6e949fab6e33b3478adc228 : SUCCESS in 5m 31s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/123c198de92b4d508064d37acc8afa81 : SUCCESS in 7m 36s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/eee7693ff8d145159121822ff6145f81 : SUCCESS in 6m 04s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/540dc0337eac47a98eb3470504c56e36 : SUCCESS in 7m 14s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/00777d2a03ed4822b024656186507a7d : SUCCESS in 7m 11s","accounts_in_message":[],"_revision_number":3},{"id":"80672b540f2c73d2f51b303494d8b23eb749663d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2021-07-29 10:44:29.000000000","message":"Uploaded patch set 4: Patch Set 3 was rebased.","accounts_in_message":[],"_revision_number":4},{"id":"c1179c3c05558cfaea444f639394443d658e24e3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-29 12:14:39.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/73884c63bfd84f08bc462f8abd8ac3a8 : SUCCESS in 1h 16m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/55259ca30511442688ee1ba942b9bdb4 : SUCCESS in 6m 05s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/36fe333384734391815f31c2390d1ee9 : SUCCESS in 6m 33s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8b6d4f60a7634b37920590816db16df1 : SUCCESS in 5m 56s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/22ed0e20dd754a78a5ab609333058123 : SUCCESS in 6m 27s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d73c2125acae4311892c76eb2f13a435 : SUCCESS in 6m 07s","accounts_in_message":[],"_revision_number":4},{"id":"2f01993b45720e7ebefb1020fac90f04ddd565a0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2021-07-29 15:37:53.000000000","message":"Uploaded patch set 5: Patch Set 4 was rebased.","accounts_in_message":[],"_revision_number":5},{"id":"a1278a7b8a4a287f9c807a93f0614304ff8626a0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-29 16:57:15.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/e9ecae4d9a014fa89c54179410e7ba78 : SUCCESS in 1h 17m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/35722a86bf0f48cd96c5e95d1c785da3 : SUCCESS in 7m 08s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9e7348cb80244499a0fc67969bcf3be0 : SUCCESS in 5m 12s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/710b4bf7ab55444d9c35191d7531011b : SUCCESS in 6m 27s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/8c791f95f54c4cfaa4435727de470d18 : SUCCESS in 5m 55s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/00e171a3e7aa40d991978d6734e6aff3 : SUCCESS in 6m 35s","accounts_in_message":[],"_revision_number":5},{"id":"ceb63ddf968a095fac8fbea22835a7bfe9565d66","author":{"_account_id":11816,"name":"mike bayer","display_name":"Mike Bayer","email":"mike_mp@zzzcomputing.com","username":"zzzeek","status":"Red Hat"},"date":"2021-08-09 12:38:45.000000000","message":"Patch Set 5: Code-Review+2","accounts_in_message":[],"_revision_number":5},{"id":"8121d3f78361b2da2dc0de0c32212b7d92e99284","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-08-10 08:31:09.000000000","message":"Patch Set 5: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":5},{"id":"aaafa355415a03cf995a6b6858f8b1b10be3449b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-10 13:04:42.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"69fa3e855655792548a2a9ad80513789b830abdb","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-10 14:38:50.000000000","message":"Patch Set 5: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/47214061e50d4233b0b47887f9d9b46d : SUCCESS in 1h 30m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b96b6bf3b3234b1f8b8a7c1adcd919eb : SUCCESS in 4m 53s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c6cadbd6c2764ba8a698e1e404e5da8a : SUCCESS in 5m 00s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/33a9be0fd2b747669f830e0a06b0c0a0 : SUCCESS in 4m 54s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b7af207339814fcd985f94bbecb13a82 : SUCCESS in 5m 07s","accounts_in_message":[],"_revision_number":5},{"id":"6b413ce2b3e6af4bb40c8c92fb830f533c79c0d5","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-10 14:38:51.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":5},{"id":"c02604336c4d2b963a9197dc7c0e8e1dea253a8f","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-10 14:40:20.000000000","message":"Patch Set 5:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/787a387066c94926ab749658a113e118 : SUCCESS in 1m 01s","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"538e0a23abd8dfb7a2fe4a47655e759a50138b13","revisions":{"f3a8cdfe81f3666ac2ed012b276cc2cdb3b47ab2":{"kind":"REWORK","_number":1,"created":"2021-07-27 17:30:19.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/64/802564/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.db","ref":"refs/changes/64/802564/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.db refs/changes/64/802564/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.db refs/changes/64/802564/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.db refs/changes/64/802564/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.db refs/changes/64/802564/1"}}},"commit":{"parents":[{"commit":"16e321d44517443327ab4dd3cf16542c0ed8ecb3","subject":"tests: Clean up base test","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.db/commit/16e321d44517443327ab4dd3cf16542c0ed8ecb3"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2021-07-16 11:37:24.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2021-07-16 12:01:45.000000000","tz":60},"subject":"tests: Enfeeble \u0027oslo_db.tests.utils.BaseTestCase\u0027","message":"tests: Enfeeble \u0027oslo_db.tests.utils.BaseTestCase\u0027\n\nWe have quite a few base test classes in use. Start fixing these by\nremoving the logic from one of these. We will remove the class entirely\nin a future change.\n\nChange-Id: I6dae26aeb4b2a5cc4f39ed637807a27e44d40025\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.db/commit/f3a8cdfe81f3666ac2ed012b276cc2cdb3b47ab2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.db/commit/f3a8cdfe81f3666ac2ed012b276cc2cdb3b47ab2"}]},"branch":"refs/heads/master"},"729b0f821b547e546c9576486383c192e66196f0":{"kind":"REWORK","_number":2,"created":"2021-07-28 09:07:17.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/64/802564/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.db","ref":"refs/changes/64/802564/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.db refs/changes/64/802564/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.db refs/changes/64/802564/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.db refs/changes/64/802564/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.db refs/changes/64/802564/2"}}},"commit":{"parents":[{"commit":"16e321d44517443327ab4dd3cf16542c0ed8ecb3","subject":"tests: Clean up base test","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.db/commit/16e321d44517443327ab4dd3cf16542c0ed8ecb3"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2021-07-16 11:37:24.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2021-07-28 09:07:08.000000000","tz":0},"subject":"tests: Enfeeble \u0027oslo_db.tests.utils.BaseTestCase\u0027","message":"tests: Enfeeble \u0027oslo_db.tests.utils.BaseTestCase\u0027\n\nWe have quite a few base test classes in use. Start fixing these by\nremoving the logic from one of these. We will remove the class entirely\nin a future change.\n\nChange-Id: I6dae26aeb4b2a5cc4f39ed637807a27e44d40025\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.db/commit/729b0f821b547e546c9576486383c192e66196f0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.db/commit/729b0f821b547e546c9576486383c192e66196f0"}]},"branch":"refs/heads/master"},"8f22c9d564b903696814f092f67be702dc580c02":{"kind":"NO_CHANGE","_number":3,"created":"2021-07-28 14:38:18.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/64/802564/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.db","ref":"refs/changes/64/802564/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.db refs/changes/64/802564/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.db refs/changes/64/802564/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.db refs/changes/64/802564/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.db refs/changes/64/802564/3"}}},"commit":{"parents":[{"commit":"16e321d44517443327ab4dd3cf16542c0ed8ecb3","subject":"tests: Clean up base test","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.db/commit/16e321d44517443327ab4dd3cf16542c0ed8ecb3"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2021-07-16 11:37:24.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2021-07-28 13:52:42.000000000","tz":60},"subject":"tests: Enfeeble \u0027oslo_db.tests.utils.BaseTestCase\u0027","message":"tests: Enfeeble \u0027oslo_db.tests.utils.BaseTestCase\u0027\n\nWe have quite a few base test classes in use. Start fixing these by\nremoving the logic from one of these. We will remove the class entirely\nin a future change.\n\nChange-Id: I6dae26aeb4b2a5cc4f39ed637807a27e44d40025\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.db/commit/8f22c9d564b903696814f092f67be702dc580c02"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.db/commit/8f22c9d564b903696814f092f67be702dc580c02"}]},"branch":"refs/heads/master"},"520661251540589e2ddf0f6bc5652899b985a804":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2021-07-29 10:44:29.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/64/802564/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.db","ref":"refs/changes/64/802564/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.db refs/changes/64/802564/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.db refs/changes/64/802564/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.db refs/changes/64/802564/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.db refs/changes/64/802564/4"}}},"commit":{"parents":[{"commit":"e0042db52be3591c39ac05973e8fb677527018b8","subject":"tests: Clean up base test","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.db/commit/e0042db52be3591c39ac05973e8fb677527018b8"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2021-07-16 11:37:24.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2021-07-29 09:54:51.000000000","tz":60},"subject":"tests: Enfeeble \u0027oslo_db.tests.utils.BaseTestCase\u0027","message":"tests: Enfeeble \u0027oslo_db.tests.utils.BaseTestCase\u0027\n\nWe have quite a few base test classes in use. Start fixing these by\nremoving the logic from one of these. We will remove the class entirely\nin a future change.\n\nChange-Id: I6dae26aeb4b2a5cc4f39ed637807a27e44d40025\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.db/commit/520661251540589e2ddf0f6bc5652899b985a804"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.db/commit/520661251540589e2ddf0f6bc5652899b985a804"}]},"branch":"refs/heads/master"},"538e0a23abd8dfb7a2fe4a47655e759a50138b13":{"kind":"TRIVIAL_REBASE","_number":5,"created":"2021-07-29 15:37:53.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/64/802564/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.db","ref":"refs/changes/64/802564/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.db refs/changes/64/802564/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.db refs/changes/64/802564/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.db refs/changes/64/802564/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.db refs/changes/64/802564/5"}}},"commit":{"parents":[{"commit":"73e376d2fd5c96a9a3ddd1aea43197f82dd767c2","subject":"tests: Clean up base test","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.db/commit/73e376d2fd5c96a9a3ddd1aea43197f82dd767c2"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2021-07-16 11:37:24.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2021-07-29 15:36:38.000000000","tz":60},"subject":"tests: Enfeeble \u0027oslo_db.tests.utils.BaseTestCase\u0027","message":"tests: Enfeeble \u0027oslo_db.tests.utils.BaseTestCase\u0027\n\nWe have quite a few base test classes in use. Start fixing these by\nremoving the logic from one of these. We will remove the class entirely\nin a future change.\n\nChange-Id: I6dae26aeb4b2a5cc4f39ed637807a27e44d40025\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.db/commit/538e0a23abd8dfb7a2fe4a47655e759a50138b13"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.db/commit/538e0a23abd8dfb7a2fe4a47655e759a50138b13"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
