)]}'
{"id":"openstack%2Foslo.db~802566","triplet_id":"openstack%2Foslo.db~master~I34b395e6d50f4e4151178c327d94308e6f5d5b6e","project":"openstack/oslo.db","branch":"master","topic":"sqlalchemy-20","hashtags":[],"change_id":"I34b395e6d50f4e4151178c327d94308e6f5d5b6e","subject":"tests: Enable SADeprecationWarning warnings","status":"MERGED","created":"2021-07-27 17:30:19.000000000","updated":"2021-08-10 14:38:54.000000000","submitted":"2021-08-10 14:38:54.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"802566-sqlalchemy-20","meta_rev_id":"a0f03b86431f1a8f51481b9f1a2eb02fa2d370a8","_number":802566,"virtual_id_number":802566,"owner":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11816,"name":"mike bayer","display_name":"Mike Bayer","email":"mike_mp@zzzcomputing.com","username":"zzzeek","status":"Red Hat"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-08-10 14:38:53.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11816,"name":"mike bayer","display_name":"Mike Bayer","email":"mike_mp@zzzcomputing.com","username":"zzzeek","status":"Red Hat"},"all":[{"value":2,"date":"2021-08-09 12:45:02.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11816,"name":"mike bayer","display_name":"Mike Bayer","email":"mike_mp@zzzcomputing.com","username":"zzzeek","status":"Red Hat"},{"value":2,"date":"2021-08-10 09:07:27.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"all":[{"value":0,"_account_id":11816,"name":"mike bayer","display_name":"Mike Bayer","email":"mike_mp@zzzcomputing.com","username":"zzzeek","status":"Red Hat"},{"value":1,"date":"2021-08-10 12:31:32.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":11816,"name":"mike bayer","display_name":"Mike Bayer","email":"mike_mp@zzzcomputing.com","username":"zzzeek","status":"Red Hat"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11816,"name":"mike bayer","display_name":"Mike Bayer","email":"mike_mp@zzzcomputing.com","username":"zzzeek","status":"Red Hat"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-07-27 18:56:30.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-08-09 12:45:02.000000000","updated_by":{"_account_id":11816,"name":"mike bayer","display_name":"Mike Bayer","email":"mike_mp@zzzcomputing.com","username":"zzzeek","status":"Red Hat"},"reviewer":{"_account_id":11816,"name":"mike bayer","display_name":"Mike Bayer","email":"mike_mp@zzzcomputing.com","username":"zzzeek","status":"Red Hat"},"state":"REVIEWER"},{"updated":"2021-08-10 09:07:27.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"}],"messages":[{"id":"22ea0c347c503203801f29d327ce9b46c64be612","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2021-07-27 17:30:19.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"bacbb2bcef9fa97d49fd1e026ab3ed0617132925","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-27 18:56:30.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/75f55f1dfb9d4a1bab0ff4dc34aeea92 : SUCCESS in 1h 19m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3bf1a19adc964b419c225b391c76c8a5 : SUCCESS in 5m 20s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1fbc16f196864fcaafeaca9b63f771da : SUCCESS in 5m 57s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/dbf9032906fc4e70bb17a71ab698dea8 : SUCCESS in 5m 35s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/da2189e676824a279bd81f0107952fd9 : SUCCESS in 5m 52s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7c78beebb4aa423d86e1250b284abe86 : SUCCESS in 5m 24s","accounts_in_message":[],"_revision_number":1},{"id":"420aef9f29f76f551ffc4bbc93ca8c6f3adeea4c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2021-07-28 14:38:18.000000000","message":"Uploaded patch set 2: New patch set was added with same tree, parent, and commit message as Patch Set 1.","accounts_in_message":[],"_revision_number":2},{"id":"ca84a5ec1a6755204f7bf5380ca5c282aecea08c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-28 15:58:20.000000000","message":"Patch Set 2:\n\nBuild succeeded (check pipeline).\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/d4d7dabcd80f465cbddaa0965fcd6020 : SUCCESS in 1h 17m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/18c8e40b86d74e92a6b12f81750df121 : SUCCESS in 5m 33s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9aa69290a5d742f5a232957495476fbf : SUCCESS in 5m 22s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ee41bcb9638e499d902e1f440d6f88bf : SUCCESS in 6m 08s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/cf787caea92840afb11d459089dbf562 : SUCCESS in 7m 34s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fea3acb06da84414934d1c974fa9f06a : SUCCESS in 8m 45s","accounts_in_message":[],"_revision_number":2},{"id":"781569b562129adb2424a429eaf4fe6ebe4be77b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2021-07-29 10:44:29.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"ce410d644776fd78fc5657ad622b825b76da78fe","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-29 12:23:21.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/f2d63bf1b7034f70b61ded66727e5bb8 : SUCCESS in 1h 32m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/27a80d704c0840958719f9b91aad3836 : SUCCESS in 4m 49s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2a7c2fef2ed74110827d80fecc825303 : SUCCESS in 5m 23s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e758e2c9505e45adb181bc53d7aff09a : SUCCESS in 5m 22s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/55e0bffb95804719a8128eb9042cd3fc : SUCCESS in 5m 44s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d5124feacd9346718f4ac0e9b9d2725b : SUCCESS in 5m 48s","accounts_in_message":[],"_revision_number":3},{"id":"a972349e769f4faa05ccbb3fc3a07c0f3c7b3f69","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2021-07-29 15:37:53.000000000","message":"Uploaded patch set 4: Patch Set 3 was rebased.","accounts_in_message":[],"_revision_number":4},{"id":"da4eb297933131c1d5788ce78cbc6f979eab335a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-29 17:00:50.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/165dccf9b5064597aa5953012ef40cf6 : SUCCESS in 1h 20m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/432fa17bfe3741139285b9852b4909c7 : SUCCESS in 6m 09s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8547b38a644a491e9c1b0e61718db08f : SUCCESS in 7m 19s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/cd491ac705c14133b247758438501370 : SUCCESS in 6m 32s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d8a1bbfb3a0b45eeac94d4ad62993c75 : SUCCESS in 6m 45s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a5d67fabe9094938b76e9d2bdf85500c : SUCCESS in 6m 52s","accounts_in_message":[],"_revision_number":4},{"id":"cfb5a159aa835cc8b2971815ae225f19445a0546","author":{"_account_id":11816,"name":"mike bayer","display_name":"Mike Bayer","email":"mike_mp@zzzcomputing.com","username":"zzzeek","status":"Red Hat"},"date":"2021-08-09 12:45:02.000000000","message":"Patch Set 4: Code-Review+2\n\n(1 comment)\n\nfor that _copy() stuff, we should do a sweep to see where those really old oslo.db functions are even being used now, I would assume nova used them at some point.    if we need to keep them longer term we should vendor the _copy() functions locally.","accounts_in_message":[],"_revision_number":4},{"id":"01bf2cbf9cd157061549a3d0699f131d799ffff7","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-08-10 09:07:27.000000000","message":"Patch Set 4: Code-Review+2\n\n\u003e Patch Set 4: Code-Review+2\n\u003e \n\u003e (1 comment)\n\u003e \n\u003e for that _copy() stuff, we should do a sweep to see where those really old oslo.db functions are even being used now, I would assume nova used them at some point.    if we need to keep them longer term we should vendor the _copy() functions locally.\n\nI agree with Mike.\n\nAt first glance the functions that use those `_copy()` are only used internally and not called outside oslo.db [1], even within Nova.\n\n[1] https://codesearch.opendev.org/?q\u003dchange_deleted_column_type_to_id_type\u0026i\u003dnope\u0026literal\u003dnope\u0026files\u003d\u0026excludeFiles\u003d\u0026repos\u003d","accounts_in_message":[],"_revision_number":4},{"id":"9ec872111853487ae6fefbe11d4d48e5ec7d2965","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2021-08-10 10:21:52.000000000","message":"Patch Set 4:\n\n\u003e Patch Set 4: Code-Review+2\n\u003e \n\u003e \u003e Patch Set 4: Code-Review+2\n\u003e \u003e \n\u003e \u003e (1 comment)\n\u003e \u003e \n\u003e \u003e for that _copy() stuff, we should do a sweep to see where those really old oslo.db functions are even being used now, I would assume nova used them at some point.    if we need to keep them longer term we should vendor the _copy() functions locally.\n\u003e \n\u003e I agree with Mike.\n\u003e \n\u003e At first glance the functions that use those `_copy()` are only used internally and not called outside oslo.db [1], even within Nova.\n\u003e \n\u003e [1] https://codesearch.opendev.org/?q\u003dchange_deleted_column_type_to_id_type\u0026i\u003dnope\u0026literal\u003dnope\u0026files\u003d\u0026excludeFiles\u003d\u0026repos\u003d\n\nYes, agreed that these can probably go. I plan to remove them like I\u0027ve removed so much else, however the series was already large enough without adding more patches. Just trust that I\u0027m generally on top of these things 😄","accounts_in_message":[],"_revision_number":4},{"id":"a483a486823638cc6037f607c8867402cf4d991b","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-08-10 12:31:32.000000000","message":"Patch Set 4: Workflow+1\n\n\u003e Patch Set 4:\n\u003e \n\u003e \u003e Patch Set 4: Code-Review+2\n\u003e \u003e \n\u003e \u003e \u003e Patch Set 4: Code-Review+2\n\u003e \u003e \u003e \n\u003e \u003e \u003e (1 comment)\n\u003e \u003e \u003e \n\u003e \u003e \u003e for that _copy() stuff, we should do a sweep to see where those really old oslo.db functions are even being used now, I would assume nova used them at some point.    if we need to keep them longer term we should vendor the _copy() functions locally.\n\u003e \u003e \n\u003e \u003e I agree with Mike.\n\u003e \u003e \n\u003e \u003e At first glance the functions that use those `_copy()` are only used internally and not called outside oslo.db [1], even within Nova.\n\u003e \u003e \n\u003e \u003e [1] https://codesearch.opendev.org/?q\u003dchange_deleted_column_type_to_id_type\u0026i\u003dnope\u0026literal\u003dnope\u0026files\u003d\u0026excludeFiles\u003d\u0026repos\u003d\n\u003e \n\u003e Yes, agreed that these can probably go. I plan to remove them like I\u0027ve removed so much else, however the series was already large enough without adding more patches. Just trust that I\u0027m generally on top of these things 😄\n\nOk, then let\u0027s go with that","accounts_in_message":[],"_revision_number":4},{"id":"d0c37d0c34010e4f71422ec666f9ce6d39049852","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-10 13:04:46.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"9b0426466d7337230023075c683ee48cbbf99385","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-10 14:38:53.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/10bab8de8df64ed4bb8ce6246bf6d9ea : SUCCESS in 1h 24m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bce8e62fe6c84635b1599df1b8b847f7 : SUCCESS in 5m 53s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/07a280154e444ebea25df9be7c073e0d : SUCCESS in 5m 23s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2ff2d2192f9f427c99c3ad8bf430f677 : SUCCESS in 6m 16s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a5b9acbcc8d94660b2259309d2c5e84e : SUCCESS in 5m 50s","accounts_in_message":[],"_revision_number":4},{"id":"a0f03b86431f1a8f51481b9f1a2eb02fa2d370a8","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-10 14:38:54.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"40bce5a2baf75dc87dd591e0f71a00f221a2ba92","revisions":{"63537971b75323db85923228615752fe45067efc":{"kind":"REWORK","_number":1,"created":"2021-07-27 17:30:19.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/66/802566/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.db","ref":"refs/changes/66/802566/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.db refs/changes/66/802566/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.db refs/changes/66/802566/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.db refs/changes/66/802566/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.db refs/changes/66/802566/1"}}},"commit":{"parents":[{"commit":"8566910a70a0761050b34a0b1ecbc1a186874601","subject":"tests: Use common base class","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.db/commit/8566910a70a0761050b34a0b1ecbc1a186874601"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2021-07-16 11:04:14.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2021-07-16 16:52:55.000000000","tz":60},"subject":"tests: Enable SADeprecationWarning warnings","message":"tests: Enable SADeprecationWarning warnings\n\nHighlight use of deprecated SQLAlchemy APIs to ensure we keep on top of\nthings. This requires resolving the following SADeprecationWarning\nwarnings:\n\n  The from_engine() method on Inspector is deprecated and will be\n  removed in a future release.  Please use the sqlalchemy.inspect()\n  function on an Engine or Connection in order to acquire an Inspector.\n\n  The Column.copy() method is deprecated and will be removed in a future\n  release.\n\n  The ColumnCollectionConstraint.copy() method is deprecated and will be\n  removed in a future release.\n\nThere are more warnings to be resolved related to SQLAlchemy 2.0, but\nthose require a special environment option (SQLALCHEMY_WARN_20) to\ntrigger and a lot of work to resolve. We\u0027ll address those in a series of\nfollow-ups.\n\nChange-Id: I34b395e6d50f4e4151178c327d94308e6f5d5b6e\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.db/commit/63537971b75323db85923228615752fe45067efc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.db/commit/63537971b75323db85923228615752fe45067efc"}]},"branch":"refs/heads/master"},"e29d0e28d8a56e2762887da7d80c1caa6278ac28":{"kind":"NO_CHANGE","_number":2,"created":"2021-07-28 14:38:18.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/66/802566/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.db","ref":"refs/changes/66/802566/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.db refs/changes/66/802566/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.db refs/changes/66/802566/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.db refs/changes/66/802566/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.db refs/changes/66/802566/2"}}},"commit":{"parents":[{"commit":"abb249cf7837b1f4b382d7a0a4d6dd96b4979d61","subject":"tests: Use common base class","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.db/commit/abb249cf7837b1f4b382d7a0a4d6dd96b4979d61"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2021-07-16 11:04:14.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2021-07-28 13:52:57.000000000","tz":60},"subject":"tests: Enable SADeprecationWarning warnings","message":"tests: Enable SADeprecationWarning warnings\n\nHighlight use of deprecated SQLAlchemy APIs to ensure we keep on top of\nthings. This requires resolving the following SADeprecationWarning\nwarnings:\n\n  The from_engine() method on Inspector is deprecated and will be\n  removed in a future release.  Please use the sqlalchemy.inspect()\n  function on an Engine or Connection in order to acquire an Inspector.\n\n  The Column.copy() method is deprecated and will be removed in a future\n  release.\n\n  The ColumnCollectionConstraint.copy() method is deprecated and will be\n  removed in a future release.\n\nThere are more warnings to be resolved related to SQLAlchemy 2.0, but\nthose require a special environment option (SQLALCHEMY_WARN_20) to\ntrigger and a lot of work to resolve. We\u0027ll address those in a series of\nfollow-ups.\n\nChange-Id: I34b395e6d50f4e4151178c327d94308e6f5d5b6e\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.db/commit/e29d0e28d8a56e2762887da7d80c1caa6278ac28"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.db/commit/e29d0e28d8a56e2762887da7d80c1caa6278ac28"}]},"branch":"refs/heads/master"},"6232f120d4ceda7ee7fdb7aee59a5db53c51267b":{"kind":"REWORK","_number":3,"created":"2021-07-29 10:44:29.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/66/802566/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.db","ref":"refs/changes/66/802566/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.db refs/changes/66/802566/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.db refs/changes/66/802566/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.db refs/changes/66/802566/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.db refs/changes/66/802566/3"}}},"commit":{"parents":[{"commit":"ccbeb04c0a7e86c92b71ea1f71bacbbf9c25294a","subject":"tests: Use common base class","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.db/commit/ccbeb04c0a7e86c92b71ea1f71bacbbf9c25294a"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2021-07-16 11:04:14.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2021-07-29 09:57:32.000000000","tz":60},"subject":"tests: Enable SADeprecationWarning warnings","message":"tests: Enable SADeprecationWarning warnings\n\nHighlight use of deprecated SQLAlchemy APIs to ensure we keep on top of\nthings. This requires resolving the following SADeprecationWarning\nwarnings:\n\n  The from_engine() method on Inspector is deprecated and will be\n  removed in a future release.  Please use the sqlalchemy.inspect()\n  function on an Engine or Connection in order to acquire an Inspector.\n\n  The Column.copy() method is deprecated and will be removed in a future\n  release.\n\n  The ColumnCollectionConstraint.copy() method is deprecated and will be\n  removed in a future release.\n\nThere are more warnings to be resolved related to SQLAlchemy 2.0, but\nthose require a special environment option (SQLALCHEMY_WARN_20) to\ntrigger and a lot of work to resolve. We\u0027ll address those in a series of\nfollow-ups.\n\nChange-Id: I34b395e6d50f4e4151178c327d94308e6f5d5b6e\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.db/commit/6232f120d4ceda7ee7fdb7aee59a5db53c51267b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.db/commit/6232f120d4ceda7ee7fdb7aee59a5db53c51267b"}]},"branch":"refs/heads/master"},"40bce5a2baf75dc87dd591e0f71a00f221a2ba92":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2021-07-29 15:37:53.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/66/802566/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.db","ref":"refs/changes/66/802566/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.db refs/changes/66/802566/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.db refs/changes/66/802566/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.db refs/changes/66/802566/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.db refs/changes/66/802566/4"}}},"commit":{"parents":[{"commit":"a6007a98b1ea111c3d41c8c00578543fa47fcab0","subject":"tests: Use common base class","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.db/commit/a6007a98b1ea111c3d41c8c00578543fa47fcab0"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2021-07-16 11:04:14.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2021-07-29 15:37:02.000000000","tz":60},"subject":"tests: Enable SADeprecationWarning warnings","message":"tests: Enable SADeprecationWarning warnings\n\nHighlight use of deprecated SQLAlchemy APIs to ensure we keep on top of\nthings. This requires resolving the following SADeprecationWarning\nwarnings:\n\n  The from_engine() method on Inspector is deprecated and will be\n  removed in a future release.  Please use the sqlalchemy.inspect()\n  function on an Engine or Connection in order to acquire an Inspector.\n\n  The Column.copy() method is deprecated and will be removed in a future\n  release.\n\n  The ColumnCollectionConstraint.copy() method is deprecated and will be\n  removed in a future release.\n\nThere are more warnings to be resolved related to SQLAlchemy 2.0, but\nthose require a special environment option (SQLALCHEMY_WARN_20) to\ntrigger and a lot of work to resolve. We\u0027ll address those in a series of\nfollow-ups.\n\nChange-Id: I34b395e6d50f4e4151178c327d94308e6f5d5b6e\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.db/commit/40bce5a2baf75dc87dd591e0f71a00f221a2ba92"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.db/commit/40bce5a2baf75dc87dd591e0f71a00f221a2ba92"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
