)]}'
{"id":"openstack%2Foslo.db~802567","triplet_id":"openstack%2Foslo.db~master~I6efb6e1d35542cd8df922c86e134fe1e0b8c2e52","project":"openstack/oslo.db","branch":"master","topic":"sqlalchemy-20","hashtags":[],"change_id":"I6efb6e1d35542cd8df922c86e134fe1e0b8c2e52","subject":"utils: Deprecate sqlalchemy-migrate-related functions","status":"MERGED","created":"2021-07-27 17:30:19.000000000","updated":"2021-08-10 14:38:55.000000000","submitted":"2021-08-10 14:38:55.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"802567-sqlalchemy-20","meta_rev_id":"fd315045a158eb05bcf979d85e8efa6d394a25d6","_number":802567,"virtual_id_number":802567,"owner":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11816,"name":"mike bayer","display_name":"Mike Bayer","email":"mike_mp@zzzcomputing.com","username":"zzzeek","status":"Red Hat"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-08-10 14:38:55.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11816,"name":"mike bayer","display_name":"Mike Bayer","email":"mike_mp@zzzcomputing.com","username":"zzzeek","status":"Red Hat"},"all":[{"value":2,"date":"2021-08-09 12:45:37.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11816,"name":"mike bayer","display_name":"Mike Bayer","email":"mike_mp@zzzcomputing.com","username":"zzzeek","status":"Red Hat"},{"value":2,"date":"2021-08-10 12:32:12.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"all":[{"value":0,"_account_id":11816,"name":"mike bayer","display_name":"Mike Bayer","email":"mike_mp@zzzcomputing.com","username":"zzzeek","status":"Red Hat"},{"value":1,"date":"2021-08-10 12:32:12.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":11816,"name":"mike bayer","display_name":"Mike Bayer","email":"mike_mp@zzzcomputing.com","username":"zzzeek","status":"Red Hat"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11816,"name":"mike bayer","display_name":"Mike Bayer","email":"mike_mp@zzzcomputing.com","username":"zzzeek","status":"Red Hat"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-07-27 19:10:43.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-08-09 12:45:37.000000000","updated_by":{"_account_id":11816,"name":"mike bayer","display_name":"Mike Bayer","email":"mike_mp@zzzcomputing.com","username":"zzzeek","status":"Red Hat"},"reviewer":{"_account_id":11816,"name":"mike bayer","display_name":"Mike Bayer","email":"mike_mp@zzzcomputing.com","username":"zzzeek","status":"Red Hat"},"state":"REVIEWER"},{"updated":"2021-08-10 12:32:12.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"}],"messages":[{"id":"785047a61d7e8e03f153930ccbbb8e5a5d5e91aa","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2021-07-27 17:30:19.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"75b045a36b18fd6d7340fd1baa66322d8963949c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-27 19:10:43.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/b671c68e72ea4c76b536bd2d86e29361 : SUCCESS in 1h 28m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1a38bf90c23444c08c679b68fc3a90dd : SUCCESS in 5m 03s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/93e950fd9ab84745a18f6275b16e8a86 : SUCCESS in 5m 28s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b806f74959fd4d8790435cb917eaaa62 : SUCCESS in 5m 16s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/17fab0a59ca547fea5b6aefd97d43630 : SUCCESS in 6m 00s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d72ed9bf5a3a43ae9f671ac0e5c80919 : SUCCESS in 5m 57s","accounts_in_message":[],"_revision_number":1},{"id":"f1aa65287178a126eb676cdd4d7a3ed0c46a2a9a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2021-07-28 14:38:18.000000000","message":"Uploaded patch set 2: New patch set was added with same tree, parent, and commit message as Patch Set 1.","accounts_in_message":[],"_revision_number":2},{"id":"9cd268d48b692cc23378f5b7827a299fedbc9c57","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-28 16:11:12.000000000","message":"Patch Set 2:\n\nBuild succeeded (check pipeline).\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/0c86692e24d8492b82f737f459a6c609 : SUCCESS in 1h 30m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/69c3e7792eaf455b9c557d731bf701f4 : SUCCESS in 5m 31s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/17b7aa9c977440e5b7a8c9c5da055ec6 : SUCCESS in 6m 12s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0a26cc6e369b4047a80da156cb095606 : SUCCESS in 6m 27s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/9f8f442559334655984add3af737e8d5 : SUCCESS in 6m 25s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d9bce9665a7c46908a69828ee966ae3c : SUCCESS in 6m 06s","accounts_in_message":[],"_revision_number":2},{"id":"0eb7b2704fa58f4f21ed9082cde8461955bf9c26","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2021-07-29 10:44:29.000000000","message":"Uploaded patch set 3: Patch Set 2 was rebased.","accounts_in_message":[],"_revision_number":3},{"id":"5eb420c9d31769f7d06894d35d392b095f348831","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-29 12:32:14.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/da6809b70d7b4f249f3873d6b8c31699 : FAILURE in 1h 45m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/67cc0c4c1aa249078b1a0701928e7ef0 : SUCCESS in 6m 12s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/579b7d1fdf3d46d3bf2b9d0b0870e13a : SUCCESS in 6m 50s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e526031a13fa40f39918a33a94171c16 : SUCCESS in 6m 05s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/78bdb42513dc4c6a86d6de41d86fbf5e : SUCCESS in 5m 35s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3c60a1c0846e44d980e92d12f71e2916 : SUCCESS in 5m 38s","accounts_in_message":[],"_revision_number":3},{"id":"b5faab82935d88d2c8ef2c69d767fae9688bc194","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2021-07-29 15:37:53.000000000","message":"Uploaded patch set 4: Patch Set 3 was rebased.","accounts_in_message":[],"_revision_number":4},{"id":"5f97acc9a18b6b97d6578f36b27d189e09822765","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-29 17:06:37.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/87c173ec21784047bc2df59f2ebda3c9 : SUCCESS in 1h 26m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cec08d2f76fd44c4ab7d62bf98509195 : SUCCESS in 6m 05s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/da222ae418974f018e883b4a82b631dd : SUCCESS in 7m 07s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e15ea88dd00b48feb96cb937f0ef4d6d : SUCCESS in 6m 43s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/989dc764e8fd4fd885f04293f9a8e8e4 : SUCCESS in 10m 03s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/32cd92d5b0dd480fbd7be9cf78cb6ec0 : SUCCESS in 6m 30s","accounts_in_message":[],"_revision_number":4},{"id":"8ce6643b449b47a7735395e54aff9d4836abba5a","author":{"_account_id":11816,"name":"mike bayer","display_name":"Mike Bayer","email":"mike_mp@zzzcomputing.com","username":"zzzeek","status":"Red Hat"},"date":"2021-08-09 12:45:37.000000000","message":"Patch Set 4: Code-Review+2\n\nsee now I am following these in relation chain order, makes much more sense :)","accounts_in_message":[],"_revision_number":4},{"id":"4113ff9a921b1c724d6bf391e8565987d7d611f5","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-08-10 12:32:12.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"5c5c4b36df37617dfd9d1319e9c243ab0bb647fe","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-10 13:04:50.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"03c5843fd98f10e0fcf7ba4dfa55435df1917c0d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-10 14:38:55.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/0bf19c8a8b9640d78cd0b8e2e4314503 : SUCCESS in 1h 20m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8861cb16f209403db6cbb0e4cf5be96a : SUCCESS in 4m 53s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b0e1884a21a84f74a58ed4a837c61cce : SUCCESS in 5m 07s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a3f8f97a7c2c4aafbee4e027366836b0 : SUCCESS in 5m 51s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a088bbd7f3e34f08aff224cc7c8c16bc : SUCCESS in 6m 24s","accounts_in_message":[],"_revision_number":4},{"id":"fd315045a158eb05bcf979d85e8efa6d394a25d6","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-10 14:38:55.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"da002a82f113afc1eb9f0329d567ae9db573091b","revisions":{"e4cfef8b0ede412f859d2097892f2936bebefd5f":{"kind":"REWORK","_number":1,"created":"2021-07-27 17:30:19.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/67/802567/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.db","ref":"refs/changes/67/802567/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.db refs/changes/67/802567/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.db refs/changes/67/802567/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.db refs/changes/67/802567/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.db refs/changes/67/802567/1"}}},"commit":{"parents":[{"commit":"63537971b75323db85923228615752fe45067efc","subject":"tests: Enable SADeprecationWarning warnings","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.db/commit/63537971b75323db85923228615752fe45067efc"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2021-07-16 12:19:41.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2021-07-16 16:53:32.000000000","tz":60},"subject":"utils: Deprecate sqlalchemy-migrate-related functions","message":"utils: Deprecate sqlalchemy-migrate-related functions\n\nThese are using private APIs of SQLAlchemy but, more importantly, are\nnot necessary in an alembic-only world. Mark them deprecated for\nremoval.\n\nChange-Id: I6efb6e1d35542cd8df922c86e134fe1e0b8c2e52\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.db/commit/e4cfef8b0ede412f859d2097892f2936bebefd5f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.db/commit/e4cfef8b0ede412f859d2097892f2936bebefd5f"}]},"branch":"refs/heads/master"},"d79d376d59f5e31d0b31f0688fd634405190c528":{"kind":"NO_CHANGE","_number":2,"created":"2021-07-28 14:38:18.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/67/802567/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.db","ref":"refs/changes/67/802567/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.db refs/changes/67/802567/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.db refs/changes/67/802567/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.db refs/changes/67/802567/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.db refs/changes/67/802567/2"}}},"commit":{"parents":[{"commit":"e29d0e28d8a56e2762887da7d80c1caa6278ac28","subject":"tests: Enable SADeprecationWarning warnings","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.db/commit/e29d0e28d8a56e2762887da7d80c1caa6278ac28"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2021-07-16 12:19:41.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2021-07-28 13:52:57.000000000","tz":60},"subject":"utils: Deprecate sqlalchemy-migrate-related functions","message":"utils: Deprecate sqlalchemy-migrate-related functions\n\nThese are using private APIs of SQLAlchemy but, more importantly, are\nnot necessary in an alembic-only world. Mark them deprecated for\nremoval.\n\nChange-Id: I6efb6e1d35542cd8df922c86e134fe1e0b8c2e52\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.db/commit/d79d376d59f5e31d0b31f0688fd634405190c528"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.db/commit/d79d376d59f5e31d0b31f0688fd634405190c528"}]},"branch":"refs/heads/master"},"7f431c872e993965c125c40dbfff2619ebb8a050":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2021-07-29 10:44:29.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/67/802567/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.db","ref":"refs/changes/67/802567/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.db refs/changes/67/802567/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.db refs/changes/67/802567/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.db refs/changes/67/802567/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.db refs/changes/67/802567/3"}}},"commit":{"parents":[{"commit":"6232f120d4ceda7ee7fdb7aee59a5db53c51267b","subject":"tests: Enable SADeprecationWarning warnings","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.db/commit/6232f120d4ceda7ee7fdb7aee59a5db53c51267b"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2021-07-16 12:19:41.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2021-07-29 09:57:34.000000000","tz":60},"subject":"utils: Deprecate sqlalchemy-migrate-related functions","message":"utils: Deprecate sqlalchemy-migrate-related functions\n\nThese are using private APIs of SQLAlchemy but, more importantly, are\nnot necessary in an alembic-only world. Mark them deprecated for\nremoval.\n\nChange-Id: I6efb6e1d35542cd8df922c86e134fe1e0b8c2e52\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.db/commit/7f431c872e993965c125c40dbfff2619ebb8a050"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.db/commit/7f431c872e993965c125c40dbfff2619ebb8a050"}]},"branch":"refs/heads/master"},"da002a82f113afc1eb9f0329d567ae9db573091b":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2021-07-29 15:37:53.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/67/802567/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.db","ref":"refs/changes/67/802567/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.db refs/changes/67/802567/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.db refs/changes/67/802567/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.db refs/changes/67/802567/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.db refs/changes/67/802567/4"}}},"commit":{"parents":[{"commit":"40bce5a2baf75dc87dd591e0f71a00f221a2ba92","subject":"tests: Enable SADeprecationWarning warnings","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.db/commit/40bce5a2baf75dc87dd591e0f71a00f221a2ba92"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2021-07-16 12:19:41.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2021-07-29 15:37:02.000000000","tz":60},"subject":"utils: Deprecate sqlalchemy-migrate-related functions","message":"utils: Deprecate sqlalchemy-migrate-related functions\n\nThese are using private APIs of SQLAlchemy but, more importantly, are\nnot necessary in an alembic-only world. Mark them deprecated for\nremoval.\n\nChange-Id: I6efb6e1d35542cd8df922c86e134fe1e0b8c2e52\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.db/commit/da002a82f113afc1eb9f0329d567ae9db573091b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.db/commit/da002a82f113afc1eb9f0329d567ae9db573091b"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
