)]}'
{"id":"openstack%2Foslo.db~802571","triplet_id":"openstack%2Foslo.db~master~I1faa8c957649a04aa080518651045b432c6bd372","project":"openstack/oslo.db","branch":"master","topic":"sqlalchemy-20","hashtags":[],"change_id":"I1faa8c957649a04aa080518651045b432c6bd372","subject":"Don\u0027t pass strings to Connection.execute()","status":"MERGED","created":"2021-07-27 17:30:19.000000000","updated":"2021-08-11 14:46:12.000000000","submitted":"2021-08-11 14:44:43.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":7,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"802571-sqlalchemy-20","meta_rev_id":"1b5dbda3b27df0a197c5ebcee28606087ca11793","_number":802571,"virtual_id_number":802571,"owner":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11816,"name":"mike bayer","display_name":"Mike Bayer","email":"mike_mp@zzzcomputing.com","username":"zzzeek","status":"Red Hat"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-08-11 14:44:43.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11816,"name":"mike bayer","display_name":"Mike Bayer","email":"mike_mp@zzzcomputing.com","username":"zzzeek","status":"Red Hat"},"all":[{"value":2,"date":"2021-08-11 14:46:12.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":11816,"name":"mike bayer","display_name":"Mike Bayer","email":"mike_mp@zzzcomputing.com","username":"zzzeek","status":"Red Hat"},{"value":2,"date":"2021-08-11 12:22:47.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"all":[{"value":0,"_account_id":11816,"name":"mike bayer","display_name":"Mike Bayer","email":"mike_mp@zzzcomputing.com","username":"zzzeek","status":"Red Hat"},{"value":1,"date":"2021-08-11 12:22:47.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":11816,"name":"mike bayer","display_name":"Mike Bayer","email":"mike_mp@zzzcomputing.com","username":"zzzeek","status":"Red Hat"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11816,"name":"mike bayer","display_name":"Mike Bayer","email":"mike_mp@zzzcomputing.com","username":"zzzeek","status":"Red Hat"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-07-27 18:54:06.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-07-27 19:03:56.000000000","updated_by":{"_account_id":11816,"name":"mike bayer","display_name":"Mike Bayer","email":"mike_mp@zzzcomputing.com","username":"zzzeek","status":"Red Hat"},"reviewer":{"_account_id":11816,"name":"mike bayer","display_name":"Mike Bayer","email":"mike_mp@zzzcomputing.com","username":"zzzeek","status":"Red Hat"},"state":"REVIEWER"},{"updated":"2021-08-11 12:22:47.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"}],"messages":[{"id":"7b466c9d5f2b6f9a5c91d406f6f5225876148421","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2021-07-27 17:30:19.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"a7f55babe364c6314a989d183db2ffcacf7ca861","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-27 18:54:06.000000000","message":"Patch Set 1: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/055487714dc34fe6b1e90bb81c30dcd9 : SUCCESS in 1h 13m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3753d811b311473fa780e3ed18989852 : FAILURE in 5m 41s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/73723999af0e4fb3927320d88c237038 : SUCCESS in 5m 39s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f9e32c44c72f4154b3c0c24c5cf8cee2 : SUCCESS in 6m 04s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/57fa3dc90563420787e088864c278886 : SUCCESS in 7m 51s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5a8af479b3ba4d2db084816cd164fbd3 : SUCCESS in 6m 03s","accounts_in_message":[],"_revision_number":1},{"id":"bc2b442bff01c97254dc9b891d90c98f0107ccb9","author":{"_account_id":11816,"name":"mike bayer","display_name":"Mike Bayer","email":"mike_mp@zzzcomputing.com","username":"zzzeek","status":"Red Hat"},"date":"2021-07-27 19:03:56.000000000","message":"Patch Set 1: Code-Review-1\n\n(3 comments)\n\nI\u0027m assuming the engine.scalar() calls go away elsewhere but there\u0027s some changes to index.create() where the \"self.engine\" argument is removed, that seems wrong because it\u0027s moving in the wrong direction","accounts_in_message":[],"_revision_number":1},{"id":"2c6d3359c316c4a1b4e983b13026d7c18293d238","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2021-07-28 14:14:37.000000000","message":"Patch Set 1:\n\n(4 comments)","accounts_in_message":[],"_revision_number":1},{"id":"4572b1c54fbbd6d8d2805d53c341ff28b27410cc","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2021-07-28 14:38:18.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"5f5e323eaf31d5f65c4c7353566c98a01d872bf9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-28 16:02:26.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/c1198cc474ae40efb593d645dfc926fd : SUCCESS in 1h 19m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ebfac717c9a54d9a95c79302329b86b6 : SUCCESS in 6m 15s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3f9f043fdf5749d4ad628e19f2690f10 : SUCCESS in 5m 38s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/00b84b98f1e440139695f8ef28d64d8e : SUCCESS in 6m 40s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/967fc5770b734e6b94662edede6e65fa : SUCCESS in 7m 17s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/350b0b3170bf41e2b96ad6cbe93fdf05 : SUCCESS in 7m 29s","accounts_in_message":[],"_revision_number":2},{"id":"273a67203dee0dfcbf62dc9fe2ec62e6e8865c76","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2021-07-29 10:44:29.000000000","message":"Uploaded patch set 3: Patch Set 2 was rebased.","accounts_in_message":[],"_revision_number":3},{"id":"d1f563cadf3e7c5c8192fe0d4447179efb345745","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-29 12:17:40.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/31883182e2e44785ab65ac163061d390 : SUCCESS in 1h 28m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3932e657776345feb63e4b4f1cf5d641 : SUCCESS in 6m 07s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f954458cc2144c08a9be4cc1334c1905 : SUCCESS in 6m 26s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f3011598a99747f78ca1fd3194807e90 : SUCCESS in 5m 29s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ed6903578189412d99b1961cc787510a : SUCCESS in 7m 55s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6e8127e102a74343a4d1f18e2aa6ad43 : SUCCESS in 6m 14s","accounts_in_message":[],"_revision_number":3},{"id":"b7c1029a5f2f01869dd9d8923bef8e3aa58ab438","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2021-07-29 15:37:53.000000000","message":"Uploaded patch set 4: Patch Set 3 was rebased.","accounts_in_message":[],"_revision_number":4},{"id":"57430fdcdeb1d07d5aad27131b07a29d5f54b5b5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-29 17:28:14.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/e46eaa6e5b2c48b1b0f4573931bcaa55 : SUCCESS in 1h 46m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/74a845f045ef4c85b022288b29f0cfb2 : SUCCESS in 6m 00s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5425666d3ea14552ab80865ff338baa9 : SUCCESS in 7m 33s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5ac0e4b77c5043709465782e7e78774a : SUCCESS in 7m 19s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a607718c8d6f40358ffde76a76fc990a : SUCCESS in 7m 27s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/69e364d67f354621a5836bf1bf964874 : SUCCESS in 6m 59s","accounts_in_message":[],"_revision_number":4},{"id":"3bc7b0ee34738dd219a4db11e3f4d3ae609fc6fe","author":{"_account_id":11816,"name":"mike bayer","display_name":"Mike Bayer","email":"mike_mp@zzzcomputing.com","username":"zzzeek","status":"Red Hat"},"date":"2021-07-29 17:32:21.000000000","message":"Patch Set 4: Code-Review+2","accounts_in_message":[],"_revision_number":4},{"id":"2d7b852dbf58db8a0e24acd2294d24b03733c4a8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2021-08-10 16:45:25.000000000","message":"Uploaded patch set 5: Patch Set 4 was rebased.","accounts_in_message":[],"_revision_number":5},{"id":"d0543f6ccf04eff540e121ab8fa13b0de0d67ab1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-10 18:07:17.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/fa44d7907ee14ef0add12ca59adb22e9 : SUCCESS in 1h 17m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5c55790547484a95b1122e854c8bb4de : SUCCESS in 5m 19s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f72b88a7c0c448ada80764af27685a58 : SUCCESS in 6m 13s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/15fd82b443e84a429e6c9bbbf324e93d : SUCCESS in 5m 43s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/1914f1bff0bb45feae82ac6efb8e5d18 : SUCCESS in 6m 07s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/245f4e04919048339b3528c7c4dd175a : SUCCESS in 6m 14s","accounts_in_message":[],"_revision_number":5},{"id":"d2ecb0cfdad6af4e9c004f6d837755e40b14e047","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-08-11 12:22:47.000000000","message":"Patch Set 5: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":5},{"id":"b4b0c7155c3f14077ecc7052060a3bfae158389d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-11 12:23:19.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"1240e89da7bd9faeaccaedd6e97e9e7da5fdd20c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-11 14:44:43.000000000","message":"Patch Set 5: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/2bc500e563b14b97bad2ec18f05d0f04 : SUCCESS in 1h 32m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/52e387e0639948808256b9f0a20668e8 : SUCCESS in 5m 13s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9f53210c75a1462599d833784adf9d28 : SUCCESS in 4m 48s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/194b75622c044c3ba2418e6ed3eb5e28 : SUCCESS in 5m 22s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/eb444da1f2df4e939a1436706eec1def : SUCCESS in 5m 07s","accounts_in_message":[],"_revision_number":5},{"id":"cb8f086cc05a524e68719914a4df8850dcec02af","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-11 14:44:43.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":5},{"id":"cb9b43196d905cea257f38dcf109fb9d57e6d6d0","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-11 14:46:12.000000000","message":"Patch Set 5:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1bafa2dd01c84fb1aa05be46d256a10c : SUCCESS in 1m 09s","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"872bc1ddccf85b2a2a571838809130749737e101","revisions":{"d5a7aa79c7a56b27c160c2d281a4e2522b4abd1e":{"kind":"REWORK","_number":1,"created":"2021-07-27 17:30:19.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/71/802571/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.db","ref":"refs/changes/71/802571/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.db refs/changes/71/802571/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.db refs/changes/71/802571/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.db refs/changes/71/802571/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.db refs/changes/71/802571/1"}}},"commit":{"parents":[{"commit":"cd4a951f0b0eec2d8ea2c1dd1ccc702c51ea9705","subject":"Remove use of MetaData.bind argument","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.db/commit/cd4a951f0b0eec2d8ea2c1dd1ccc702c51ea9705"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2021-07-16 15:52:00.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2021-07-19 14:29:33.000000000","tz":60},"subject":"Don\u0027t pass strings to Connection.execute()","message":"Don\u0027t pass strings to Connection.execute()\n\nResolve the following RemovedIn20Warning warning:\n\n  Passing a string to Connection.execute() is deprecated and will be\n  removed in version 2.0.  Use the text() construct, or the\n  Connection.exec_driver_sql() method to invoke a driver-level SQL\n  string.\n\nChange-Id: I1faa8c957649a04aa080518651045b432c6bd372\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.db/commit/d5a7aa79c7a56b27c160c2d281a4e2522b4abd1e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.db/commit/d5a7aa79c7a56b27c160c2d281a4e2522b4abd1e"}]},"branch":"refs/heads/master"},"5f5abe556c9f9acb665b202b110da1747c8be30f":{"kind":"REWORK","_number":2,"created":"2021-07-28 14:38:18.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/71/802571/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.db","ref":"refs/changes/71/802571/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.db refs/changes/71/802571/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.db refs/changes/71/802571/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.db refs/changes/71/802571/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.db refs/changes/71/802571/2"}}},"commit":{"parents":[{"commit":"169544e9f073c63266e39b2fa7b9dd1b8e4a8354","subject":"Remove use of MetaData.bind argument","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.db/commit/169544e9f073c63266e39b2fa7b9dd1b8e4a8354"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2021-07-16 15:52:00.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2021-07-28 14:14:12.000000000","tz":60},"subject":"Don\u0027t pass strings to Connection.execute()","message":"Don\u0027t pass strings to Connection.execute()\n\nResolve the following RemovedIn20Warning warning:\n\n  Passing a string to Connection.execute() is deprecated and will be\n  removed in version 2.0.  Use the text() construct, or the\n  Connection.exec_driver_sql() method to invoke a driver-level SQL\n  string.\n\nChange-Id: I1faa8c957649a04aa080518651045b432c6bd372\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.db/commit/5f5abe556c9f9acb665b202b110da1747c8be30f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.db/commit/5f5abe556c9f9acb665b202b110da1747c8be30f"}]},"branch":"refs/heads/master"},"4f1b09070098e213505da34e549834f0b2d5669d":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2021-07-29 10:44:29.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/71/802571/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.db","ref":"refs/changes/71/802571/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.db refs/changes/71/802571/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.db refs/changes/71/802571/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.db refs/changes/71/802571/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.db refs/changes/71/802571/3"}}},"commit":{"parents":[{"commit":"ab614879f9b152b81f72e4fe40e208270facdbd5","subject":"Remove use of MetaData.bind argument","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.db/commit/ab614879f9b152b81f72e4fe40e208270facdbd5"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2021-07-16 15:52:00.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2021-07-29 09:57:34.000000000","tz":60},"subject":"Don\u0027t pass strings to Connection.execute()","message":"Don\u0027t pass strings to Connection.execute()\n\nResolve the following RemovedIn20Warning warning:\n\n  Passing a string to Connection.execute() is deprecated and will be\n  removed in version 2.0.  Use the text() construct, or the\n  Connection.exec_driver_sql() method to invoke a driver-level SQL\n  string.\n\nChange-Id: I1faa8c957649a04aa080518651045b432c6bd372\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.db/commit/4f1b09070098e213505da34e549834f0b2d5669d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.db/commit/4f1b09070098e213505da34e549834f0b2d5669d"}]},"branch":"refs/heads/master"},"3a15ea3b1b7cfb2cdeca4e1ccfc721b9ad08c4f9":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2021-07-29 15:37:53.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/71/802571/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.db","ref":"refs/changes/71/802571/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.db refs/changes/71/802571/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.db refs/changes/71/802571/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.db refs/changes/71/802571/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.db refs/changes/71/802571/4"}}},"commit":{"parents":[{"commit":"c85142a945b5cc1e61c9c5b08ff537d5048d80ce","subject":"Remove use of MetaData.bind argument","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.db/commit/c85142a945b5cc1e61c9c5b08ff537d5048d80ce"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2021-07-16 15:52:00.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2021-07-29 15:37:46.000000000","tz":60},"subject":"Don\u0027t pass strings to Connection.execute()","message":"Don\u0027t pass strings to Connection.execute()\n\nResolve the following RemovedIn20Warning warning:\n\n  Passing a string to Connection.execute() is deprecated and will be\n  removed in version 2.0.  Use the text() construct, or the\n  Connection.exec_driver_sql() method to invoke a driver-level SQL\n  string.\n\nChange-Id: I1faa8c957649a04aa080518651045b432c6bd372\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.db/commit/3a15ea3b1b7cfb2cdeca4e1ccfc721b9ad08c4f9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.db/commit/3a15ea3b1b7cfb2cdeca4e1ccfc721b9ad08c4f9"}]},"branch":"refs/heads/master"},"872bc1ddccf85b2a2a571838809130749737e101":{"kind":"TRIVIAL_REBASE","_number":5,"created":"2021-08-10 16:45:25.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/71/802571/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.db","ref":"refs/changes/71/802571/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.db refs/changes/71/802571/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.db refs/changes/71/802571/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.db refs/changes/71/802571/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.db refs/changes/71/802571/5"}}},"commit":{"parents":[{"commit":"22b44ee18b1585bc2943a29ddaa18051cb1344ed","subject":"Remove use of MetaData.bind argument","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.db/commit/22b44ee18b1585bc2943a29ddaa18051cb1344ed"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2021-07-16 15:52:00.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2021-08-10 15:51:17.000000000","tz":60},"subject":"Don\u0027t pass strings to Connection.execute()","message":"Don\u0027t pass strings to Connection.execute()\n\nResolve the following RemovedIn20Warning warning:\n\n  Passing a string to Connection.execute() is deprecated and will be\n  removed in version 2.0.  Use the text() construct, or the\n  Connection.exec_driver_sql() method to invoke a driver-level SQL\n  string.\n\nChange-Id: I1faa8c957649a04aa080518651045b432c6bd372\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.db/commit/872bc1ddccf85b2a2a571838809130749737e101"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.db/commit/872bc1ddccf85b2a2a571838809130749737e101"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":11816,"name":"mike bayer","display_name":"Mike Bayer","email":"mike_mp@zzzcomputing.com","username":"zzzeek","status":"Red Hat"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
