)]}'
{"id":"openstack%2Foslo.db~802574","triplet_id":"openstack%2Foslo.db~master~I426605aa8f4332fd0b7d9f6fa0e97d7e02c7cbdc","project":"openstack/oslo.db","branch":"master","topic":"sqlalchemy-20","hashtags":[],"change_id":"I426605aa8f4332fd0b7d9f6fa0e97d7e02c7cbdc","subject":"Replace use of update.values parameter","status":"MERGED","created":"2021-07-27 17:30:19.000000000","updated":"2021-08-17 12:08:08.000000000","submitted":"2021-08-17 12:06:42.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"802574-sqlalchemy-20","meta_rev_id":"ea191fc1177dd0c09be1148c76c2ac4bb2a3cdd2","_number":802574,"virtual_id_number":802574,"owner":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11816,"name":"mike bayer","display_name":"Mike Bayer","email":"mike_mp@zzzcomputing.com","username":"zzzeek","status":"Red Hat"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-08-17 12:06:42.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11816,"name":"mike bayer","display_name":"Mike Bayer","email":"mike_mp@zzzcomputing.com","username":"zzzeek","status":"Red Hat"},"all":[{"value":2,"date":"2021-08-10 16:55:43.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11816,"name":"mike bayer","display_name":"Mike Bayer","email":"mike_mp@zzzcomputing.com","username":"zzzeek","status":"Red Hat"},{"value":2,"date":"2021-08-11 12:27:16.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"all":[{"value":0,"_account_id":11816,"name":"mike bayer","display_name":"Mike Bayer","email":"mike_mp@zzzcomputing.com","username":"zzzeek","status":"Red Hat"},{"value":1,"date":"2021-08-11 12:27:16.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":11816,"name":"mike bayer","display_name":"Mike Bayer","email":"mike_mp@zzzcomputing.com","username":"zzzeek","status":"Red Hat"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11816,"name":"mike bayer","display_name":"Mike Bayer","email":"mike_mp@zzzcomputing.com","username":"zzzeek","status":"Red Hat"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-07-27 18:45:53.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-08-10 16:55:43.000000000","updated_by":{"_account_id":11816,"name":"mike bayer","display_name":"Mike Bayer","email":"mike_mp@zzzcomputing.com","username":"zzzeek","status":"Red Hat"},"reviewer":{"_account_id":11816,"name":"mike bayer","display_name":"Mike Bayer","email":"mike_mp@zzzcomputing.com","username":"zzzeek","status":"Red Hat"},"state":"REVIEWER"},{"updated":"2021-08-11 12:27:16.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"}],"messages":[{"id":"394c8b15feca0b364a1e22c26149ac0332c4ab4c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2021-07-27 17:30:19.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"6d4ec616157d67ccce075d2a8af45f7effd82d12","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-27 18:45:53.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/39b78be820ee4e45b6f1305f66d04328 : FAILURE in 1h 03m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ec4d199ac3834a9e9c1cfb3f5a21cc39 : FAILURE in 5m 45s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8446050f4d4940fb9af9f5ea16e9c7ac : SUCCESS in 6m 01s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ca939a3232b44675a3034bb7dd56013a : SUCCESS in 5m 45s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b7e3c961287d436e9f58045796184a45 : SUCCESS in 6m 09s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f34ec51dbf204b06b9dca211cb52ffd3 : SUCCESS in 6m 45s\n\nWarning:\n  Comments left for invalid file oslo_db/sqlalchemy/engines.py","accounts_in_message":[],"_revision_number":1},{"id":"b2efc12ceec9348057b40675bf01d9a7f28f472f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2021-07-28 14:38:18.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"534f952b8809d70a818c3b355eef3472e5619e5f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-28 16:03:34.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/863d23a10a074837ac7ad5dc80837560 : SUCCESS in 1h 21m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/41e7af0633d641d9ad5830cb3254ae55 : SUCCESS in 6m 26s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/142c97d1aeb845368027f1f1d37a077e : SUCCESS in 7m 00s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d1259edc7fdc4592a55f7cd8e2b47f61 : SUCCESS in 7m 16s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/73d67225f86f4219ba4d8a4d8fcd5181 : SUCCESS in 6m 37s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fca037a152c24c408204f0c22562608b : SUCCESS in 7m 02s","accounts_in_message":[],"_revision_number":2},{"id":"881b68455ebd71c14b212153bf781140699512ba","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2021-07-29 10:44:29.000000000","message":"Uploaded patch set 3: Patch Set 2 was rebased.","accounts_in_message":[],"_revision_number":3},{"id":"d372eb37db123fb2287298c1c61e2df789527ecd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-29 12:00:25.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/f29502ba86e04aacb3d7f2bae2b9e09b : SUCCESS in 1h 10m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/439754a47f5c44c6890f46c131863c2b : SUCCESS in 7m 01s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/828fa1e9192241afa524904dcabafacb : SUCCESS in 5m 43s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0b488a2d5de743b09f4ad040548373f1 : SUCCESS in 5m 40s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/38b1d04364b74dca8c883c9ddd1ea3f6 : SUCCESS in 5m 44s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/263a04cbe35e48a68a44b7f7f8a3618d : SUCCESS in 5m 14s","accounts_in_message":[],"_revision_number":3},{"id":"fe6df98e7b6566110ed6cc3a75de355ab44ee0f3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2021-07-29 15:37:53.000000000","message":"Uploaded patch set 4: Patch Set 3 was rebased.","accounts_in_message":[],"_revision_number":4},{"id":"8bc49744e297b54544e006ca87fdc4554f5873ec","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-29 17:03:07.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/22c971c90ad042ffa82b39d8711a7fa4 : SUCCESS in 1h 22m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7b92a549adbf47a08d2c2dfab64ce47e : SUCCESS in 7m 02s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/07a9f56c889d4a3b8e6c93ff1c86ba0b : SUCCESS in 7m 19s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/41c3034a1f404690ac7d905233e0a4ae : SUCCESS in 8m 41s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/dc5a310a29f5477099d480c06a8b979d : SUCCESS in 7m 15s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/922a926d21144907bf2c48bf9dca8653 : SUCCESS in 7m 24s","accounts_in_message":[],"_revision_number":4},{"id":"0196ce1a0bc674151d643796a2f998a184a16245","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2021-08-10 16:45:25.000000000","message":"Uploaded patch set 5: Patch Set 4 was rebased.","accounts_in_message":[],"_revision_number":5},{"id":"10edf415e6b535135f41d969ca384258924d2410","author":{"_account_id":11816,"name":"mike bayer","display_name":"Mike Bayer","email":"mike_mp@zzzcomputing.com","username":"zzzeek","status":"Red Hat"},"date":"2021-08-10 16:55:43.000000000","message":"Patch Set 5: Code-Review+2","accounts_in_message":[],"_revision_number":5},{"id":"04a43fc07d305798bac1bbf63cc285cafe4af0f8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-10 18:12:10.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/2a9c1a2a7b0546f4a4d86c4145955859 : SUCCESS in 1h 24m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a0d53e626fed401c849c31a08087865b : SUCCESS in 6m 03s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9fe684c31c9f404aa90008ab25b582b5 : SUCCESS in 7m 43s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f5b78b85ca69438e9ff2f16df982944d : SUCCESS in 6m 51s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/8bb6d8207b3e4cc9b20896f13e24152d : SUCCESS in 7m 17s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/df4ec544729740c49d57222bd22603fa : SUCCESS in 5m 50s","accounts_in_message":[],"_revision_number":5},{"id":"b0ae218f1d77c23adcd445c86bd1d1fb18f9f8ac","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-08-11 12:27:16.000000000","message":"Patch Set 5: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":5},{"id":"2699bceb15ef5d1b4bb832113088c51bc84e8450","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-11 12:29:37.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"c50218e0e7c60c83ed7f2eec3d9cff40ba7946e4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-11 14:44:46.000000000","message":"Patch Set 5: Verified-2\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":5},{"id":"90e23919ee43e086ac14f44ee2925a8c519c316d","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2021-08-11 15:47:30.000000000","message":"Patch Set 5:\n\nrecheck","accounts_in_message":[],"_revision_number":5},{"id":"4429aa5eed299e009cf216f762f8b2bdf5ee9d34","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-11 17:04:53.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/967ea2559b284aef89e5c8058f0c2eeb : SUCCESS in 1h 15m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1dfb7beeca29490a8632d8a383633e7b : SUCCESS in 5m 28s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ab976033118742fd9a08ba1c45267595 : SUCCESS in 5m 19s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/95007ec226b448c2af0061294356fcc6 : SUCCESS in 6m 04s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/3840fe3dcf274f228349fb093f98d323 : SUCCESS in 5m 44s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b59203c6fbb3456c857ed82949c37c08 : SUCCESS in 6m 00s","accounts_in_message":[],"_revision_number":5},{"id":"aadcde5e546b19b7ab377351ae78d8117c132e6c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-11 17:05:34.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"02c78ce5c9a2f9759107db6d2820065d70f42046","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-11 18:25:41.000000000","message":"Patch Set 5: Verified-2\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":5},{"id":"3497d03b7a7ad45199922a3e210056832ba6b91a","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-08-17 06:26:26.000000000","message":"Patch Set 5:\n\nThis patch seems to need to be rebased","accounts_in_message":[],"_revision_number":5},{"id":"86c36658b656f2cb345da62bc1cf318cdfc555f2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-17 07:47:14.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/679cd27b28e1467b983eb32a6426c6ab : SUCCESS in 1h 14m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/54218f5e20db4b95adabc531213100d3 : SUCCESS in 4m 32s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/acb1d3e038f445d581ce30ad0ded86fd : SUCCESS in 4m 55s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3ce18bfc0fe54851955b6f2d7da89246 : SUCCESS in 5m 53s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/fd05c6e1454a4ecbb40499d138f62e36 : SUCCESS in 5m 12s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c0a647d7c24842c0bb91f11ba341675a : SUCCESS in 5m 22s","accounts_in_message":[],"_revision_number":5},{"id":"47915b227c7f159e60ce4d57050ad8b155b620ca","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-17 07:47:42.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"42707cf131e49f80b493c74abe4d376b0fe1be05","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-17 08:19:27.000000000","message":"Patch Set 5: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/095d3aa03eb24e219268f6e8369be72c : FAILURE in 30m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3c42436d51ee4bbeb03a29faf6c92d83 : SUCCESS in 6m 23s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e09056a5fa5e48aa99c9ee830fa0f5dd : SUCCESS in 7m 32s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5e9a8c31e0824c6a8894d25139ac9dc2 : SUCCESS in 5m 48s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/05567ee5dfe54db281521ea78ddc1c71 : SUCCESS in 5m 27s","accounts_in_message":[],"_revision_number":5},{"id":"0d4c30bb03e7245497c3ac2588730dcd89d9163b","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2021-08-17 09:08:10.000000000","message":"Patch Set 5:\n\nrecheck","accounts_in_message":[],"_revision_number":5},{"id":"4dc65583345d5a2f2c4b60f2c73ee25426838382","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-17 10:38:59.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/7a4e1a2eec174367bdfce31b5634e19f : SUCCESS in 1h 29m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cd538a5484d14db3a636d3d5af246e45 : SUCCESS in 5m 32s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/50fb278457654cf09d7e436e13b862e2 : SUCCESS in 5m 34s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/87ac6851ccf040b58c5e8f90fab3d9ba : SUCCESS in 6m 39s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f3323e5230204dc2874b2034f186f8f3 : SUCCESS in 6m 23s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/17b1333ab53b4fecbb3a210ee957cd8e : SUCCESS in 5m 25s","accounts_in_message":[],"_revision_number":5},{"id":"3822c88f89b06d0bc2919c29f753abcf775ae0d1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-17 10:39:28.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"616bce4f21ee978699cd9caf925f16e74317c275","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-17 12:06:42.000000000","message":"Patch Set 5: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/ff2e6f58c97d4427af51c83e735088be : SUCCESS in 1h 26m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9a970bcea2cd4587aa457c5fe5321528 : SUCCESS in 6m 36s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3ad5ec1d6e2d4946a3727df3446777d2 : SUCCESS in 5m 12s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/85dbd740d9314f6a954673665542f636 : SUCCESS in 7m 11s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/26d971807a774beb844634776fd2891b : SUCCESS in 4m 56s","accounts_in_message":[],"_revision_number":5},{"id":"f85b0b771a628554a55eacae05fcebcd790f9a58","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-17 12:06:42.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":5},{"id":"ea191fc1177dd0c09be1148c76c2ac4bb2a3cdd2","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-17 12:08:08.000000000","message":"Patch Set 5:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b2bcfa49e987414e9c75420128cfb004 : SUCCESS in 59s","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"38c2effbf4a04158a7a79833099f168f5cfa5048","revisions":{"fc158694924f756b96dc40a9b9b61ddfe814cedc":{"kind":"REWORK","_number":1,"created":"2021-07-27 17:30:19.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/74/802574/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.db","ref":"refs/changes/74/802574/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.db refs/changes/74/802574/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.db refs/changes/74/802574/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.db refs/changes/74/802574/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.db refs/changes/74/802574/1"}}},"commit":{"parents":[{"commit":"5eb7ab3c417fa13e512c02df823e66282ac5c537","subject":"Replace use of insert.values parameter","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.db/commit/5eb7ab3c417fa13e512c02df823e66282ac5c537"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2021-07-16 18:02:50.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2021-07-19 14:29:33.000000000","tz":60},"subject":"Replace use of update.values parameter","message":"Replace use of update.values parameter\n\nResolve the following RemovedIn20Warning warning:\n\n  The update.values parameter will be removed in SQLAlchemy 2.0.  Please\n  refer to the Update.values() method.\n\nChange-Id: I426605aa8f4332fd0b7d9f6fa0e97d7e02c7cbdc\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.db/commit/fc158694924f756b96dc40a9b9b61ddfe814cedc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.db/commit/fc158694924f756b96dc40a9b9b61ddfe814cedc"}]},"branch":"refs/heads/master"},"e1b5eea02558a9f4717315f4cde2fca319eb0575":{"kind":"REWORK","_number":2,"created":"2021-07-28 14:38:18.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/74/802574/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.db","ref":"refs/changes/74/802574/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.db refs/changes/74/802574/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.db refs/changes/74/802574/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.db refs/changes/74/802574/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.db refs/changes/74/802574/2"}}},"commit":{"parents":[{"commit":"56cea5622b6d7dd473814c22755163c673caa043","subject":"Replace use of insert.values parameter","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.db/commit/56cea5622b6d7dd473814c22755163c673caa043"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2021-07-16 18:02:50.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2021-07-28 14:38:01.000000000","tz":60},"subject":"Replace use of update.values parameter","message":"Replace use of update.values parameter\n\nResolve the following RemovedIn20Warning warning:\n\n  The update.values parameter will be removed in SQLAlchemy 2.0.  Please\n  refer to the Update.values() method.\n\nChange-Id: I426605aa8f4332fd0b7d9f6fa0e97d7e02c7cbdc\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.db/commit/e1b5eea02558a9f4717315f4cde2fca319eb0575"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.db/commit/e1b5eea02558a9f4717315f4cde2fca319eb0575"}]},"branch":"refs/heads/master"},"e1b4ccaf7db37be123fd38ffbcacaede17b8e272":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2021-07-29 10:44:29.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/74/802574/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.db","ref":"refs/changes/74/802574/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.db refs/changes/74/802574/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.db refs/changes/74/802574/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.db refs/changes/74/802574/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.db refs/changes/74/802574/3"}}},"commit":{"parents":[{"commit":"6b2f5bf562d5b3658363d85ae9db77b334a36a4d","subject":"Replace use of insert.values parameter","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.db/commit/6b2f5bf562d5b3658363d85ae9db77b334a36a4d"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2021-07-16 18:02:50.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2021-07-29 09:57:34.000000000","tz":60},"subject":"Replace use of update.values parameter","message":"Replace use of update.values parameter\n\nResolve the following RemovedIn20Warning warning:\n\n  The update.values parameter will be removed in SQLAlchemy 2.0.  Please\n  refer to the Update.values() method.\n\nChange-Id: I426605aa8f4332fd0b7d9f6fa0e97d7e02c7cbdc\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.db/commit/e1b4ccaf7db37be123fd38ffbcacaede17b8e272"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.db/commit/e1b4ccaf7db37be123fd38ffbcacaede17b8e272"}]},"branch":"refs/heads/master"},"045cfc14917a071f897a0d116264107da8c85905":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2021-07-29 15:37:53.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/74/802574/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.db","ref":"refs/changes/74/802574/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.db refs/changes/74/802574/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.db refs/changes/74/802574/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.db refs/changes/74/802574/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.db refs/changes/74/802574/4"}}},"commit":{"parents":[{"commit":"f6ffff972cd602ea96c207771a47ea18fddba2d9","subject":"Replace use of insert.values parameter","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.db/commit/f6ffff972cd602ea96c207771a47ea18fddba2d9"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2021-07-16 18:02:50.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2021-07-29 15:37:46.000000000","tz":60},"subject":"Replace use of update.values parameter","message":"Replace use of update.values parameter\n\nResolve the following RemovedIn20Warning warning:\n\n  The update.values parameter will be removed in SQLAlchemy 2.0.  Please\n  refer to the Update.values() method.\n\nChange-Id: I426605aa8f4332fd0b7d9f6fa0e97d7e02c7cbdc\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.db/commit/045cfc14917a071f897a0d116264107da8c85905"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.db/commit/045cfc14917a071f897a0d116264107da8c85905"}]},"branch":"refs/heads/master"},"38c2effbf4a04158a7a79833099f168f5cfa5048":{"kind":"TRIVIAL_REBASE","_number":5,"created":"2021-08-10 16:45:25.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/74/802574/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.db","ref":"refs/changes/74/802574/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.db refs/changes/74/802574/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.db refs/changes/74/802574/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.db refs/changes/74/802574/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.db refs/changes/74/802574/5"}}},"commit":{"parents":[{"commit":"6dc4b1c415bba34c875ad69dd5c01f804ae6f223","subject":"Replace use of insert.values parameter","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.db/commit/6dc4b1c415bba34c875ad69dd5c01f804ae6f223"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2021-07-16 18:02:50.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2021-08-10 15:53:47.000000000","tz":60},"subject":"Replace use of update.values parameter","message":"Replace use of update.values parameter\n\nResolve the following RemovedIn20Warning warning:\n\n  The update.values parameter will be removed in SQLAlchemy 2.0.  Please\n  refer to the Update.values() method.\n\nChange-Id: I426605aa8f4332fd0b7d9f6fa0e97d7e02c7cbdc\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.db/commit/38c2effbf4a04158a7a79833099f168f5cfa5048"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.db/commit/38c2effbf4a04158a7a79833099f168f5cfa5048"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
