)]}'
{"id":"openstack%2Foslo.db~802578","triplet_id":"openstack%2Foslo.db~master~I8acdb54d168afa1a7eac270ad6165faa287311ec","project":"openstack/oslo.db","branch":"master","topic":"sqlalchemy-20","hashtags":[],"change_id":"I8acdb54d168afa1a7eac270ad6165faa287311ec","subject":"Don\u0027t use plain string SQL statements","status":"MERGED","created":"2021-07-27 17:30:19.000000000","updated":"2021-08-17 16:23:18.000000000","submitted":"2021-08-17 16:20:06.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"802578-sqlalchemy-20","meta_rev_id":"39555e63ffdc17639b4537f0bdac9585c6ec24bf","_number":802578,"virtual_id_number":802578,"owner":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11816,"name":"mike bayer","display_name":"Mike Bayer","email":"mike_mp@zzzcomputing.com","username":"zzzeek","status":"Red Hat"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-08-17 16:20:06.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11816,"name":"mike bayer","display_name":"Mike Bayer","email":"mike_mp@zzzcomputing.com","username":"zzzeek","status":"Red Hat"},"all":[{"value":2,"date":"2021-08-10 16:59:14.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11816,"name":"mike bayer","display_name":"Mike Bayer","email":"mike_mp@zzzcomputing.com","username":"zzzeek","status":"Red Hat"},{"value":2,"date":"2021-08-11 12:01:25.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"all":[{"value":0,"_account_id":11816,"name":"mike bayer","display_name":"Mike Bayer","email":"mike_mp@zzzcomputing.com","username":"zzzeek","status":"Red Hat"},{"value":1,"date":"2021-08-11 12:01:25.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":11816,"name":"mike bayer","display_name":"Mike Bayer","email":"mike_mp@zzzcomputing.com","username":"zzzeek","status":"Red Hat"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11816,"name":"mike bayer","display_name":"Mike Bayer","email":"mike_mp@zzzcomputing.com","username":"zzzeek","status":"Red Hat"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-07-27 19:01:27.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-08-10 16:59:14.000000000","updated_by":{"_account_id":11816,"name":"mike bayer","display_name":"Mike Bayer","email":"mike_mp@zzzcomputing.com","username":"zzzeek","status":"Red Hat"},"reviewer":{"_account_id":11816,"name":"mike bayer","display_name":"Mike Bayer","email":"mike_mp@zzzcomputing.com","username":"zzzeek","status":"Red Hat"},"state":"REVIEWER"},{"updated":"2021-08-11 12:01:25.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"}],"messages":[{"id":"b7d2398adc31edc5946c26101a2d0454b344ee43","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2021-07-27 17:30:19.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"cdb8e63b506bd9a611a58c41aa2b2f577e932c1d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-27 19:01:27.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/a99e9ee83e304fbd8afd55794e68dc88 : FAILURE in 1h 12m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a65f06278cb34488989a047ed6d56bca : FAILURE in 5m 25s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4689c5d0640e44bb9cbc39f2bb6285c5 : SUCCESS in 6m 49s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/83db2e6c3f7a4325ae0068edf592aa45 : SUCCESS in 6m 41s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/823c7077b70b4d6a98ddfad74c5860c2 : SUCCESS in 6m 58s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0034dc75a47348ad94f2dfb5204ca999 : SUCCESS in 7m 13s\n\nWarning:\n  Comments left for invalid file oslo_db/sqlalchemy/engines.py","accounts_in_message":[],"_revision_number":1},{"id":"87a12f5cdc7785803a2026dfe554f641d6296c14","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2021-07-28 14:38:18.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.","accounts_in_message":[],"_revision_number":2},{"id":"870a256f6df195b81e43a57327512853a9bc2fc0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-28 15:57:55.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/254b9374348144ec961995631377c4b3 : SUCCESS in 1h 15m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ba102aa6cbee41a88137e821e2157063 : SUCCESS in 6m 13s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6016b0ea045c461ebf0eb9fb38b90606 : SUCCESS in 6m 04s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3d1c4a22029d4fc0a2dd68ea0f0c6506 : SUCCESS in 6m 28s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/800d8e3341e6494eb5c31ae0ace3076c : SUCCESS in 6m 50s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/927be8201dca43d984937089018e4764 : SUCCESS in 7m 59s","accounts_in_message":[],"_revision_number":2},{"id":"b267f413f9ff753cc7c695472fc6d34733816609","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2021-07-29 10:44:29.000000000","message":"Uploaded patch set 3: Patch Set 2 was rebased.","accounts_in_message":[],"_revision_number":3},{"id":"020be953c9a3cf08474b4102f13024d7ac59e50f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-29 12:27:27.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/860ee18aeef3402288d771e3c97bc39a : SUCCESS in 1h 17m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f5002033f2dc4b8ca6a3a6218db24c2b : SUCCESS in 5m 48s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1f75f302e50d4519a1693e904bd4f2f4 : SUCCESS in 7m 17s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e6526f057dd3453d90ddc564574096aa : SUCCESS in 5m 57s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/12f57318b98b4eab96545eec14728bee : SUCCESS in 6m 15s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3b84cb072dbb457f91c22727fa1392b2 : SUCCESS in 5m 37s","accounts_in_message":[],"_revision_number":3},{"id":"81e626d6fc5eda0a6d96fc26c484c825203c7d7a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2021-07-29 15:37:53.000000000","message":"Uploaded patch set 4: Patch Set 3 was rebased.","accounts_in_message":[],"_revision_number":4},{"id":"c04c2d7694f6a760bbfab16aa071e09bb291479c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-29 17:04:17.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/3455b7489a0447c1b87b8a136304028e : SUCCESS in 1h 22m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/df7441f6cb394e4bb92d20cac56a9f96 : SUCCESS in 6m 33s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/56d2367603a947a4a14d05458e58d429 : SUCCESS in 7m 48s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5f183ca12d3b463d861b5c605ac53bca : SUCCESS in 7m 48s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/aaf32258268648b0bf09eed360d3b0c0 : SUCCESS in 7m 53s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/781ffe02492549e287568ef134ed4a46 : SUCCESS in 7m 45s","accounts_in_message":[],"_revision_number":4},{"id":"ff1647a4e9d2acaee409e7b2a4af0fcb0b2610d7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2021-08-10 16:45:25.000000000","message":"Uploaded patch set 5: Patch Set 4 was rebased.","accounts_in_message":[],"_revision_number":5},{"id":"ad4f9f7873abd8d0d4f6f366c60a15f989a88a7e","author":{"_account_id":11816,"name":"mike bayer","display_name":"Mike Bayer","email":"mike_mp@zzzcomputing.com","username":"zzzeek","status":"Red Hat"},"date":"2021-08-10 16:59:14.000000000","message":"Patch Set 5: Code-Review+2","accounts_in_message":[],"_revision_number":5},{"id":"a0594cc8f313e19e02fe90e0c58879c1cec7c2fe","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-10 18:01:44.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/4f382577ffa54028ad0a7cc356aab17e : SUCCESS in 1h 11m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c64085b3c34c4b99b0e828fb1b948437 : SUCCESS in 6m 04s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/dc330c2f2de34303a2553b5950a6e2fb : SUCCESS in 6m 02s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9336058460cd4dbdb54984f06c26c61d : SUCCESS in 5m 54s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d3284d3da40644a9a78837ef3e5a0a12 : SUCCESS in 6m 38s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8bb418ec78cf40aebdbbb7438cc7bf05 : SUCCESS in 7m 20s","accounts_in_message":[],"_revision_number":5},{"id":"f94ad8853d0cb6117897bbd0f4f682629aa6c686","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-08-11 12:01:25.000000000","message":"Patch Set 5: Code-Review+2 Workflow+1\n\nExcellent","accounts_in_message":[],"_revision_number":5},{"id":"9ea8b3e64d7fc61e34f9d3d23ea350aa670a0dd9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-17 07:48:00.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"64c72731b26949d102b14c07ed45265794d5e1b5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-17 08:19:32.000000000","message":"Patch Set 5: Verified-2\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":5},{"id":"28b1b9f2fa5bb7a2681bdfe89f119c3aaa8f8360","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2021-08-17 09:08:31.000000000","message":"Patch Set 5:\n\nrecheck","accounts_in_message":[],"_revision_number":5},{"id":"a994f3e6f96d7e4266c3beae03d064bd67e24f5d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-17 10:29:47.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/ec879289c2e149f7acc4639f5e8a8755 : SUCCESS in 1h 15m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/be6e1cb6f8aa4e41b53d4aa6c357b691 : SUCCESS in 4m 59s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a16df0aa50f94dd485693f30691e0aeb : SUCCESS in 5m 17s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/11fe35c1d58c40a48ddeafa5986f6bd8 : SUCCESS in 5m 26s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f1f30303ca434841aaa615f07ae2b286 : SUCCESS in 6m 26s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/be7efc93c5fd4080acccc241ce37cfb5 : SUCCESS in 6m 43s","accounts_in_message":[],"_revision_number":5},{"id":"be8454d35e1205faebd461265a0ef8cc226d4703","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-17 14:36:56.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"0b4a03ec02d59f929e446c264e407907a56348a8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-17 16:20:06.000000000","message":"Patch Set 5: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/5d7fbc77a39346cd84d90f17f361120e : SUCCESS in 1h 12m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/91ae4f65b86f44d3b7b3933ad405f5cf : SUCCESS in 4m 44s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e541e71a854d43a2a5143cb92e7a1837 : SUCCESS in 4m 52s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/eea3cb413d1d4013ac75acd26f9e18b2 : SUCCESS in 6m 06s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5edb7507aad54e9a9fcb5ebb734cb439 : SUCCESS in 5m 12s","accounts_in_message":[],"_revision_number":5},{"id":"c5cc25ccc84c6f957253778fcc8104731bc8221c","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-17 16:20:06.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":5},{"id":"39555e63ffdc17639b4537f0bdac9585c6ec24bf","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-17 16:23:18.000000000","message":"Patch Set 5:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a12c3348d80842a3a518915228c6bee7 : SUCCESS in 1m 03s","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"8e7454c240aaed0d85bb3208f0ad0771c6b70c96","revisions":{"7e1b702b1be91c10deda21bf7d1e4a295e39212b":{"kind":"REWORK","_number":1,"created":"2021-07-27 17:30:19.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/78/802578/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.db","ref":"refs/changes/78/802578/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.db refs/changes/78/802578/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.db refs/changes/78/802578/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.db refs/changes/78/802578/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.db refs/changes/78/802578/1"}}},"commit":{"parents":[{"commit":"9c1cf836a936d45b63a287a7731c98d2826ae1a2","subject":"Replace use of Table.autoload parameter","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.db/commit/9c1cf836a936d45b63a287a7731c98d2826ae1a2"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2021-07-16 18:22:37.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2021-07-19 14:29:33.000000000","tz":60},"subject":"Don\u0027t use plain string SQL statements","message":"Don\u0027t use plain string SQL statements\n\nResolve the following RemovedIn20Warning warning:\n\n  Using plain strings to indicate SQL statements without using the\n  text() construct is  deprecated and will be removed in version 2.0.\n  Ensure plain SQL statements are passed using the text() construct.\n\nChange-Id: I8acdb54d168afa1a7eac270ad6165faa287311ec\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.db/commit/7e1b702b1be91c10deda21bf7d1e4a295e39212b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.db/commit/7e1b702b1be91c10deda21bf7d1e4a295e39212b"}]},"branch":"refs/heads/master"},"51d2f39956d562cb5a900b8cb7336f526fb811fa":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2021-07-28 14:38:18.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/78/802578/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.db","ref":"refs/changes/78/802578/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.db refs/changes/78/802578/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.db refs/changes/78/802578/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.db refs/changes/78/802578/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.db refs/changes/78/802578/2"}}},"commit":{"parents":[{"commit":"b2a4158d6294f328019ffc058f826bbc48f6a438","subject":"Replace use of Table.autoload parameter","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.db/commit/b2a4158d6294f328019ffc058f826bbc48f6a438"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2021-07-16 18:22:37.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2021-07-28 14:38:06.000000000","tz":60},"subject":"Don\u0027t use plain string SQL statements","message":"Don\u0027t use plain string SQL statements\n\nResolve the following RemovedIn20Warning warning:\n\n  Using plain strings to indicate SQL statements without using the\n  text() construct is  deprecated and will be removed in version 2.0.\n  Ensure plain SQL statements are passed using the text() construct.\n\nChange-Id: I8acdb54d168afa1a7eac270ad6165faa287311ec\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.db/commit/51d2f39956d562cb5a900b8cb7336f526fb811fa"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.db/commit/51d2f39956d562cb5a900b8cb7336f526fb811fa"}]},"branch":"refs/heads/master"},"eb9ebd26c7618edc94f2f3f8a2997e8d41e92b5b":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2021-07-29 10:44:29.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/78/802578/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.db","ref":"refs/changes/78/802578/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.db refs/changes/78/802578/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.db refs/changes/78/802578/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.db refs/changes/78/802578/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.db refs/changes/78/802578/3"}}},"commit":{"parents":[{"commit":"bb86d01a209131b2c40fb0e6ce2a148a143fd819","subject":"Replace use of Table.autoload parameter","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.db/commit/bb86d01a209131b2c40fb0e6ce2a148a143fd819"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2021-07-16 18:22:37.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2021-07-29 09:57:34.000000000","tz":60},"subject":"Don\u0027t use plain string SQL statements","message":"Don\u0027t use plain string SQL statements\n\nResolve the following RemovedIn20Warning warning:\n\n  Using plain strings to indicate SQL statements without using the\n  text() construct is  deprecated and will be removed in version 2.0.\n  Ensure plain SQL statements are passed using the text() construct.\n\nChange-Id: I8acdb54d168afa1a7eac270ad6165faa287311ec\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.db/commit/eb9ebd26c7618edc94f2f3f8a2997e8d41e92b5b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.db/commit/eb9ebd26c7618edc94f2f3f8a2997e8d41e92b5b"}]},"branch":"refs/heads/master"},"c1f0f7e62a05272d788a3e339b8a722014218cb1":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2021-07-29 15:37:53.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/78/802578/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.db","ref":"refs/changes/78/802578/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.db refs/changes/78/802578/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.db refs/changes/78/802578/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.db refs/changes/78/802578/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.db refs/changes/78/802578/4"}}},"commit":{"parents":[{"commit":"0d2b79db3ed30657272cf17ae3115eb7f1a46ae8","subject":"Replace use of Table.autoload parameter","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.db/commit/0d2b79db3ed30657272cf17ae3115eb7f1a46ae8"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2021-07-16 18:22:37.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2021-07-29 15:37:46.000000000","tz":60},"subject":"Don\u0027t use plain string SQL statements","message":"Don\u0027t use plain string SQL statements\n\nResolve the following RemovedIn20Warning warning:\n\n  Using plain strings to indicate SQL statements without using the\n  text() construct is  deprecated and will be removed in version 2.0.\n  Ensure plain SQL statements are passed using the text() construct.\n\nChange-Id: I8acdb54d168afa1a7eac270ad6165faa287311ec\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.db/commit/c1f0f7e62a05272d788a3e339b8a722014218cb1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.db/commit/c1f0f7e62a05272d788a3e339b8a722014218cb1"}]},"branch":"refs/heads/master"},"8e7454c240aaed0d85bb3208f0ad0771c6b70c96":{"kind":"TRIVIAL_REBASE","_number":5,"created":"2021-08-10 16:45:25.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/78/802578/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.db","ref":"refs/changes/78/802578/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.db refs/changes/78/802578/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.db refs/changes/78/802578/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.db refs/changes/78/802578/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.db refs/changes/78/802578/5"}}},"commit":{"parents":[{"commit":"6a013c606b023b0ba7f3cb1bcc094d9adc1710f3","subject":"Replace use of Table.autoload parameter","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.db/commit/6a013c606b023b0ba7f3cb1bcc094d9adc1710f3"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2021-07-16 18:22:37.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2021-08-10 15:53:47.000000000","tz":60},"subject":"Don\u0027t use plain string SQL statements","message":"Don\u0027t use plain string SQL statements\n\nResolve the following RemovedIn20Warning warning:\n\n  Using plain strings to indicate SQL statements without using the\n  text() construct is  deprecated and will be removed in version 2.0.\n  Ensure plain SQL statements are passed using the text() construct.\n\nChange-Id: I8acdb54d168afa1a7eac270ad6165faa287311ec\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.db/commit/8e7454c240aaed0d85bb3208f0ad0771c6b70c96"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.db/commit/8e7454c240aaed0d85bb3208f0ad0771c6b70c96"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
