)]}'
{"id":"openstack%2Foslo.db~802581","triplet_id":"openstack%2Foslo.db~master~Ic3c9c5bb008b6299f2ed9a59bda0329fab9d554d","project":"openstack/oslo.db","branch":"master","topic":"sqlalchemy-20","hashtags":[],"change_id":"Ic3c9c5bb008b6299f2ed9a59bda0329fab9d554d","subject":"Replace use of Engine.scalar()","status":"MERGED","created":"2021-07-27 17:30:19.000000000","updated":"2021-11-12 12:20:45.000000000","submitted":"2021-11-12 12:20:45.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":4,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"802581-sqlalchemy-20","meta_rev_id":"f17d1661cbd73b5b7c91af8693979a76d756df2d","_number":802581,"virtual_id_number":802581,"owner":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":11816,"name":"mike bayer","display_name":"Mike Bayer","email":"mike_mp@zzzcomputing.com","username":"zzzeek","status":"Red Hat"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-11-12 12:20:44.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11816,"name":"mike bayer","display_name":"Mike Bayer","email":"mike_mp@zzzcomputing.com","username":"zzzeek","status":"Red Hat"},"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":2,"date":"2021-08-24 13:03:45.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11816,"name":"mike bayer","display_name":"Mike Bayer","email":"mike_mp@zzzcomputing.com","username":"zzzeek","status":"Red Hat"},{"value":2,"date":"2021-08-11 12:32:01.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":1,"date":"2021-11-12 10:20:33.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":11816,"name":"mike bayer","display_name":"Mike Bayer","email":"mike_mp@zzzcomputing.com","username":"zzzeek","status":"Red Hat"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":11816,"name":"mike bayer","display_name":"Mike Bayer","email":"mike_mp@zzzcomputing.com","username":"zzzeek","status":"Red Hat"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11816,"name":"mike bayer","display_name":"Mike Bayer","email":"mike_mp@zzzcomputing.com","username":"zzzeek","status":"Red Hat"},{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-07-27 19:07:59.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-08-11 12:32:01.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2021-08-24 13:03:45.000000000","updated_by":{"_account_id":11816,"name":"mike bayer","display_name":"Mike Bayer","email":"mike_mp@zzzcomputing.com","username":"zzzeek","status":"Red Hat"},"reviewer":{"_account_id":11816,"name":"mike bayer","display_name":"Mike Bayer","email":"mike_mp@zzzcomputing.com","username":"zzzeek","status":"Red Hat"},"state":"REVIEWER"}],"messages":[{"id":"71808d37f77b6361e3453e6b50503d7a861578b0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2021-07-27 17:30:19.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"62c434536038215d22873ef643975400020c04b4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-27 19:07:59.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/623a440540fb46fba7866eaf46248fdd : FAILURE in 1h 16m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4da7655dc5234ec9866931b1d382fd16 : FAILURE in 7m 00s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/93a3bdfb86a94c9ea715a151b3bade15 : SUCCESS in 6m 05s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8e4b0cb1931443faa0ed06e2966a03fa : SUCCESS in 6m 04s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/db21efebf49f48d0a2e75a180f53cc65 : SUCCESS in 6m 52s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/521580cae1d74756ab7788121b8ba437 : SUCCESS in 6m 12s\n\nWarning:\n  Comments left for invalid file oslo_db/sqlalchemy/engines.py","accounts_in_message":[],"_revision_number":1},{"id":"a98f1771f0a3edf79974182e2f6a3cedaa0623b8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2021-07-28 14:38:18.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.","accounts_in_message":[],"_revision_number":2},{"id":"55b1d0f2d8d4983308b043ccee9c8459a11408bf","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-28 16:20:13.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/d7cdf767ffb04df589b3dc4478733a3b : SUCCESS in 1h 33m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fc74b174c01b4031a9551972df3b7664 : SUCCESS in 6m 17s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/12af93a118e84c669166fb8ee7d545d6 : SUCCESS in 6m 57s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/526b7c87700b4dd18ed73b43f238ea95 : SUCCESS in 6m 44s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/35ddefde52bb45c78708d020bc02902a : SUCCESS in 6m 43s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/180124ea538c411e8a0fb76bda135076 : SUCCESS in 6m 30s","accounts_in_message":[],"_revision_number":2},{"id":"933a2c4862d91185bb0c4af67deb08b9ea276150","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2021-07-29 10:44:29.000000000","message":"Uploaded patch set 3: Patch Set 2 was rebased.","accounts_in_message":[],"_revision_number":3},{"id":"24ec5b4774f56500cf4ad7313c9e273d70aa3a3e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-29 12:09:12.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/353e4d2e5aa144e2b548462ee7fcfed7 : SUCCESS in 1h 15m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ba55e36b604d403295965dbf8ac8e685 : SUCCESS in 5m 20s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c9a092c9fe534c049cd638190defbf9f : SUCCESS in 5m 39s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a92fb52502fe4e129f6ade094a2feb7e : SUCCESS in 6m 17s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/7867c87e4da143afb7b50ee8e2075bbb : SUCCESS in 6m 29s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b2c0fb87a29f423ebb5ac9ab3b0a44cf : SUCCESS in 7m 33s","accounts_in_message":[],"_revision_number":3},{"id":"e7074981860b2dfc5a5100f239b67f67b37cd63a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2021-07-29 15:37:53.000000000","message":"Uploaded patch set 4: Patch Set 3 was rebased.","accounts_in_message":[],"_revision_number":4},{"id":"d082d2b773e5bc0d5090f1b054fc86c72797363a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-29 17:04:22.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/4f9ba66c91bd4950b4d3eb74c89cd14a : SUCCESS in 1h 23m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ee2119209f5d480cb666fb1a7d4d002a : SUCCESS in 7m 10s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/07754fdd78e543349e977850050c90b4 : SUCCESS in 6m 43s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/cbd1421f37c045e49de0267377ce49f1 : SUCCESS in 8m 08s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/98275041d08344619fdd8f6a3bdc997c : SUCCESS in 8m 07s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f4bc3f34d44746079de905848c557df8 : SUCCESS in 7m 07s","accounts_in_message":[],"_revision_number":4},{"id":"2326b41062770a88cbc8a1b63aed4d3b5ee6ce55","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2021-08-10 16:45:25.000000000","message":"Uploaded patch set 5: Patch Set 4 was rebased.","accounts_in_message":[],"_revision_number":5},{"id":"514bb5e8f2d8e855fc95ac4f1719ce221321348f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-10 17:57:56.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/a769bf09ed3949a4bfa678ad1707a978 : SUCCESS in 1h 06m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3ce0aa56700241e8b6849f7a3d89b967 : SUCCESS in 5m 40s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/68d95d59e8d74698bb85d42a088c02f4 : SUCCESS in 6m 54s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/93ad3b6829b94db78f9fc68546037357 : SUCCESS in 5m 45s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d201a5015cdb482a9542fbe02410d8b9 : SUCCESS in 6m 57s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/08ade96e179740c38372b7ad1a3f3ff8 : SUCCESS in 6m 06s","accounts_in_message":[],"_revision_number":5},{"id":"883cc9a852b9c1a902db51868176ed8683d7ba0f","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-08-11 12:32:01.000000000","message":"Patch Set 5: Code-Review+2","accounts_in_message":[],"_revision_number":5},{"id":"9662fc564b2820263defdc0f6091255ad95b76bf","author":{"_account_id":11816,"name":"mike bayer","display_name":"Mike Bayer","email":"mike_mp@zzzcomputing.com","username":"zzzeek","status":"Red Hat"},"date":"2021-08-24 13:03:45.000000000","message":"Patch Set 5: Code-Review+2\n\n(2 comments)\n\nI\u0027m +2ing this because I know subsequent merges fix this further.  it seems a little tricky to me to do it this way but we can just get through it :)","accounts_in_message":[],"_revision_number":5},{"id":"8a25567f9f970f4446c09791dd9c21f257642b01","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2021-08-25 10:18:06.000000000","message":"Patch Set 5:\n\n(2 comments)","accounts_in_message":[],"_revision_number":5},{"id":"cb1a557af3e254742f90e3f4c85939e3e9616b10","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2021-11-12 10:20:33.000000000","message":"Patch Set 5: Workflow+1","accounts_in_message":[],"_revision_number":5},{"id":"2b198792abec98459aeb43cc9636f30db6ca0de4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-12 10:21:44.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"c5c4ecd5604dcef82069737a945f4883cc8fc3cd","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-12 12:20:44.000000000","message":"Patch Set 5: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/1122d6d3450747388b00a395007648fc : SUCCESS in 1h 22m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2dfd284516834246b9356e6489e9cbf8 : SUCCESS in 5m 47s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0ee6139b77834b6298d7655f6f228d49 : SUCCESS in 5m 12s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ecf726fbe47042968c9b5a80e84cdd69 : SUCCESS in 6m 04s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0961cdd0ef7c4bf7b19bc92b9d43baad : SUCCESS in 6m 12s","accounts_in_message":[],"_revision_number":5},{"id":"f17d1661cbd73b5b7c91af8693979a76d756df2d","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-12 12:20:45.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"fadc0ef9022a360016d2e9af756fb93c9c308467","revisions":{"5743f8c1ecb933580a76ddf2d365c973f5be6d51":{"kind":"REWORK","_number":1,"created":"2021-07-27 17:30:19.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/81/802581/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.db","ref":"refs/changes/81/802581/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.db refs/changes/81/802581/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.db refs/changes/81/802581/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.db refs/changes/81/802581/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.db refs/changes/81/802581/1"}}},"commit":{"parents":[{"commit":"a30ca84ec5f159e41de6b9948306189dde4db2ec","subject":"Don\u0027t use the \u0027Row.keys()\u0027 method","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.db/commit/a30ca84ec5f159e41de6b9948306189dde4db2ec"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2021-07-19 09:40:32.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2021-07-19 14:29:33.000000000","tz":60},"subject":"Replace use of Engine.scalar()","message":"Replace use of Engine.scalar()\n\nResolve the following RemovedIn20Warning warning:\n\n  The Engine.scalar() method is considered legacy as of the 1.x series\n  of SQLAlchemy and will be removed in 2.0. All statement execution in\n  SQLAlchemy 2.0 is performed by the Connection.execute() method of\n  Connection, or in the ORM by the Session.execute() method of Session;\n  the Result.scalar() method can then be used to return a scalar result.\n\nChange-Id: Ic3c9c5bb008b6299f2ed9a59bda0329fab9d554d\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.db/commit/5743f8c1ecb933580a76ddf2d365c973f5be6d51"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.db/commit/5743f8c1ecb933580a76ddf2d365c973f5be6d51"}]},"branch":"refs/heads/master"},"c66ae42aec4b6f1209f70209ee58eeea7611faa6":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2021-07-28 14:38:18.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/81/802581/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.db","ref":"refs/changes/81/802581/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.db refs/changes/81/802581/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.db refs/changes/81/802581/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.db refs/changes/81/802581/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.db refs/changes/81/802581/2"}}},"commit":{"parents":[{"commit":"8f259eb2ea58442bb610180cfeb658ecd7d956ad","subject":"Don\u0027t use the \u0027Row.keys()\u0027 method","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.db/commit/8f259eb2ea58442bb610180cfeb658ecd7d956ad"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2021-07-19 09:40:32.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2021-07-28 14:38:06.000000000","tz":60},"subject":"Replace use of Engine.scalar()","message":"Replace use of Engine.scalar()\n\nResolve the following RemovedIn20Warning warning:\n\n  The Engine.scalar() method is considered legacy as of the 1.x series\n  of SQLAlchemy and will be removed in 2.0. All statement execution in\n  SQLAlchemy 2.0 is performed by the Connection.execute() method of\n  Connection, or in the ORM by the Session.execute() method of Session;\n  the Result.scalar() method can then be used to return a scalar result.\n\nChange-Id: Ic3c9c5bb008b6299f2ed9a59bda0329fab9d554d\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.db/commit/c66ae42aec4b6f1209f70209ee58eeea7611faa6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.db/commit/c66ae42aec4b6f1209f70209ee58eeea7611faa6"}]},"branch":"refs/heads/master"},"f7a9143e5d6afcf4430f465ac6aefcc228093356":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2021-07-29 10:44:29.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/81/802581/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.db","ref":"refs/changes/81/802581/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.db refs/changes/81/802581/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.db refs/changes/81/802581/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.db refs/changes/81/802581/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.db refs/changes/81/802581/3"}}},"commit":{"parents":[{"commit":"d07a437173a6abbab575b5b31cca10b0efdca672","subject":"Don\u0027t use the \u0027Row.keys()\u0027 method","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.db/commit/d07a437173a6abbab575b5b31cca10b0efdca672"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2021-07-19 09:40:32.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2021-07-29 09:57:34.000000000","tz":60},"subject":"Replace use of Engine.scalar()","message":"Replace use of Engine.scalar()\n\nResolve the following RemovedIn20Warning warning:\n\n  The Engine.scalar() method is considered legacy as of the 1.x series\n  of SQLAlchemy and will be removed in 2.0. All statement execution in\n  SQLAlchemy 2.0 is performed by the Connection.execute() method of\n  Connection, or in the ORM by the Session.execute() method of Session;\n  the Result.scalar() method can then be used to return a scalar result.\n\nChange-Id: Ic3c9c5bb008b6299f2ed9a59bda0329fab9d554d\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.db/commit/f7a9143e5d6afcf4430f465ac6aefcc228093356"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.db/commit/f7a9143e5d6afcf4430f465ac6aefcc228093356"}]},"branch":"refs/heads/master"},"7b18b0284b75c28d67502e5c95592dc550f46cef":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2021-07-29 15:37:53.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/81/802581/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.db","ref":"refs/changes/81/802581/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.db refs/changes/81/802581/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.db refs/changes/81/802581/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.db refs/changes/81/802581/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.db refs/changes/81/802581/4"}}},"commit":{"parents":[{"commit":"5018ffbcc3afbc26fda4ffb8ffc908e466c5f837","subject":"Don\u0027t use the \u0027Row.keys()\u0027 method","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.db/commit/5018ffbcc3afbc26fda4ffb8ffc908e466c5f837"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2021-07-19 09:40:32.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2021-07-29 15:37:46.000000000","tz":60},"subject":"Replace use of Engine.scalar()","message":"Replace use of Engine.scalar()\n\nResolve the following RemovedIn20Warning warning:\n\n  The Engine.scalar() method is considered legacy as of the 1.x series\n  of SQLAlchemy and will be removed in 2.0. All statement execution in\n  SQLAlchemy 2.0 is performed by the Connection.execute() method of\n  Connection, or in the ORM by the Session.execute() method of Session;\n  the Result.scalar() method can then be used to return a scalar result.\n\nChange-Id: Ic3c9c5bb008b6299f2ed9a59bda0329fab9d554d\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.db/commit/7b18b0284b75c28d67502e5c95592dc550f46cef"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.db/commit/7b18b0284b75c28d67502e5c95592dc550f46cef"}]},"branch":"refs/heads/master"},"fadc0ef9022a360016d2e9af756fb93c9c308467":{"kind":"TRIVIAL_REBASE","_number":5,"created":"2021-08-10 16:45:25.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/81/802581/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.db","ref":"refs/changes/81/802581/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.db refs/changes/81/802581/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.db refs/changes/81/802581/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.db refs/changes/81/802581/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.db refs/changes/81/802581/5"}}},"commit":{"parents":[{"commit":"8772fcc349226b152e8a04a8ed3e883231412043","subject":"Don\u0027t use the \u0027Row.keys()\u0027 method","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.db/commit/8772fcc349226b152e8a04a8ed3e883231412043"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2021-07-19 09:40:32.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2021-08-10 15:53:47.000000000","tz":60},"subject":"Replace use of Engine.scalar()","message":"Replace use of Engine.scalar()\n\nResolve the following RemovedIn20Warning warning:\n\n  The Engine.scalar() method is considered legacy as of the 1.x series\n  of SQLAlchemy and will be removed in 2.0. All statement execution in\n  SQLAlchemy 2.0 is performed by the Connection.execute() method of\n  Connection, or in the ORM by the Session.execute() method of Session;\n  the Result.scalar() method can then be used to return a scalar result.\n\nChange-Id: Ic3c9c5bb008b6299f2ed9a59bda0329fab9d554d\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.db/commit/fadc0ef9022a360016d2e9af756fb93c9c308467"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.db/commit/fadc0ef9022a360016d2e9af756fb93c9c308467"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":11816,"name":"mike bayer","display_name":"Mike Bayer","email":"mike_mp@zzzcomputing.com","username":"zzzeek","status":"Red Hat"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
