)]}'
{"id":"openstack%2Foslo.db~802583","triplet_id":"openstack%2Foslo.db~master~Ie0acba4a315c85ec7236e44a22449e0ad920ca9b","project":"openstack/oslo.db","branch":"master","topic":"sqlalchemy-20","hashtags":[],"change_id":"Ie0acba4a315c85ec7236e44a22449e0ad920ca9b","subject":"Replace use of Executable.execute method","status":"MERGED","created":"2021-07-27 17:30:19.000000000","updated":"2021-11-12 12:24:29.000000000","submitted":"2021-11-12 12:20:49.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":4,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"802583-sqlalchemy-20","meta_rev_id":"d4a4a17d5255841b8d51147205444f43f0b9664f","_number":802583,"virtual_id_number":802583,"owner":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":11816,"name":"mike bayer","display_name":"Mike Bayer","email":"mike_mp@zzzcomputing.com","username":"zzzeek","status":"Red Hat"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-11-12 12:20:48.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11816,"name":"mike bayer","display_name":"Mike Bayer","email":"mike_mp@zzzcomputing.com","username":"zzzeek","status":"Red Hat"},"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":2,"date":"2021-08-24 13:05:02.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11816,"name":"mike bayer","display_name":"Mike Bayer","email":"mike_mp@zzzcomputing.com","username":"zzzeek","status":"Red Hat"},{"value":2,"date":"2021-08-11 12:43:16.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":1,"date":"2021-11-12 10:20:45.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":11816,"name":"mike bayer","display_name":"Mike Bayer","email":"mike_mp@zzzcomputing.com","username":"zzzeek","status":"Red Hat"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":11816,"name":"mike bayer","display_name":"Mike Bayer","email":"mike_mp@zzzcomputing.com","username":"zzzeek","status":"Red Hat"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11816,"name":"mike bayer","display_name":"Mike Bayer","email":"mike_mp@zzzcomputing.com","username":"zzzeek","status":"Red Hat"},{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-07-27 18:53:46.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-08-11 12:43:16.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2021-08-24 13:05:02.000000000","updated_by":{"_account_id":11816,"name":"mike bayer","display_name":"Mike Bayer","email":"mike_mp@zzzcomputing.com","username":"zzzeek","status":"Red Hat"},"reviewer":{"_account_id":11816,"name":"mike bayer","display_name":"Mike Bayer","email":"mike_mp@zzzcomputing.com","username":"zzzeek","status":"Red Hat"},"state":"REVIEWER"}],"messages":[{"id":"25243432a0de0a9bc03e61bf43d0f689751d67ba","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2021-07-27 17:30:19.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"5ffffa38a0d76b2f49c83449190c1c2c61b62d2a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-27 18:53:46.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/724546b680cf4b03b3ebbd4bd7515ff8 : FAILURE in 1h 06m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d0b4e061a6f24a0a95d42c60863271eb : FAILURE in 5m 48s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/978e96e48b594206ae509b174797a321 : SUCCESS in 6m 03s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/59888752bd1c42d682f92ec70563b483 : SUCCESS in 6m 20s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/672d75b1b31e463781ef3740bdf27999 : SUCCESS in 7m 40s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8b42bbba08de40a9a9f51af4dbaec1f0 : SUCCESS in 6m 01s\n\nWarning:\n  Comments left for invalid file oslo_db/sqlalchemy/engines.py","accounts_in_message":[],"_revision_number":1},{"id":"4ae93415b92c20d6564327cd4d73f752284bf454","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2021-07-28 14:38:18.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.","accounts_in_message":[],"_revision_number":2},{"id":"e4f89b263e363deeec95db88ed550e798a871e26","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-28 16:20:26.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/cfe84b473161464b8dd84cb7c8c72b06 : SUCCESS in 1h 31m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c275e953f701466889ea89344cbbd14b : SUCCESS in 6m 46s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5089d5db94f34acb9ad36c0f241cbe58 : SUCCESS in 7m 15s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d6abb30b20544e7e9cfcafd6de5d90fe : SUCCESS in 6m 40s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/5776f1390dc94514b1594b89c44894c7 : SUCCESS in 7m 15s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/87436ff68fa74cc8aea7c46d5d08afe9 : SUCCESS in 7m 21s","accounts_in_message":[],"_revision_number":2},{"id":"4f84a700bdf111389f0d38e34394c117a53e6c3f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2021-07-29 10:44:29.000000000","message":"Uploaded patch set 3: Patch Set 2 was rebased.","accounts_in_message":[],"_revision_number":3},{"id":"3b1e922a8bd5ae20f0b9aae2a7f7493278215faf","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-29 12:23:07.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/fba33ec18ad84f4c8f09cbe2610c1fbf : SUCCESS in 1h 23m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2a228f8d2aeb4d48a7b156443b5cda08 : SUCCESS in 6m 20s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1791b95903aa4109aed87db4a4351953 : SUCCESS in 6m 15s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/165a90c57e0547e28c5216b31eee9d58 : SUCCESS in 5m 41s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a0fa12060ffc44cba511cf7bc361bd56 : SUCCESS in 6m 49s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/738779696d9b40f3bd0f25601c8b82ae : SUCCESS in 6m 09s","accounts_in_message":[],"_revision_number":3},{"id":"484a26373ca5e7a8ff9a12a27c299342dd65d696","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2021-07-29 15:37:53.000000000","message":"Uploaded patch set 4: Patch Set 3 was rebased.","accounts_in_message":[],"_revision_number":4},{"id":"cac956c64f8fa583b5e99bd0b2e67db8cbf14012","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-29 17:10:58.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/48a1b310ac8a40399895c18c1681b1f0 : SUCCESS in 1h 24m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/45dd997225bc4f17a2680a39278a8718 : SUCCESS in 7m 22s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/02ca13b929e04d17921e85422683f8b9 : SUCCESS in 7m 23s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/4a059758381d43468ca390a8ddbd6c83 : SUCCESS in 7m 29s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/dc76e26bac904cb0b8346f0fe68751cb : SUCCESS in 7m 24s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d5746899314344d6a9489c2a2089a34f : SUCCESS in 8m 00s","accounts_in_message":[],"_revision_number":4},{"id":"c2e21c89a5d03604f41ef2277be6ef61f340288d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2021-08-10 16:45:25.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"c2778c77729a0b07ce2fa38f61e414ce64a9af4b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-10 18:20:17.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/578c336465bb45348ec2a178bd923a11 : SUCCESS in 1h 19m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/25a5f7eafcfe4920b185bb621867ed47 : SUCCESS in 6m 17s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/149a4e23a90947cabdff1c9f12155c49 : SUCCESS in 6m 08s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8a33746ca49f4b828e7c577c868741ec : SUCCESS in 6m 08s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/9a80bcbdeb3d4c74b7c9efeb78ec9b24 : SUCCESS in 6m 49s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/50e672d25c8b45dea4fccb0f52853f86 : SUCCESS in 6m 44s","accounts_in_message":[],"_revision_number":5},{"id":"62aaf82baf28ed746e05e5a7bda552672a9b095e","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-08-11 12:43:16.000000000","message":"Patch Set 5: Code-Review+2","accounts_in_message":[],"_revision_number":5},{"id":"fa1f1c677acdb79cbf99f63968f75839adb88455","author":{"_account_id":11816,"name":"mike bayer","display_name":"Mike Bayer","email":"mike_mp@zzzcomputing.com","username":"zzzeek","status":"Red Hat"},"date":"2021-08-24 13:05:02.000000000","message":"Patch Set 5: Code-Review+2\n\n(2 comments)","accounts_in_message":[],"_revision_number":5},{"id":"643e6bd97c82514e87f2e1ebc9aa2d4d8662c0cd","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2021-08-25 10:16:22.000000000","message":"Patch Set 5:\n\n(2 comments)","accounts_in_message":[],"_revision_number":5},{"id":"96cefada1247542135602dbefb647b9dfdc6bffc","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2021-11-12 10:20:45.000000000","message":"Patch Set 5: Workflow+1","accounts_in_message":[],"_revision_number":5},{"id":"4ae3af123420fa91f165631699efe35daeacd1fb","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-12 10:22:14.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"7ffddd152fd7589e13bb74240c96879500ea4db3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-12 12:20:48.000000000","message":"Patch Set 5: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/265e4dd62a1d4339a554e405f2207355 : SUCCESS in 1h 22m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6b2fc2f280a64a91b2d7bcc601458078 : SUCCESS in 5m 45s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f71e029fc5d64f7dbedc0a9d4653e167 : SUCCESS in 5m 46s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/15553868263a439cab2b850364d9582f : SUCCESS in 6m 37s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d691ffe6eceb4b93afe42f85863d9c9a : SUCCESS in 5m 29s","accounts_in_message":[],"_revision_number":5},{"id":"63f147da3fc31ca87faa8552f765f088147b2467","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-12 12:20:49.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":5},{"id":"d4a4a17d5255841b8d51147205444f43f0b9664f","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-12 12:24:29.000000000","message":"Patch Set 5:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b6e5497d5ddb4da498d0c5bcca799d0e : SUCCESS in 1m 06s","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"62d77fc6df9bf27a334613d42a2ffa14c9711928","revisions":{"20ab6c380f1db0abc69cdd42c15a8f6e996b8521":{"kind":"REWORK","_number":1,"created":"2021-07-27 17:30:19.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/83/802583/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.db","ref":"refs/changes/83/802583/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.db refs/changes/83/802583/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.db refs/changes/83/802583/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.db refs/changes/83/802583/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.db refs/changes/83/802583/1"}}},"commit":{"parents":[{"commit":"9280e726730d2c22b6eeedc7c9dab4dc57b4e4dd","subject":"Remove unnecessary warning filter","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.db/commit/9280e726730d2c22b6eeedc7c9dab4dc57b4e4dd"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2021-07-19 09:53:01.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2021-07-19 16:44:14.000000000","tz":60},"subject":"Replace use of Executable.execute method","message":"Replace use of Executable.execute method\n\nResolve the following RemovedIn20Warning warning:\n\n  The Executable.execute() method is considered legacy as of the 1.x\n  series of SQLAlchemy and will be removed in 2.0. All statement\n  execution in SQLAlchemy 2.0 is performed by the Connection.execute()\n  method of Connection, or in the ORM by the Session.execute() method of\n  Session.\n\nChange-Id: Ie0acba4a315c85ec7236e44a22449e0ad920ca9b\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.db/commit/20ab6c380f1db0abc69cdd42c15a8f6e996b8521"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.db/commit/20ab6c380f1db0abc69cdd42c15a8f6e996b8521"}]},"branch":"refs/heads/master"},"e9fc6a0eaa0a99c5d70dffca292cbe2ec225019c":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2021-07-28 14:38:18.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/83/802583/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.db","ref":"refs/changes/83/802583/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.db refs/changes/83/802583/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.db refs/changes/83/802583/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.db refs/changes/83/802583/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.db refs/changes/83/802583/2"}}},"commit":{"parents":[{"commit":"3346d87187afc3d02decc30b550aa14d364f9677","subject":"Remove unnecessary warning filter","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.db/commit/3346d87187afc3d02decc30b550aa14d364f9677"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2021-07-19 09:53:01.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2021-07-28 14:38:06.000000000","tz":60},"subject":"Replace use of Executable.execute method","message":"Replace use of Executable.execute method\n\nResolve the following RemovedIn20Warning warning:\n\n  The Executable.execute() method is considered legacy as of the 1.x\n  series of SQLAlchemy and will be removed in 2.0. All statement\n  execution in SQLAlchemy 2.0 is performed by the Connection.execute()\n  method of Connection, or in the ORM by the Session.execute() method of\n  Session.\n\nChange-Id: Ie0acba4a315c85ec7236e44a22449e0ad920ca9b\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.db/commit/e9fc6a0eaa0a99c5d70dffca292cbe2ec225019c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.db/commit/e9fc6a0eaa0a99c5d70dffca292cbe2ec225019c"}]},"branch":"refs/heads/master"},"471eb88d2790e95d564b448002c2465f68f2e041":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2021-07-29 10:44:29.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/83/802583/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.db","ref":"refs/changes/83/802583/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.db refs/changes/83/802583/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.db refs/changes/83/802583/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.db refs/changes/83/802583/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.db refs/changes/83/802583/3"}}},"commit":{"parents":[{"commit":"2fb4a5aa6aaa7d98861a9ae735c69f850a1f7f08","subject":"Remove unnecessary warning filter","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.db/commit/2fb4a5aa6aaa7d98861a9ae735c69f850a1f7f08"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2021-07-19 09:53:01.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2021-07-29 09:57:34.000000000","tz":60},"subject":"Replace use of Executable.execute method","message":"Replace use of Executable.execute method\n\nResolve the following RemovedIn20Warning warning:\n\n  The Executable.execute() method is considered legacy as of the 1.x\n  series of SQLAlchemy and will be removed in 2.0. All statement\n  execution in SQLAlchemy 2.0 is performed by the Connection.execute()\n  method of Connection, or in the ORM by the Session.execute() method of\n  Session.\n\nChange-Id: Ie0acba4a315c85ec7236e44a22449e0ad920ca9b\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.db/commit/471eb88d2790e95d564b448002c2465f68f2e041"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.db/commit/471eb88d2790e95d564b448002c2465f68f2e041"}]},"branch":"refs/heads/master"},"9659ec980cc8973d4c00e12f1a5c9e2781d7c8b3":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2021-07-29 15:37:53.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/83/802583/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.db","ref":"refs/changes/83/802583/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.db refs/changes/83/802583/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.db refs/changes/83/802583/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.db refs/changes/83/802583/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.db refs/changes/83/802583/4"}}},"commit":{"parents":[{"commit":"26d30ec7303c3bef86c4135b77f06cee7295ad03","subject":"Remove unnecessary warning filter","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.db/commit/26d30ec7303c3bef86c4135b77f06cee7295ad03"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2021-07-19 09:53:01.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2021-07-29 15:37:46.000000000","tz":60},"subject":"Replace use of Executable.execute method","message":"Replace use of Executable.execute method\n\nResolve the following RemovedIn20Warning warning:\n\n  The Executable.execute() method is considered legacy as of the 1.x\n  series of SQLAlchemy and will be removed in 2.0. All statement\n  execution in SQLAlchemy 2.0 is performed by the Connection.execute()\n  method of Connection, or in the ORM by the Session.execute() method of\n  Session.\n\nChange-Id: Ie0acba4a315c85ec7236e44a22449e0ad920ca9b\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.db/commit/9659ec980cc8973d4c00e12f1a5c9e2781d7c8b3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.db/commit/9659ec980cc8973d4c00e12f1a5c9e2781d7c8b3"}]},"branch":"refs/heads/master"},"62d77fc6df9bf27a334613d42a2ffa14c9711928":{"kind":"REWORK","_number":5,"created":"2021-08-10 16:45:25.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/83/802583/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.db","ref":"refs/changes/83/802583/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.db refs/changes/83/802583/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.db refs/changes/83/802583/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.db refs/changes/83/802583/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.db refs/changes/83/802583/5"}}},"commit":{"parents":[{"commit":"c320df474636ac22e9e778b1cce8975afae8100c","subject":"Remove unnecessary warning filter","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.db/commit/c320df474636ac22e9e778b1cce8975afae8100c"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2021-07-19 09:53:01.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2021-08-10 16:37:05.000000000","tz":60},"subject":"Replace use of Executable.execute method","message":"Replace use of Executable.execute method\n\nResolve the following RemovedIn20Warning warning:\n\n  The Executable.execute() method is considered legacy as of the 1.x\n  series of SQLAlchemy and will be removed in 2.0. All statement\n  execution in SQLAlchemy 2.0 is performed by the Connection.execute()\n  method of Connection, or in the ORM by the Session.execute() method of\n  Session.\n\nChange-Id: Ie0acba4a315c85ec7236e44a22449e0ad920ca9b\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.db/commit/62d77fc6df9bf27a334613d42a2ffa14c9711928"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.db/commit/62d77fc6df9bf27a334613d42a2ffa14c9711928"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":11816,"name":"mike bayer","display_name":"Mike Bayer","email":"mike_mp@zzzcomputing.com","username":"zzzeek","status":"Red Hat"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
