)]}'
{"id":"openstack%2Foslo.db~802584","triplet_id":"openstack%2Foslo.db~master~I44675fce86337696b6494abc03e8058af32686c6","project":"openstack/oslo.db","branch":"master","topic":"sqlalchemy-20","attention_set":{},"removed_from_attention_set":{"15334":{"account":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"last_update":"2021-11-12 18:46:06.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I44675fce86337696b6494abc03e8058af32686c6","subject":"Don\u0027t pass kwargs to connection.execute()","status":"MERGED","created":"2021-07-27 17:30:19.000000000","updated":"2021-11-12 18:47:54.000000000","submitted":"2021-11-12 18:46:06.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"802584-sqlalchemy-20","meta_rev_id":"2d9de637baeab921afcf2392889885cbd0a008b3","_number":802584,"virtual_id_number":802584,"owner":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":11816,"name":"mike bayer","display_name":"Mike Bayer","email":"mike_mp@zzzcomputing.com","username":"zzzeek","status":"Red Hat"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-11-12 18:46:05.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11816,"name":"mike bayer","display_name":"Mike Bayer","email":"mike_mp@zzzcomputing.com","username":"zzzeek","status":"Red Hat"},"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":2,"date":"2021-08-24 13:05:31.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11816,"name":"mike bayer","display_name":"Mike Bayer","email":"mike_mp@zzzcomputing.com","username":"zzzeek","status":"Red Hat"},{"value":2,"date":"2021-08-11 12:45:40.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":1,"date":"2021-11-12 10:20:51.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":11816,"name":"mike bayer","display_name":"Mike Bayer","email":"mike_mp@zzzcomputing.com","username":"zzzeek","status":"Red Hat"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":11816,"name":"mike bayer","display_name":"Mike Bayer","email":"mike_mp@zzzcomputing.com","username":"zzzeek","status":"Red Hat"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11816,"name":"mike bayer","display_name":"Mike Bayer","email":"mike_mp@zzzcomputing.com","username":"zzzeek","status":"Red Hat"},{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-07-27 19:01:07.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-08-11 12:45:40.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2021-08-24 13:05:31.000000000","updated_by":{"_account_id":11816,"name":"mike bayer","display_name":"Mike Bayer","email":"mike_mp@zzzcomputing.com","username":"zzzeek","status":"Red Hat"},"reviewer":{"_account_id":11816,"name":"mike bayer","display_name":"Mike Bayer","email":"mike_mp@zzzcomputing.com","username":"zzzeek","status":"Red Hat"},"state":"REVIEWER"}],"messages":[{"id":"5b286bf7bef628f29460f38452eb8b1b0a04dcfb","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2021-07-27 17:30:19.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"837e987395ed75dd47fcd708ad1c7e749ef53fe2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-27 19:01:07.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/414e8a06f442478a9b8e59b311528b22 : FAILURE in 58m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a71c7ef3c4bc4cae88f8a7041fa65606 : FAILURE in 6m 40s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c26d746884ee4fe9953b90bc57451b81 : FAILURE in 7m 07s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5f6e5e6c60d547769d9dc52e150780ec : FAILURE in 7m 17s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/610ab6ca69ff47ae899bd37d5b343f28 : FAILURE in 6m 48s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8ff14dffe51c4cc4962949e9a1560e81 : SUCCESS in 7m 04s\n\nWarning:\n  Comments left for invalid file oslo_db/sqlalchemy/engines.py","accounts_in_message":[],"_revision_number":1},{"id":"ff7eb301caf8eba9657c5b3ab66183988a26c7e8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2021-07-28 14:38:18.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.","accounts_in_message":[],"_revision_number":2},{"id":"b30d080c499e5ae2bde66407d07304dbc76b6aa9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-28 16:00:07.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/6914331e78ff4287bc26188f2be6a811 : SUCCESS in 1h 15m 56s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9a9e0ce38fc3493dbe991c8b5c151523 : SUCCESS in 7m 12s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c8b5196c848b44229140b10ed3b2f7a5 : FAILURE in 7m 25s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7a18305f9c9e477091fcfa59c0102ea6 : FAILURE in 8m 38s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/21dfb2b685644e5e82e3877c77145e91 : FAILURE in 8m 09s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4a6e35470d664f67a8be5e98d476861a : SUCCESS in 7m 26s","accounts_in_message":[],"_revision_number":2},{"id":"11ca893103114299958b00097f9fa30a8dcebf15","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2021-07-29 10:44:29.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"8a43de6524cfff9a39dcdd572936cf5d56f5f439","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-29 12:20:59.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/ccf4f67fc2d5494285bb0583925809dd : SUCCESS in 1h 26m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7b3afdb0ba3f4eef8db95dbead585109 : SUCCESS in 6m 08s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ebea93d296ad447e997deb08b9831ddc : SUCCESS in 7m 08s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/640ba869192f4b8b81c8ee4a752a0be6 : SUCCESS in 6m 11s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f468acd747234e298f48d108d5f1ef5a : SUCCESS in 6m 53s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1094c390b273497fbbb4f95d9e86fd80 : SUCCESS in 6m 34s","accounts_in_message":[],"_revision_number":3},{"id":"c7cc714c03cc262117d4ae0a517d38a7f42323c0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2021-07-29 15:37:53.000000000","message":"Uploaded patch set 4: Patch Set 3 was rebased.","accounts_in_message":[],"_revision_number":4},{"id":"b87b28774bdd940adcb34cad4ed301f30ed6da47","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-29 17:08:22.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/d67a52eba4b74a5ab5182ade86bcc249 : SUCCESS in 1h 25m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/380e94d2c4334bba8315c6d0e8347e4e : SUCCESS in 9m 20s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e5aeb58a36ed4178bbd3ffd040f5ae17 : SUCCESS in 7m 44s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6b1d6511aebc41148ef7b6a348abae05 : SUCCESS in 7m 52s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ef64e8b79d8d4b2688777ec3aab482e3 : SUCCESS in 8m 29s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f041e334a02e4b92a3931aa17a7e9c6e : SUCCESS in 9m 37s","accounts_in_message":[],"_revision_number":4},{"id":"a7ffb5139d449bad3baf64ae339e727a4eb370da","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2021-08-10 16:45:25.000000000","message":"Uploaded patch set 5: Patch Set 4 was rebased.","accounts_in_message":[],"_revision_number":5},{"id":"08ffd6bc959a3b212fb5e3bb064cf502280a4998","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-10 18:38:06.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/7d312d246f85444d9ae3e0c0d111e535 : SUCCESS in 1h 17m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d001ceac1e6247a5b7730b038d254774 : SUCCESS in 5m 28s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4a7daac106dd4eb7a1e95cfe2770eb79 : SUCCESS in 5m 31s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/409d309481524201bbd9f9a6a814c06c : SUCCESS in 7m 34s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/1c1fb97e87274145b0d4242dcf47c1eb : SUCCESS in 6m 23s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b7fc54003e6b4ad39a646b4acdbe31c3 : SUCCESS in 6m 12s","accounts_in_message":[],"_revision_number":5},{"id":"fb1f35c1827509a25c85a8aecac20c50847a1517","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-08-11 12:45:40.000000000","message":"Patch Set 5: Code-Review+2","accounts_in_message":[],"_revision_number":5},{"id":"44a67141e7b75124ac74f73df934574de7895aa2","author":{"_account_id":11816,"name":"mike bayer","display_name":"Mike Bayer","email":"mike_mp@zzzcomputing.com","username":"zzzeek","status":"Red Hat"},"date":"2021-08-24 13:05:31.000000000","message":"Patch Set 5: Code-Review+2","accounts_in_message":[],"_revision_number":5},{"id":"f4f2c9213b21db96fe19e4da5f023fb196629680","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2021-11-12 10:20:51.000000000","message":"Patch Set 5: Workflow+1","accounts_in_message":[],"_revision_number":5},{"id":"56909f7075102454fb0f947df4686988e162ed6c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-12 10:22:15.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"2948a3f9eb62dca1c6a2989f6da9d1a1b74696ee","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-12 12:20:51.000000000","message":"Patch Set 5: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/49f5784744e846d8a011c4be690b9505 : FAILURE in 1h 00m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/36154bbf18f14404b52f251dfeed721b : SUCCESS in 5m 29s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/55fc40a090d64158952bb21f4b87b1a3 : SUCCESS in 5m 55s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/945b0c9e80ec4183b95344aef77ad6d4 : SUCCESS in 5m 45s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/530f9266dd564052b16d70891819a7af : SUCCESS in 6m 41s","accounts_in_message":[],"_revision_number":5},{"id":"1a9925f23ac77c30b2cadb6eed0adba80b1a0554","author":{"_account_id":11816,"name":"mike bayer","display_name":"Mike Bayer","email":"mike_mp@zzzcomputing.com","username":"zzzeek","status":"Red Hat"},"date":"2021-11-12 15:37:41.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"49b62ecfbce765bfffe1237a0aa4ebb27b4afb95","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-12 17:23:04.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/31cc7d3a8f0a4b83b91068ceed5c4a5f : SUCCESS in 1h 42m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1393dfd63f2f4b3aaa0e139b55328648 : SUCCESS in 5m 54s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/92bc1310fcc446919b055bf36396c936 : SUCCESS in 6m 32s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0ca06bcf4b844226a16924b1c9754975 : SUCCESS in 6m 23s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/7b6f11f38ecf4e86bff24e162f0c3267 : SUCCESS in 6m 22s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/162a672b1566481dad4f9e3b0728ce77 : SUCCESS in 5m 06s","accounts_in_message":[],"_revision_number":5},{"id":"a8644b3ddea0b1d8d3be12021cca392c01acafae","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-12 17:24:14.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"f2066a99fdd418a6cfe8d3f89a9b050336399a0b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-12 18:46:05.000000000","message":"Patch Set 5: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/917389b496b0494a85cc42f0246d5a67 : SUCCESS in 1h 20m 56s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/95468e1586f0401b84ed3580148d6a2a : SUCCESS in 5m 32s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/32181c8749214d32a8936192071e650c : SUCCESS in 6m 24s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5b85cd7b14a141db95e93444cd37d55d : SUCCESS in 7m 00s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c79535223f674e8fb7bee548a02cb0f5 : SUCCESS in 5m 52s","accounts_in_message":[],"_revision_number":5},{"id":"655fcd554999f38a5cdab8fe6df7479076ad9a99","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-12 18:46:06.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":5},{"id":"2d9de637baeab921afcf2392889885cbd0a008b3","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-12 18:47:54.000000000","message":"Patch Set 5:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1d3626d732864b0e85f2200baef0a20c : SUCCESS in 1m 02s","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"e1039e0849890c6bd10516004c6dab89db1c9111","revisions":{"9177ba989412fcb44d3db519c6a27300195d5f57":{"kind":"REWORK","_number":1,"created":"2021-07-27 17:30:19.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/84/802584/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.db","ref":"refs/changes/84/802584/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.db refs/changes/84/802584/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.db refs/changes/84/802584/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.db refs/changes/84/802584/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.db refs/changes/84/802584/1"}}},"commit":{"parents":[{"commit":"20ab6c380f1db0abc69cdd42c15a8f6e996b8521","subject":"Replace use of Executable.execute method","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.db/commit/20ab6c380f1db0abc69cdd42c15a8f6e996b8521"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2021-07-19 10:36:01.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2021-07-19 16:44:17.000000000","tz":60},"subject":"Don\u0027t pass kwargs to connection.execute()","message":"Don\u0027t pass kwargs to connection.execute()\n\nResolve the following RemovedIn20Warning warning:\n\n  The connection.execute() method in SQLAlchemy 2.0 will accept\n  parameters as a single dictionary or a single sequence of dictionaries\n  only. Parameters passed as keyword arguments, tuples or positionally\n  oriented dictionaries and/or tuples will no longer be accepted.\n\nChange-Id: I44675fce86337696b6494abc03e8058af32686c6\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.db/commit/9177ba989412fcb44d3db519c6a27300195d5f57"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.db/commit/9177ba989412fcb44d3db519c6a27300195d5f57"}]},"branch":"refs/heads/master"},"371dea3368bb5c738490532afc817d69e13b8439":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2021-07-28 14:38:18.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/84/802584/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.db","ref":"refs/changes/84/802584/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.db refs/changes/84/802584/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.db refs/changes/84/802584/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.db refs/changes/84/802584/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.db refs/changes/84/802584/2"}}},"commit":{"parents":[{"commit":"e9fc6a0eaa0a99c5d70dffca292cbe2ec225019c","subject":"Replace use of Executable.execute method","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.db/commit/e9fc6a0eaa0a99c5d70dffca292cbe2ec225019c"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2021-07-19 10:36:01.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2021-07-28 14:38:06.000000000","tz":60},"subject":"Don\u0027t pass kwargs to connection.execute()","message":"Don\u0027t pass kwargs to connection.execute()\n\nResolve the following RemovedIn20Warning warning:\n\n  The connection.execute() method in SQLAlchemy 2.0 will accept\n  parameters as a single dictionary or a single sequence of dictionaries\n  only. Parameters passed as keyword arguments, tuples or positionally\n  oriented dictionaries and/or tuples will no longer be accepted.\n\nChange-Id: I44675fce86337696b6494abc03e8058af32686c6\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.db/commit/371dea3368bb5c738490532afc817d69e13b8439"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.db/commit/371dea3368bb5c738490532afc817d69e13b8439"}]},"branch":"refs/heads/master"},"b928ed659fab428fb21ffc0d2c55e598a0bea1c1":{"kind":"REWORK","_number":3,"created":"2021-07-29 10:44:29.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/84/802584/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.db","ref":"refs/changes/84/802584/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.db refs/changes/84/802584/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.db refs/changes/84/802584/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.db refs/changes/84/802584/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.db refs/changes/84/802584/3"}}},"commit":{"parents":[{"commit":"471eb88d2790e95d564b448002c2465f68f2e041","subject":"Replace use of Executable.execute method","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.db/commit/471eb88d2790e95d564b448002c2465f68f2e041"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2021-07-19 10:36:01.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2021-07-29 10:23:04.000000000","tz":60},"subject":"Don\u0027t pass kwargs to connection.execute()","message":"Don\u0027t pass kwargs to connection.execute()\n\nResolve the following RemovedIn20Warning warning:\n\n  The connection.execute() method in SQLAlchemy 2.0 will accept\n  parameters as a single dictionary or a single sequence of dictionaries\n  only. Parameters passed as keyword arguments, tuples or positionally\n  oriented dictionaries and/or tuples will no longer be accepted.\n\nChange-Id: I44675fce86337696b6494abc03e8058af32686c6\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.db/commit/b928ed659fab428fb21ffc0d2c55e598a0bea1c1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.db/commit/b928ed659fab428fb21ffc0d2c55e598a0bea1c1"}]},"branch":"refs/heads/master"},"33f67199f11bac21261a7549a1b593f4d2e2341b":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2021-07-29 15:37:53.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/84/802584/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.db","ref":"refs/changes/84/802584/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.db refs/changes/84/802584/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.db refs/changes/84/802584/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.db refs/changes/84/802584/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.db refs/changes/84/802584/4"}}},"commit":{"parents":[{"commit":"9659ec980cc8973d4c00e12f1a5c9e2781d7c8b3","subject":"Replace use of Executable.execute method","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.db/commit/9659ec980cc8973d4c00e12f1a5c9e2781d7c8b3"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2021-07-19 10:36:01.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2021-07-29 15:37:46.000000000","tz":60},"subject":"Don\u0027t pass kwargs to connection.execute()","message":"Don\u0027t pass kwargs to connection.execute()\n\nResolve the following RemovedIn20Warning warning:\n\n  The connection.execute() method in SQLAlchemy 2.0 will accept\n  parameters as a single dictionary or a single sequence of dictionaries\n  only. Parameters passed as keyword arguments, tuples or positionally\n  oriented dictionaries and/or tuples will no longer be accepted.\n\nChange-Id: I44675fce86337696b6494abc03e8058af32686c6\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.db/commit/33f67199f11bac21261a7549a1b593f4d2e2341b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.db/commit/33f67199f11bac21261a7549a1b593f4d2e2341b"}]},"branch":"refs/heads/master"},"e1039e0849890c6bd10516004c6dab89db1c9111":{"kind":"TRIVIAL_REBASE","_number":5,"created":"2021-08-10 16:45:25.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/84/802584/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.db","ref":"refs/changes/84/802584/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.db refs/changes/84/802584/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.db refs/changes/84/802584/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.db refs/changes/84/802584/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.db refs/changes/84/802584/5"}}},"commit":{"parents":[{"commit":"62d77fc6df9bf27a334613d42a2ffa14c9711928","subject":"Replace use of Executable.execute method","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.db/commit/62d77fc6df9bf27a334613d42a2ffa14c9711928"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2021-07-19 10:36:01.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2021-08-10 16:37:45.000000000","tz":60},"subject":"Don\u0027t pass kwargs to connection.execute()","message":"Don\u0027t pass kwargs to connection.execute()\n\nResolve the following RemovedIn20Warning warning:\n\n  The connection.execute() method in SQLAlchemy 2.0 will accept\n  parameters as a single dictionary or a single sequence of dictionaries\n  only. Parameters passed as keyword arguments, tuples or positionally\n  oriented dictionaries and/or tuples will no longer be accepted.\n\nChange-Id: I44675fce86337696b6494abc03e8058af32686c6\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.db/commit/e1039e0849890c6bd10516004c6dab89db1c9111"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.db/commit/e1039e0849890c6bd10516004c6dab89db1c9111"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":11816,"name":"mike bayer","display_name":"Mike Bayer","email":"mike_mp@zzzcomputing.com","username":"zzzeek","status":"Red Hat"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
